)]}'
{"id":"openstack%2Frequirements~675401","triplet_id":"openstack%2Frequirements~master~I4bec7591cd7b5e9ca4ba84517554081b057acfef","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4bec7591cd7b5e9ca4ba84517554081b057acfef","subject":"update constraint for python-designateclient to new release 2.12.0","status":"MERGED","created":"2019-08-08 16:19:17.000000000","updated":"2019-08-09 00:31:43.000000000","submitted":"2019-08-09 00:31:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675401-1565310703340-92a0a85e","meta_rev_id":"7c201416e376b9ab6c273d678695a2acc2959f01","_number":675401,"virtual_id_number":675401,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-08-09 00:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-08 21:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-08 20:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-08 21:59:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 20:01:09.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-08-08 21:59:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-09 00:31:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b847b63c33f0f64c2d625db2552a75cd6cedc9ae","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-08 16:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c41f59a3446326b1d380ed5b266304215d6cfbe","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-08-08 20:01:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf4151c6e77a7cc923cf80289ab108e99e2f5522","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-08 21:59:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f7b65d2702d62a0a462cee6e729c448ec829793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:35:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/01/675401/1/check/openstack-tox-docs/d06647e/html/ : SUCCESS in 3m 04s\n- openstack-tox-pypy https://logs.opendev.org/01/675401/1/check/openstack-tox-pypy/cec4bce/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/01/675401/1/check/openstack-tox-pep8/63206de/ : SUCCESS in 4m 36s\n- openstack-tox-py27 https://logs.opendev.org/01/675401/1/check/openstack-tox-py27/945a985/ : SUCCESS in 3m 30s\n- openstack-tox-py36 https://logs.opendev.org/01/675401/1/check/openstack-tox-py36/6c15052/ : SUCCESS in 4m 49s\n- openstack-tox-py37 https://logs.opendev.org/01/675401/1/check/openstack-tox-py37/e17b590/ : SUCCESS in 4m 07s\n- openstack-tox-validate https://logs.opendev.org/01/675401/1/check/openstack-tox-validate/50b204c/ : SUCCESS in 3m 38s\n- requirements-tox-babel https://logs.opendev.org/01/675401/1/check/requirements-tox-babel/a98e156/ : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://logs.opendev.org/01/675401/1/check/requirements-tox-py27-check-uc/08ae7f4/ : SUCCESS in 8m 27s\n- requirements-tox-py36-check-uc https://logs.opendev.org/01/675401/1/check/requirements-tox-py36-check-uc/f7dba31/ : SUCCESS in 9m 30s\n- cross-cinder-py27 https://logs.opendev.org/01/675401/1/check/cross-cinder-py27/3bc54b3/ : SUCCESS in 10m 02s\n- cross-cinder-py36 https://logs.opendev.org/01/675401/1/check/cross-cinder-py36/9882b77/ : SUCCESS in 8m 18s\n- cross-glance-py27 https://logs.opendev.org/01/675401/1/check/cross-glance-py27/3da0125/ : SUCCESS in 4m 11s\n- cross-glance-py36 https://logs.opendev.org/01/675401/1/check/cross-glance-py36/2b9a206/ : SUCCESS in 4m 30s\n- cross-horizon-py27 https://logs.opendev.org/01/675401/1/check/cross-horizon-py27/f0d6d44/ : SUCCESS in 9m 06s\n- cross-horizon-py36 https://logs.opendev.org/01/675401/1/check/cross-horizon-py36/d253c82/ : SUCCESS in 10m 26s\n- cross-horizon-npm https://logs.opendev.org/01/675401/1/check/cross-horizon-npm/3ff8724/ : SUCCESS in 6m 59s\n- cross-keystone-py27 https://logs.opendev.org/01/675401/1/check/cross-keystone-py27/43e1688/ : SUCCESS in 32m 25s\n- cross-keystone-py36 https://logs.opendev.org/01/675401/1/check/cross-keystone-py36/ef9bd16/ : SUCCESS in 30m 37s\n- cross-neutron-py27 https://logs.opendev.org/01/675401/1/check/cross-neutron-py27/5a26953/ : SUCCESS in 28m 32s\n- cross-neutron-py36 https://logs.opendev.org/01/675401/1/check/cross-neutron-py36/7097224/ : SUCCESS in 26m 34s\n- cross-nova-py27 https://logs.opendev.org/01/675401/1/check/cross-nova-py27/2642d14/ : SUCCESS in 16m 32s\n- cross-nova-py36 https://logs.opendev.org/01/675401/1/check/cross-nova-py36/4e23781/ : SUCCESS in 15m 20s\n- cross-nova-functional https://logs.opendev.org/01/675401/1/check/cross-nova-functional/542ce9a/ : SUCCESS in 24m 12s\n- cross-osc-tox-docs https://logs.opendev.org/01/675401/1/check/cross-osc-tox-docs/11e7e0e/html/ : SUCCESS in 5m 28s\n- cross-swift-py27 https://logs.opendev.org/01/675401/1/check/cross-swift-py27/5642e71/ : SUCCESS in 18m 16s\n- cross-swift-py36 https://logs.opendev.org/01/675401/1/check/cross-swift-py36/a4132a9/ : SUCCESS in 19m 24s\n- requirements-integration https://logs.opendev.org/01/675401/1/check/requirements-integration/51fc894/ : SUCCESS in 59m 53s\n- neutron-grenade-multinode https://logs.opendev.org/01/675401/1/check/neutron-grenade-multinode/9d4c8da/ : SUCCESS in 1h 11m 44s\n- tempest-full https://logs.opendev.org/01/675401/1/check/tempest-full/55ca11d/ : SUCCESS in 1h 53m 34s\n- neutron-grenade https://logs.opendev.org/01/675401/1/check/neutron-grenade/545529f/ : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":1},{"id":"84924f414f27b2167a613ede4977ab347997f5ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:36:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"180ad0b4e12da7a27ab62c8b1a26433833011c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 00:31:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec43bb55b8cbfeb76a00e3c77e7c4bb932a4ea08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 00:31:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/01/675401/1/gate/openstack-tox-docs/01e6f50/html/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://logs.opendev.org/01/675401/1/gate/openstack-tox-pep8/9aef8d9/ : SUCCESS in 3m 46s\n- openstack-tox-py27 https://logs.opendev.org/01/675401/1/gate/openstack-tox-py27/8f8e9fd/ : SUCCESS in 3m 51s\n- openstack-tox-py36 https://logs.opendev.org/01/675401/1/gate/openstack-tox-py36/06ba170/ : SUCCESS in 3m 18s\n- openstack-tox-py37 https://logs.opendev.org/01/675401/1/gate/openstack-tox-py37/7820e98/ : SUCCESS in 10m 04s\n- openstack-tox-validate https://logs.opendev.org/01/675401/1/gate/openstack-tox-validate/2f95e53/ : SUCCESS in 7m 58s\n- requirements-tox-babel https://logs.opendev.org/01/675401/1/gate/requirements-tox-babel/8b5f52c/ : SUCCESS in 6m 25s\n- requirements-tox-py27-check-uc https://logs.opendev.org/01/675401/1/gate/requirements-tox-py27-check-uc/5483a77/ : SUCCESS in 21m 14s\n- requirements-tox-py36-check-uc https://logs.opendev.org/01/675401/1/gate/requirements-tox-py36-check-uc/c01e834/ : SUCCESS in 21m 22s\n- cross-cinder-py27 https://logs.opendev.org/01/675401/1/gate/cross-cinder-py27/93a8a45/ : SUCCESS in 9m 31s\n- cross-cinder-py36 https://logs.opendev.org/01/675401/1/gate/cross-cinder-py36/4f9fb3b/ : SUCCESS in 8m 27s\n- cross-glance-py27 https://logs.opendev.org/01/675401/1/gate/cross-glance-py27/3e639fc/ : SUCCESS in 9m 04s\n- cross-glance-py36 https://logs.opendev.org/01/675401/1/gate/cross-glance-py36/2e0938e/ : SUCCESS in 6m 04s\n- cross-horizon-py27 https://logs.opendev.org/01/675401/1/gate/cross-horizon-py27/b583966/ : SUCCESS in 11m 44s\n- cross-horizon-py36 https://logs.opendev.org/01/675401/1/gate/cross-horizon-py36/b2505d4/ : SUCCESS in 9m 46s\n- cross-horizon-npm https://logs.opendev.org/01/675401/1/gate/cross-horizon-npm/36f7d5a/ : SUCCESS in 7m 11s\n- cross-keystone-py27 https://logs.opendev.org/01/675401/1/gate/cross-keystone-py27/35321ca/ : SUCCESS in 29m 44s\n- cross-keystone-py36 https://logs.opendev.org/01/675401/1/gate/cross-keystone-py36/43d77fa/ : SUCCESS in 28m 20s\n- cross-neutron-py27 https://logs.opendev.org/01/675401/1/gate/cross-neutron-py27/cef44ae/ : SUCCESS in 25m 41s\n- cross-neutron-py36 https://logs.opendev.org/01/675401/1/gate/cross-neutron-py36/db34395/ : SUCCESS in 24m 23s\n- cross-nova-py27 https://logs.opendev.org/01/675401/1/gate/cross-nova-py27/1cbe4f2/ : SUCCESS in 25m 11s\n- cross-nova-py36 https://logs.opendev.org/01/675401/1/gate/cross-nova-py36/c7aa461/ : SUCCESS in 23m 58s\n- cross-nova-functional https://logs.opendev.org/01/675401/1/gate/cross-nova-functional/3bf75c8/ : SUCCESS in 28m 54s\n- cross-osc-tox-docs https://logs.opendev.org/01/675401/1/gate/cross-osc-tox-docs/6cfbe1f/html/ : SUCCESS in 4m 58s\n- cross-swift-py27 https://logs.opendev.org/01/675401/1/gate/cross-swift-py27/9fb1b9a/ : SUCCESS in 8m 45s\n- cross-swift-py36 https://logs.opendev.org/01/675401/1/gate/cross-swift-py36/20362c9/ : SUCCESS in 7m 18s\n- requirements-integration https://logs.opendev.org/01/675401/1/gate/requirements-integration/df4032b/ : SUCCESS in 59m 23s\n- neutron-grenade-multinode https://logs.opendev.org/01/675401/1/gate/neutron-grenade-multinode/0f5e9c8/ : SUCCESS in 1h 15m 35s\n- tempest-full https://logs.opendev.org/01/675401/1/gate/tempest-full/0726f61/ : SUCCESS in 1h 49m 51s\n- neutron-grenade https://logs.opendev.org/01/675401/1/gate/neutron-grenade/7408713/ : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"553c279f8e0ce643315564f89b169f5f2b82e09e","revisions":{"553c279f8e0ce643315564f89b169f5f2b82e09e":{"kind":"REWORK","_number":1,"created":"2019-08-08 16:19:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/675401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/675401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/675401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/675401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/675401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/675401/1"}}},"commit":{"parents":[{"commit":"f938e71e8c79362d5ff31af4f87a3c0f45c33f46","subject":"Merge \"update constraint for pbr to new release 5.4.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f938e71e8c79362d5ff31af4f87a3c0f45c33f46"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-08 16:19:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-08 16:19:15.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 2.12.0","message":"update constraint for python-designateclient to new release 2.12.0\n\nChange-Id: I4bec7591cd7b5e9ca4ba84517554081b057acfef\nmeta:version: 2.12.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Dirk Mueller \u003cdirk@dmllr.de\u003e\nmeta:release:Commit: Dirk Mueller \u003cdirk@dmllr.de\u003e\nmeta:release:Change-Id: I4019cfff92b0174721177aaa175c3ef4da2dffaa\nmeta:release:Code-Review+1: Graham Hayes \u003cgr@ham.ie\u003e\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/553c279f8e0ce643315564f89b169f5f2b82e09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/553c279f8e0ce643315564f89b169f5f2b82e09e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
