)]}'
{"id":"openstack%2Frequirements~676401","triplet_id":"openstack%2Frequirements~master~I7e731a30e12a8a3a5ee0b27524064e87275c010d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7e731a30e12a8a3a5ee0b27524064e87275c010d","subject":"update constraint for monasca-common to new release 2.16.0","status":"MERGED","created":"2019-08-14 12:21:53.000000000","updated":"2019-08-15 00:24:56.000000000","submitted":"2019-08-15 00:24:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676401-1565828696144-070cffcc","meta_rev_id":"a253e9ff4468901c10ba159117bd363d50ffa1f1","_number":676401,"virtual_id_number":676401,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-15 00:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-14 16:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-14 21:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-14 21:47:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-14 13:33:07.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 13:33:07.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-08-14 16:42:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-14 21:47:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-15 00:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6390c9f56ad902e117abef3f739d45b315058785","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-14 12:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"060754148253553e20f653a3c824eeeb92eecbab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d0f97115ec48d5a6cea8bf1fef7ddb : SUCCESS in 5m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/affbd175bb114df9a06f3572b209e013 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df57dd6d7cc4e9894b802be63bca7d6 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef5fb698b34341b3b9f2c11bbc3e6475 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80fffa848a4e485c8aeebd65421c8f30 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ffddb050f34458390af14960560e421 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/312446530e9b4a72aecdba0a1644afd4 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/195c3f87ec904059b22ad0ab47ca9bcb : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cdc203b907b744ba8f300e73c41cce92 : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c777da8879a0438db66a4cf01415917d : SUCCESS in 7m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9fadfb61499e485f9dae13bbe93f9bcc : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2d9f5887c4c48249e1ce2d4c6fd785e : SUCCESS in 8m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/632a605ebeb949de899d90601a79588f : SUCCESS in 5m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b1059579cdd4ff58ed272a84a13c113 : SUCCESS in 8m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9229e2bbae4147f3bd17df3dc40e85e3 : SUCCESS in 11m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ebe541e189442858ebb83491fcf2a26 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b60e05405374453b940e74f83ec1cb9 : SUCCESS in 6m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cf784f61e2f94c59a1a856d755890b73 : SUCCESS in 26m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aefc4296630f41a6bc3320172448d53a : SUCCESS in 28m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/82f0d967d700449a9fae166a33af6bae : SUCCESS in 19m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/caf583e8d5584ade9c332d0ce5e9a109 : SUCCESS in 25m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b4b781b820904d3d975c7a593f4a1cf5 : SUCCESS in 20m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ba254899f9f4918aae4fbd265a930e4 : SUCCESS in 16m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/234dc54e6998465a87effbd29671f6db : SUCCESS in 21m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e6e5d4ffe5e4911ba900ce9780be939 : SUCCESS in 5m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d0c3007ffb143cea68e49462eb9c3c4 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d750ce5990045069bc0ef6c84aff34d : SUCCESS in 7m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/94d5e804630e4e0f8e016f1f44c95c78 : SUCCESS in 1h 09m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba2b2f95b8714c5097b12b84ba5f1856 : SUCCESS in 1h 16m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cfa4ae69a33b444ea48a7179ff9ee5cf : SUCCESS in 1h 43m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4269a90cb5e64cc186735a137f35109f : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d8d00135f2793f5d3db2bb42d701590ca8c6b040","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 16:42:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5fd5a2a2cf39ac8faa702dcc86aa62a20aa3fcf5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-14 21:47:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2271f296aade6be2cfcfa2b061d4b0b499ee6e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:48:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c111cff238e77c7900fe918a06ac89f5381dabb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:24:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4006607fdaf5428785d64c706a7f83d1 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4eab23fd1545d48c36aab7de5df87a : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13d8a2ee19264d25b6bf8d3f7ed49156 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03303bfc440642c0b35c6692f04bd815 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e33770187cf4be1a80ddcb6ea122166 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50f537174986492fb2b6989ecb32996f : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71967da408034a399ee564fc229b99ec : SUCCESS in 4m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/69bc1a75a459469e96940a4a7da851a2 : SUCCESS in 10m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/efede83a2ac4485ea3883132aa0fc54f : SUCCESS in 8m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/492c77c7c3bd4370a7f6c1c4dde3f4de : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/738c1a72e58b49df800a4a1ce6806478 : SUCCESS in 8m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/46dff3a3c3b3464b81315ee3509d0ae2 : SUCCESS in 4m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5041e34d3cb84b17836478e7b934887e : SUCCESS in 6m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/34abbb61487047cb9fd2b3f9d1f6630c : SUCCESS in 9m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c74303f890504832a65fc921768689eb : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2dc7a9de80dc47629c4d56c9a9c9430c : SUCCESS in 5m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8c7c5a7a5dcb450ebe10ba6c1cbf3881 : SUCCESS in 27m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/00347414ea5d42b098363fc99054e892 : SUCCESS in 24m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f2d25f2c30d84d76a331e87a94b6cde7 : SUCCESS in 25m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d57d50805f8045a69a103b724a3d67b4 : SUCCESS in 22m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/abfcb511916d45a2beabdab346cc55b9 : SUCCESS in 16m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8ca09439fe214c92b8204d2f498e31f4 : SUCCESS in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5756f678eaff497aa7c56f6db5c3ba6a : SUCCESS in 18m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee1552079a0c4e6f936a9fbbd581623e : SUCCESS in 4m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2fdcad80cf954f9eb521630c28b5e40d : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/88f8066a0e9b4f7e86d71199474cc8f2 : SUCCESS in 8m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6a8f0b691d6549379afde6fc9a97ef13 : SUCCESS in 52m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77b3a8799ceb4fb3a8d36fc5e8e08234 : SUCCESS in 1h 10m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e3c5afd394ed4962afc1a940329efa44 : SUCCESS in 1h 40m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b41f2b7778514783b304d9f5855b3c88 : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7740d3d6fb153ca910ca28f2727e94fbb56c6122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:24:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19ad1c5ed79d7087c3ddc32dd5040abdd81b8cf6","revisions":{"19ad1c5ed79d7087c3ddc32dd5040abdd81b8cf6":{"kind":"REWORK","_number":1,"created":"2019-08-14 12:21:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/676401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/676401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/676401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/676401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/676401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/676401/1"}}},"commit":{"parents":[{"commit":"9d9cf9a640553a386a2ed150ef2cba4fea86865c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d9cf9a640553a386a2ed150ef2cba4fea86865c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-14 12:21:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-14 12:21:51.000000000","tz":0},"subject":"update constraint for monasca-common to new release 2.16.0","message":"update constraint for monasca-common to new release 2.16.0\n\nChange-Id: I7e731a30e12a8a3a5ee0b27524064e87275c010d\nmeta:version: 2.16.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Witek Bedyk \u003cwitold.bedyk@suse.com\u003e\nmeta:release:Commit: Witek Bedyk \u003cwitold.bedyk@suse.com\u003e\nmeta:release:Change-Id: I7459352de6ab12a95b7014b274e9ab8936e480bb\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19ad1c5ed79d7087c3ddc32dd5040abdd81b8cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19ad1c5ed79d7087c3ddc32dd5040abdd81b8cf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
