)]}'
{"id":"openstack%2Frequirements~677052","triplet_id":"openstack%2Frequirements~master~I79baeee85f93adb2adfd01b1f02ca511073c8d78","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I79baeee85f93adb2adfd01b1f02ca511073c8d78","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-08-17 06:27:58.000000000","updated":"2019-08-20 06:20:14.000000000","submitted":"2019-08-20 06:20:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677052-1566282014301-5d7c1590","meta_rev_id":"701203a273a4cb34e82ce3eee5014144d0b6046e","_number":677052,"virtual_id_number":677052,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-20 06:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-19 22:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-19 22:26:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 22:26:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-20 06:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5561ec180bada64a78dd812c6bbeee19757af10a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-17 06:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a0b27dd3988ea0275305be8cd255794bf180ae8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 07:53:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/947c85093032461881247d39c751e281 : SUCCESS in 4m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/34c97b338aac4affb0a964c2928656d5 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3586d5794e7148848156496c66ef03c7 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7cc3dc507c345a78e9366a5542db22b : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b62e9d73a234934b5a2f4738753e02f : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60f4eadf15fd48ce8ec75f33fbea902b : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de8ac5847d334c538a96a1d7bf00fb50 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a87ec56010984632bddf470425e8161c : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c848fe39d24240f89eb982485223aba6 : FAILURE in 9m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/244b978a2c7f4e839ca6219a47c33111 : FAILURE in 7m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2d23b3e5607f496da838830337a92708 : SUCCESS in 14m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb7ade9efc0f4c3582e2a0e4e7bed475 : SUCCESS in 9m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b39e013763f24b8ebefe022853788a84 : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a08f9dae0e634f768a5fd57890fe99d0 : SUCCESS in 4m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2b94dc4441d84e8e93ea7048e03a1e8d : SUCCESS in 11m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2830a5c1c8f244b89b0f2311f9428427 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84cbd82103d74c28bd9cad51322cb5d3 : SUCCESS in 7m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/eaab25bda3654da19d1af01b2ad06388 : FAILURE in 31m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3a65c27ba3854d16afd179f8d2447b0d : FAILURE in 29m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d595f547da7346aaaaaf3e1381812083 : FAILURE in 23m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/45d9885313544006ac2c67867859456d : FAILURE in 25m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dd48cbd9ccb74b8f959c0873d4af1f8f : SUCCESS in 11m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/26c5bf98bb0e49b08446412f856d57e3 : FAILURE in 11m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/983ffa8421914124a57e775dc9795f91 : SUCCESS in 22m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22ca187458944a0b995a1c3c453827d4 : SUCCESS in 5m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9f7bed23005540529337289220744df8 : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87d79ffd5415427685c4987c8cdaae2f : SUCCESS in 9m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5232f5d5aa5a4557ae899c750e654800 : SUCCESS in 55m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a90a79969e04f05bef24cc731533381 : FAILURE in 1h 15m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1cc9c3afc8d7410eb8043a5ef95b6dd8 : FAILURE in 1h 20m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1c2226c9546f484f91bc1d2518d21dfe : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e1139de161fc11a64b3f83d2ff684368b99e8267","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-17 16:22:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a3bd1c70c3cea7fac4704c8f7b98ea0154bcff3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-17 16:22:40.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"518df4c743b323830e053d1f54e7a752b32d688e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-17 16:31:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"386c2b5eb2884280dc4c987f83ccf142c844b5b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-17 16:32:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2316a961176e1e78457ac09b3dc90d42d10787b5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-17 16:37:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"850ac2e9f51b80ac5083a7641c2365813368ea3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 18:19:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319146cf747b40bb8297e1e1b8c87119 : SUCCESS in 4m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a48fbdcafc7c48c9baf905d77ebec1f3 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eae2db23df044159da84df757929c0a : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb09074b5828422195ff833becb268f5 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d4b00d035ea4c7d941bc5df8d8036c4 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b32bf526e0fa44f5bc834d666c1bf26f : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39e1730681444efb97cb74fe74b28fb2 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/245d7bdd12ed4024a4588591454649f4 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d8f8af8f0bf74200a58364258089e39e : SUCCESS in 7m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2241a4d0acb45408d06263fbbb70093 : SUCCESS in 7m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9def43d247f34361b74dfdb4ab01251c : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d47c4bdc098c493e80647e1bbeb76514 : SUCCESS in 9m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/09fca901b3f64ddea39905de72cb0007 : SUCCESS in 4m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ffd57cac861d477b87eecd382cdc1e3a : FAILURE in 7m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3dda3baf078d440cbc4f7ac6f2818d7f : SUCCESS in 10m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6152ae1c1f514a7993e4ef22de93b1e5 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/408f4e86774245cab61f482326c203a8 : SUCCESS in 7m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6bec65e970b840ab8a4672a6e839c44d : SUCCESS in 35m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/931e99f4052a41f7a707895a87168e9f : SUCCESS in 22m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a0a3c758459b45a88b0f39738cac2b57 : FAILURE in 22m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/716218314396409e83be3c3f4aa1c320 : FAILURE in 21m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/89d90f5f59e247dfa26929674c9f0537 : SUCCESS in 20m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bd00c99fe3de4977a0a690e1158e1809 : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6efdd07d737743c4bb2a98f0c53d1a9a : SUCCESS in 17m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9d4532109dc440a96d6fb0062c056d4 : SUCCESS in 6m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f08e1f90610437fa83b49801c5fc318 : SUCCESS in 8m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/122e2f8441f5493a84c6e6096fb86296 : SUCCESS in 10m 34s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/876e7df0450c4efbac8cec6704ad13a1 : SUCCESS in 54m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58c9bf7226e847c1ab1c8e83da515ce2 : SUCCESS in 1h 07m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/562bdfe27ff144b28bae3d5b8774eee2 : SUCCESS in 1h 42m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b3b19b51254b487badc91cf552dda689 : FAILURE in 39m 20s","accounts_in_message":[],"_revision_number":5},{"id":"91d7bc08de8d603d043ed0a408bfe77a9d38d798","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 14:11:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d4d8f8d1696cd81e815240e42c5660f1066467d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:07:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018fa662fcbf41779a10e752fd8b9a70 : SUCCESS in 3m 51s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b475c3bad8424282829dd18d40a036b5 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d225fa4e01634558b74e4a0d51adae69 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d248f3d2821490a8b68523b1817e594 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/617b60be3db141ac8d950a83da8eafe8 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d03331ad3a4467ba8c4e288a70513dd : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b8283cad46843c3a3d539fc17fa64d6 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8595354c34e341a0b37d0e83006f5802 : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d74b39d9861443c6b8edc85fbced9a2d : SUCCESS in 7m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18fd90ebb7fd435a87963f4c428e965d : SUCCESS in 8m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/44b1436d06b64ce9bf54c32ce896478d : SUCCESS in 13m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff7662633d854a199e228a84f066ae11 : SUCCESS in 8m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/80b99dddf45041888ccbb794caebb73e : SUCCESS in 5m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d5ce3927e6c440ee88eba2e60825788e : SUCCESS in 5m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/66c293edd5724485ac59c4dde6d57303 : SUCCESS in 10m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f8696dbd9dad4ca6b9b88a5c9e2c08f3 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e30f076afdba4595b60d9a73227ff03f : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/df062809c95c49e9b0de11725a88dfb5 : SUCCESS in 25m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/16caf06eb9dc439aa9fa4186d7326f22 : SUCCESS in 29m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/57a5a583811f490aa23161f744824096 : FAILURE in 23m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c75ced50948418b99dede4ceae4c829 : FAILURE in 22m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5615217ac51c42a69dbfddc20bf43a72 : SUCCESS in 16m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3cfa0582cf14dbcbcea5dd760f7fad6 : SUCCESS in 12m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19938adbfd4c4c9ca0051dd65d555e9d : SUCCESS in 19m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1d1d44b9d7642e69837404bce74ee9f : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/05bc57232c9041e5a197e807b838984b : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef297e7f2fb7475783e7fc5ee36b21d3 : SUCCESS in 9m 20s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5d77ae8c1d554b97845d66de6ca87aab : SUCCESS in 1h 00m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed7d60428d3c41a5848e0935d082118b : SUCCESS in 1h 16m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b268e2149a054b509fd35c2b8d150759 : SUCCESS in 1h 50m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/13621558f7544ce7816815ff4c9ad8a4 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":5},{"id":"5a91ad10128af0423d9c09f9662b9dd0827aa245","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 19:00:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"12e24b4c7bb7a60b3336c7e030befb52a7beacac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:10:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55faf4cdb61a42cf9326ded31b4d5a1f : SUCCESS in 4m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/72b11709ce4942138fdb107e6e4493e1 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c88d615ea64c07a69adc0a8e05ed70 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5757748ca4e4cd7b4b1d878d42ea7bb : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/993ed4cdeb9c4f41be54a789c9d20a44 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5928bbcb6df4578a9af1ced20f7fa6a : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f75bab6956e4a98a91aa91b72792646 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5c733357e244ea8962c34dce1942818 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3e5840c388a3477db58ff8f7a4668501 : SUCCESS in 7m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71e102e6aa5a42d281846e8c104e0282 : SUCCESS in 6m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/57e8ae88d2df483080ac9a1fdd883e8a : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/06227463dc544bcaa2444cc9b686da8c : SUCCESS in 10m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9ddcfcfea21d47938da1957378012009 : SUCCESS in 5m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00dd382e00ee4c7bb83272215c886137 : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3fe4c0bcf81347c7aaeaece889305c82 : SUCCESS in 10m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8a5fed69bd34d4484b23dff2014de3f : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd19a10fe69b4b58a7652982f6d51c9f : SUCCESS in 10m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b0586a830e1f458097c7dd0375dc4663 : SUCCESS in 31m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/31e164e7ba90418cb9ed8a7d4d6a21a3 : SUCCESS in 26m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/33751e8bb9c444c99c8814c631776d93 : SUCCESS in 25m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bea60e9b6ca144df92108a268e7c61ac : SUCCESS in 24m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dbf5985c4b9e4e92adc1d8a9639a348e : SUCCESS in 13m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/07ed242f205e4ffba72ff0cdcbc97ce2 : SUCCESS in 12m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/49ac2840f2034086a65c2c3b45cbb7dd : SUCCESS in 10m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23b8b2b9e07044dc850645b9176b7081 : SUCCESS in 21m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7e6d797a8544f8e82898ece006a63bf : SUCCESS in 4m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/69dff3e0a8584400b72e89aa95e31bb8 : SUCCESS in 8m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12fa8c64e468452a9d0505e5932a0b8c : SUCCESS in 11m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/933fb716267b48fcafaa69e25f2b8770 : SUCCESS in 1h 14m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bf056aa1db140aab864ebe08f9382b7 : SUCCESS in 1h 21m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50a67fa143794940b88eb14666126f8c : SUCCESS in 1h 41m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ff473c465df405aa1f90ff114f5a952 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":6},{"id":"52bb3f781572d23e2b6fbc385ea7800be0b7760b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 22:26:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a449d5b62fca7389aa935e4621c9df54f73ddcbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:27:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"493f2e9b907acb6328ac9e39eef350205aa7eb4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 00:59:17.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5375220996284366abafb9648044bd03 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d68b8efd8ceb491e89a764238ab781b7 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6d1fd1bca9342c8b4482287831c9900 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1cf424fc5c34022bf8e482169b50011 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71094f6f1ad942a8ab970026ac6a6d84 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41e67ffc787b4fc99c1f22dd8236e26e : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1596c63b80424a7f909cc1e7e5ce8494 : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1b5411fc5ebc44258f14624451a85734 : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90e5c831697b4497b568868e89dcffba : SUCCESS in 6m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ef9de82d8eec4df99199852ef412b154 : SUCCESS in 9m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b043da95ff594b378e1c9e9c568a36e4 : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ff8d43d781844f24a5ae93fd606a472a : SUCCESS in 5m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb5892d349874156aa3eb21951318780 : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f120e759860b4875ab5cd3dc016dcf5f : SUCCESS in 12m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fc926d493bb140ef9dd5edeca6527147 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/817d378fa63e45759e4d5f1c1dc268af : SUCCESS in 7m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/800cedcdd28a4f6a98b39731e8d105c9 : SUCCESS in 28m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/735432ead2e542b18473202b7e34315d : TIMED_OUT in 41m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/677bf8abf4894b98a8761bab07110eed : SUCCESS in 28m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5de4acbf50f64b16a95347197e511d82 : SUCCESS in 24m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b2eb02f9ce904f908f39d655f545294f : SUCCESS in 17m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f9a33cdbc1034aa885dc87f1f171d535 : FAILURE in 26m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c1c6ba80834d4e5aaba44ad1d640f75e : SUCCESS in 15m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/076a960d5f1f44b587f8413f24502cd5 : SUCCESS in 21m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ebe258f0391b4830be5bcfc9f9d6a38b : SUCCESS in 4m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7de64938482a4cf6adcccfdaf0c55c1c : SUCCESS in 14m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b73dd4c10c184d86a702d78da9bbac2d : SUCCESS in 10m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2a580786fd7146e5b266815d3124c1e5 : SUCCESS in 58m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a9ecdf06f094c2d88962479bcfb78d8 : SUCCESS in 1h 11m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fd1a3ca4fb6c44e18e81d2c3f6d8a882 : SUCCESS in 1h 47m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b533850a0234cf084eeca89544f8b62 : FAILURE in 43m 52s","accounts_in_message":[],"_revision_number":6},{"id":"29418c569db47b90c017e081d36bcde7dab6a2a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-20 01:36:13.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0d3e9a5d9cedf2873eebe9966961abbe2a2d04f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 03:18:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da8e9028cb149a3b4826b5f9b5260e0 : SUCCESS in 3m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/18c4e79db12544e0a1cbc4a964f48b40 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c7ee1a54264e56be02272035507014 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/199ed9b086c347ab8f929ac25770f2c1 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64f4acf737bd410e956086bcd4714d58 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b82a9561928141bf9ec63ad4d977501f : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8930d90213044e8abcf3747781da3a7c : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d72a1f86f4214857b0a1da3450b612ed : SUCCESS in 4m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b3f26379c9ee4368869d71eff0e9bd86 : SUCCESS in 8m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95e9623adafe4af0aa3e32f9e373a852 : SUCCESS in 8m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/eda27ee56a3848fb85c165ff242c5b15 : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9abba3533fde4ab99284a3f0b1a4ecbd : SUCCESS in 9m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/acbe7f347d48494598aded0528b900ce : SUCCESS in 5m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/acdea319cc5844dda8df5414e3b73fdf : SUCCESS in 5m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8346ec66edc54a5584d246860967b45d : SUCCESS in 9m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8d5903f503f4829a3f2532b18d6df6e : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6e8a365812a46b0a8da511f6ea0f5f8 : SUCCESS in 7m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/62ab88252ecc42fcbef7b34a3c1cc9e2 : SUCCESS in 29m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d18ef4e17042424f8819f68a214469f2 : SUCCESS in 24m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bc446d7d26c845e3bf474c1c1b04966b : SUCCESS in 25m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/adb7c2021ee44507b459a18f28b82b94 : SUCCESS in 21m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d41e6204d364041a48a5e72acc61c0a : SUCCESS in 14m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2dedf4b6cc04a248d39ceaf733ae8ac : SUCCESS in 13m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3b16382ea43a493497f7bb592665f958 : SUCCESS in 13m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26cf356b15a24c22a28981dacb1a6958 : SUCCESS in 19m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d6aae4871e04c559c58c718abfb5ea9 : SUCCESS in 4m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e360015c14684b4fab1c5d94160be2c8 : SUCCESS in 8m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/baa524554fa24c1a92d0cd8ad1ab6a29 : SUCCESS in 8m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a3f76efc824c4b03941ff98470633bd1 : SUCCESS in 54m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/199c08183d1d4aa4a8e7110d14dee2ba : SUCCESS in 1h 12m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d9c31c75bf534f8987b05fab35fc0d3e : SUCCESS in 1h 40m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dda7f26cbd3841bb9d25eee29db23d37 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":6},{"id":"a6a7f7a934f051274db4862ed0d788a60189dff6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 03:18:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fe10bee07e9eb2ee1366865adecdad3301d3147b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 06:20:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d578f8822944ee9b66a64f232ac3cb : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19542de9fba94989a1b06aa157ec1465 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4404540d07fb4381ae129964df1bffa4 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6058c5cd153d4a37894c7317b3a41fe1 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32de6e35a60b45ed90e55019d08d99bd : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18a2682380984f8a91e490ea64604f3c : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66fe1d0a65b6491f8c28419ec9ab7826 : SUCCESS in 5m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f045c2a5f1ab4d929d13f88b245a954a : SUCCESS in 9m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be34c5fac30b4a1bb14089e0f8a99083 : SUCCESS in 6m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/11ccc533c9fb470b9f273daa9f2874c7 : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9b4c5196f1fb40f5be52ab656c41c953 : SUCCESS in 11m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ce08fe9cd39648bcbb06c4cd78a263f2 : SUCCESS in 5m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4ae649965b74ab3bef04837bb85418d : SUCCESS in 7m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/470aad74771c4fb989e52f9ca8a16e4d : SUCCESS in 10m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/845fdbfb31894b589ae5fa4ec2656500 : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d414eb7b434d41cda8eb03a230ca6aac : SUCCESS in 8m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7fc3492eee794739bdb1ec75c32b3502 : SUCCESS in 36m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/553895727e1b4d76b00d5203ed2e9948 : SUCCESS in 29m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/57a28623797243d18e1c84402b304eaf : SUCCESS in 29m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7fd32a58975c48de8b980cc92f7c02b6 : SUCCESS in 26m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fc153b7b8aee40aea8da42cd3625f493 : SUCCESS in 24m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/258eeb8e65b64576aa87f6e571406114 : SUCCESS in 14m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7b6aa97b60e44f3ba29fddcc0fdac26d : SUCCESS in 23m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2eb7c0238b49439daf596f4fdb23eae2 : SUCCESS in 16m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f5dc61a15c2452d8a2b838450029693 : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aa82e7f7a5a74ce58b3c037e5af7581c : SUCCESS in 14m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c4dc5fab89fe4ba09af7d5361d07f15b : SUCCESS in 9m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/41a6a47fc07e431fb92b7d998249bc46 : SUCCESS in 55m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac0e31c5b8cb496db6f580531fe02a02 : SUCCESS in 1h 18m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d5ebd897a32f4cdbabbf95b705d5557b : SUCCESS in 1h 28m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d57e9710c9fb4e199c9c1c257417aa5a : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":6},{"id":"701203a273a4cb34e82ce3eee5014144d0b6046e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 06:20:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"608a7d0a445c15b0f01f7421d9b95e735111d90e","revisions":{"62a8e56708efdc108aa454a14368529861233ead":{"kind":"REWORK","_number":1,"created":"2019-08-17 06:27:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/677052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/677052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/677052/1"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I79baeee85f93adb2adfd01b1f02ca511073c8d78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62a8e56708efdc108aa454a14368529861233ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62a8e56708efdc108aa454a14368529861233ead"}]},"branch":"refs/heads/master"},"971a261d1cf69d9b559a73b9a8aab748f7221581":{"kind":"REWORK","_number":2,"created":"2019-08-17 16:22:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/677052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/677052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/677052/2"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-17 16:22:24.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I79baeee85f93adb2adfd01b1f02ca511073c8d78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/971a261d1cf69d9b559a73b9a8aab748f7221581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/971a261d1cf69d9b559a73b9a8aab748f7221581"}]},"branch":"refs/heads/master"},"8c9508a607533bd3cac8d7da7f5265fce60c3655":{"kind":"REWORK","_number":3,"created":"2019-08-17 16:31:14.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/677052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/677052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/677052/3"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-17 16:30:05.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nHeld back the following\n\nlxml\u003d\u003d\u003d4.4.1       nova tests fail https://bugs.launchpad.net/nova/+bug/1838666\ntornado\u003d\u003d\u003d5.1.1    salt is cauing this, no eta on fix (same as the last year)\nwebsockify\u003d\u003d\u003d0.9.0 tempest test failing, ping nova\nstestr\u003d\u003d\u003d2.5.0     needs merged https://github.com/mtreinish/stestr/pull/265\njsonschema\u003d\u003d\u003d3.0.2 see https://review.opendev.org/649789\noauthlib\u003d\u003d\u003d3.1.0   keystone https://bugs.launchpad.net/keystone/+bug/1839393\n\nChange-Id: I79baeee85f93adb2adfd01b1f02ca511073c8d78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c9508a607533bd3cac8d7da7f5265fce60c3655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c9508a607533bd3cac8d7da7f5265fce60c3655"}]},"branch":"refs/heads/master"},"45ad4916e37bb026197484707b697f6dc6dd7f74":{"kind":"REWORK","_number":4,"created":"2019-08-17 16:32:30.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/677052/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/677052/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/677052/4"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-17 16:32:22.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nHeld back the following\n\nlxml\u003d\u003d\u003d4.4.1       nova tests fail https://bugs.launchpad.net/nova/+bug/1838666\nwebsockify\u003d\u003d\u003d0.9.0 tempest test failing, ping nova\nstestr\u003d\u003d\u003d2.5.0     needs merged https://github.com/mtreinish/stestr/pull/265\njsonschema\u003d\u003d\u003d3.0.2 see https://review.opendev.org/649789\noauthlib\u003d\u003d\u003d3.1.0   keystone https://bugs.launchpad.net/keystone/+bug/1839393\n\nChange-Id: I79baeee85f93adb2adfd01b1f02ca511073c8d78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45ad4916e37bb026197484707b697f6dc6dd7f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45ad4916e37bb026197484707b697f6dc6dd7f74"}]},"branch":"refs/heads/master"},"d92e5dfb210a4f15b414221a06f088d501075e1c":{"kind":"REWORK","_number":5,"created":"2019-08-17 16:37:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/677052/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/677052/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/677052/5"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-17 16:36:38.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nHeld back the following\n\nlxml\u003d\u003d\u003d4.4.1        nova tests fail https://bugs.launchpad.net/nova/+bug/1838666\nwebsockify\u003d\u003d\u003d0.9.0  tempest test failing, ping nova\ntornado\u003d\u003d\u003d5.1.1     salt is cauing this, no eta on fix (same as the last year)\nstestr\u003d\u003d\u003d2.5.0      needs merged https://github.com/mtreinish/stestr/pull/265\njsonschema\u003d\u003d\u003d3.0.2  see https://review.opendev.org/649789\noauthlib\u003d\u003d\u003d3.1.0    keystone https://bugs.launchpad.net/keystone/+bug/1839393\nkubernetes\u003d\u003d\u003d10.0.1 openshift PINS this, only kuryr-tempest-plugin deps on it\n\nChange-Id: I79baeee85f93adb2adfd01b1f02ca511073c8d78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d92e5dfb210a4f15b414221a06f088d501075e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d92e5dfb210a4f15b414221a06f088d501075e1c"}]},"branch":"refs/heads/master"},"608a7d0a445c15b0f01f7421d9b95e735111d90e":{"kind":"REWORK","_number":6,"created":"2019-08-19 19:00:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/677052/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/677052/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/677052/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/677052/6"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-17 06:27:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-19 18:59:48.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nHeld back the following\n\nlxml\u003d\u003d\u003d4.4.1        nova tests fail https://bugs.launchpad.net/nova/+bug/1838666\nwebsockify\u003d\u003d\u003d0.9.0  tempest test failing, ping nova\ntornado\u003d\u003d\u003d5.1.1     salt is cauing this, no eta on fix (same as the last year)\nstestr\u003d\u003d\u003d2.5.0      needs merged https://github.com/mtreinish/stestr/pull/265\njsonschema\u003d\u003d\u003d3.0.2  see https://review.opendev.org/649789\noauthlib\u003d\u003d\u003d3.1.0    keystone https://bugs.launchpad.net/keystone/+bug/1839393\nkubernetes\u003d\u003d\u003d10.0.1 openshift PINS this, only kuryr-tempest-plugin deps on it\ntenacity\u003d\u003d\u003d5.1.1    failing tests in neutron\n\nChange-Id: I79baeee85f93adb2adfd01b1f02ca511073c8d78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/608a7d0a445c15b0f01f7421d9b95e735111d90e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/608a7d0a445c15b0f01f7421d9b95e735111d90e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
