)]}'
{"id":"openstack%2Frequirements~677070","triplet_id":"openstack%2Frequirements~master~I113c06f6e09727aa279f5ab8d2b7534a019394c2","project":"openstack/requirements","branch":"master","topic":"blacklist-kombu-4.6.4","hashtags":[],"change_id":"I113c06f6e09727aa279f5ab8d2b7534a019394c2","subject":"Blacklist kombu-4.6.4","status":"MERGED","created":"2019-08-17 17:38:58.000000000","updated":"2019-08-19 05:31:36.000000000","submitted":"2019-08-19 05:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677070-1566192696008-de1e3c07","meta_rev_id":"2955f9e299ea6223533f03fd0e1c180b0b553e34","_number":677070,"virtual_id_number":677070,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-08-19 05:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-19 04:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-08-17 23:02:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2019-08-18 23:33:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-08-18 07:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-08-18 11:17:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-18 06:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"value":2,"date":"2019-08-18 23:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-19 04:05:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-17 23:02:59.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2019-08-18 06:09:31.000000000","updated_by":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"reviewer":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"state":"REVIEWER"},{"updated":"2019-08-18 07:09:44.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-08-18 11:17:38.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-08-18 23:33:18.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2019-08-18 23:33:29.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-08-19 04:05:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-19 05:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd04aadc09ade356cce25e8fcb080a28113dbf5d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-17 17:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21902982680359214dc025d390a170086f2f1148","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-17 18:16:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4f786d518d31f0ecb019bbf0b2cdb183f681e429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 19:54:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4faee382a9ce4418a7aa2fd874e1cced : SUCCESS in 4m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/166232cb3c60495c82bcfc3282e6076a : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7d3880b23a48ac9ffaa4eb2b77ab05 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/409d47b316ae46eba2be5027d63595e5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f987a24f0a7f4e0887d1b657a9d0be9d : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70c71345c7a04d5ab83c9153a9bf80c8 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9cb4e0993b64a2888de6db804e0d8fc : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b51c012763a44af89c7f7e415506cb01 : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d30412b97efd416c884ec3a580010849 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c62734457ba14f68b5c3378809a8402e : SUCCESS in 6m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6df537f886ba4530b261f3f76ce89aa5 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b1a529945dcb439db8579a9c5d3af729 : SUCCESS in 10m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/697fae12ea194ad0bac9e3ead9c72824 : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1e1bef0fa1d24bd1be54551a8669821b : SUCCESS in 4m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cddbed41b5dd43449bd5eb86db985399 : SUCCESS in 11m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/23b89732ec5c46b78b2532a48cf78146 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5c8382b71fd4a8e8b1deefb14af2bff : SUCCESS in 9m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e7776163023f419c8d7b0bb802f98adb : SUCCESS in 26m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e12746fad2e4c8da8121daa7308b583 : SUCCESS in 25m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9c58b6bf59d44d13bb4b7cff9cb3ce88 : SUCCESS in 28m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/74a758cc0e684d829893f34326337e1b : SUCCESS in 24m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9b76633cc2734583bde9cebc10d69b98 : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9a450180427c483cbbce82a95aeb28ed : SUCCESS in 10m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cadee18ebcba4a35b8739e8292dec66b : SUCCESS in 19m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/289a613beda344c7844ddd9757ad3da4 : SUCCESS in 4m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f47a198a23c349d28804dce0d6b0a854 : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/538289a8ed1c4df79b5c454eb7dbd7a9 : SUCCESS in 8m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0dfc300ec35246c78cdbc0455d4aaa3e : SUCCESS in 44m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/681433ee1b6d4a2a84d9ec4d6d0eb33b : SUCCESS in 1h 19m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/85c26c676be847d2bfafe86c956464db : SUCCESS in 1h 20m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/313ae69b7c434661b76739beb03a27a9 : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":2},{"id":"1be3c78b70a5135d2514ed7eafc3250f57c619bb","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-08-17 23:02:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"73f92791da174609bb732ae5061d449acc0af638","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-08-18 06:09:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e1df2cc3e8c859f6034a795acc49afc35e52d537","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-08-18 07:09:44.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"4c91f87b0419572e37323baa549f6db8d0d87813","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-18 11:17:38.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks. lgtm","accounts_in_message":[],"_revision_number":2},{"id":"474bf4e13636bffcd6c5164506a8a01dd84f7306","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2019-08-18 23:33:18.000000000","message":"Patch Set 2: Code-Review+2\n\nsee \n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-release/%23openstack-release.2019-08-17.log.html#t2019-08-17T17:45:16\n\ni see that a bug has been opened and commented on upstream, so WIP on a real fix.  I believe this addresses dhellmann\u0027s concerns in irc.  Only thing i might say is to comment above this explaining why","accounts_in_message":[],"_revision_number":2},{"id":"d15cfca2e5c9145678fe4e724579586fac9fef8a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-18 23:33:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4cd8d5e39bc38ab5efe16cf45596ec8ca15651f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 04:05:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e0c00563e2114f226bd0f16d47c4dbe944d09a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 04:05:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"721cc0d66cc99a8e0ea04f164af8681565ba896f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 05:31:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b54c8467b942a2bd625e5f7308eec9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb91534e97f4fb4bc5ee8e57978f6cf : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42898cfc280045f6a906e720dd0442dc : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9411d836d90495b99d3f64a26b585bc : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ceacff0b7644c088b64dcebb396cf8b : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb385c648f9a475b810baa16b41c9729 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/737ea4d2d4104bfd962778e44f641979 : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/28565f339f7c444286d7e74b8a5cf370 : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a9abd68bed04d42a00c083072f92fee : SUCCESS in 6m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c7cff2602f524bdabb029020b160a753 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d07b1def25f64f86a6740a07ee926f67 : SUCCESS in 9m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1e5265f4bf994cdfa38d908c29a330e7 : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/766966c601424e80aa8c2bd1a190ca55 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/040b5fbf0d56440f8e352f932c0e6293 : SUCCESS in 15m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ffbe03fa510d477b8b417fb98b06144c : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf8954201589461fb61da63bce45c2f3 : SUCCESS in 11m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/791befb90a1d49f1a8661df4b146dcbc : SUCCESS in 28m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/17f2f91692494afc8492b73242cbcb68 : SUCCESS in 28m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9e49b6a2c51b40599248ffbcca1db6f7 : SUCCESS in 22m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/38bc5248bba44eba91a4867bb06ba879 : SUCCESS in 21m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5e6ae88704104ff6b5b1a7b8126d3a1b : SUCCESS in 13m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/45941b597f6346cbb225c17b76957454 : SUCCESS in 14m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/501b5ae52ead462ca7d8d3063a6c1d16 : SUCCESS in 19m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e91f66a9d60c4ab1ab211c0b236c5164 : SUCCESS in 4m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18c5e6e35ed94945bcc0368941e3ce23 : SUCCESS in 11m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2af875170e594725a8009a07925c86b9 : SUCCESS in 10m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ed9f9c01afdc468da6c3d7f5cf0a6765 : SUCCESS in 52m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d4c05b6eb6b4e68b3199441069efc8e : SUCCESS in 1h 12m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/076f38beb5d548ffb1d8a6dc77bd6d9f : SUCCESS in 1h 25m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/79e6488f70eb40a38b03412fb39970cc : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":2},{"id":"4806690ab18226d75fc9305c605292fca37344a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 05:31:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a81aa355d78054c96568de284a34394e4056097f","revisions":{"e35bc1ad37b6ec289cf6a40c8502dae16fac529c":{"kind":"REWORK","_number":1,"created":"2019-08-17 17:38:58.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/70/677070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/677070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/677070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/677070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/677070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/677070/1"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-17 17:34:08.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-17 17:34:08.000000000","tz":-300},"subject":"Blacklist kombu-4.6.4","message":"Blacklist kombu-4.6.4\n\nCommit [1] bumped the upper constraint for kombu to 4.6.4. Per issues\n[2][3] this interacts poorly with eventlet and breaks many things\nopenstack on py37 with this error:\n\nTypeError: open: path should be string, bytes or os.PathLike, not _NormalAccessor\n\nIt is possible there will be a simple way to fix those many things, but\nin the meantime, this commit blacklists 4.6.4 to get the world moving\nagain.\n\nThanks to Daniel Leaberry for doing the legwork to identify the problem\n[4].\n\n[1] Id71496bcb5c8342851f287e35931431e8a443d23\n[2] https://github.com/eventlet/eventlet/issues/534\n[3] https://github.com/nameko/nameko/issues/655\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2019-August/008553.html\n\nChange-Id: I113c06f6e09727aa279f5ab8d2b7534a019394c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e35bc1ad37b6ec289cf6a40c8502dae16fac529c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e35bc1ad37b6ec289cf6a40c8502dae16fac529c"}]},"branch":"refs/heads/master"},"a81aa355d78054c96568de284a34394e4056097f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-17 18:16:08.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/70/677070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/677070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/677070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/677070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/677070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/677070/2"}}},"commit":{"parents":[{"commit":"44356da35b485f4325cfe41d49c52aa4543f29ac","subject":"Merge \"update constraint for osc-placement to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44356da35b485f4325cfe41d49c52aa4543f29ac"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-17 17:34:08.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-17 18:16:06.000000000","tz":0},"subject":"Blacklist kombu-4.6.4","message":"Blacklist kombu-4.6.4\n\nCommit [1] bumped the upper constraint for kombu to 4.6.4. Per issues\n[2][3] this interacts poorly with eventlet and breaks many things\nopenstack on py37 (including nova gate - see the linked bug) with this\nerror:\n\nTypeError: open: path should be string, bytes or os.PathLike, not _NormalAccessor\n\nI opened an issue against kombu [4].\n\nIt is possible there will be a simple way to fix those many things, but\nin the meantime, this commit blacklists 4.6.4 to get the world moving\nagain.\n\nThanks to Daniel Leaberry for doing the legwork to identify the problem\n[5].\n\nCloses-Bug: #1840551\n\n[1] Id71496bcb5c8342851f287e35931431e8a443d23\n[2] https://github.com/eventlet/eventlet/issues/534\n[3] https://github.com/nameko/nameko/issues/655\n[4] https://github.com/celery/kombu/issues/1082\n[5] http://lists.openstack.org/pipermail/openstack-discuss/2019-August/008553.html\n\nChange-Id: I113c06f6e09727aa279f5ab8d2b7534a019394c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a81aa355d78054c96568de284a34394e4056097f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a81aa355d78054c96568de284a34394e4056097f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
