)]}'
{"id":"openstack%2Frequirements~677235","triplet_id":"openstack%2Frequirements~master~Ib225076f859c10cbcbc7a636c2381a51feed307f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib225076f859c10cbcbc7a636c2381a51feed307f","subject":"update constraint for python-freezerclient to new release 2.2.0","status":"MERGED","created":"2019-08-19 15:16:28.000000000","updated":"2019-08-19 21:44:57.000000000","submitted":"2019-08-19 21:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677235-1566251097288-a49dced6","meta_rev_id":"d9c897bf99ba9b1e8474b8c9c3ba46d22ae84891","_number":677235,"virtual_id_number":677235,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-19 21:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-19 18:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-19 18:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 18:11:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-19 21:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3cbdaf5ac9374038b3302aacca956dba0f636ab","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-19 15:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e47f2fba56307deb549034557a7100b80432be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:46:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f817909eed4e5d949ae0fcaec01f87 : SUCCESS in 5m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a27fa32a011b41db9b7b6424f0285720 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4415e4f05f45e884700121947f682d : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14b70120d6984e6d9f56829cde2dc223 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75e8575565547fdb34b34ac8e57d009 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6212aedd1d504f0da3a989096e0dbfcb : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8034b3d2ced84fecbe86288d8dc7f8f2 : SUCCESS in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09d79eac83a14b8dbffa888991214c0f : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/86d99fc760f341e48a627bc930518091 : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5efd0fc06764673b595424369a9b9a9 : SUCCESS in 9m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cf150801573d470abe87e7f63541eb33 : SUCCESS in 12m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/58177b2ef49a4a27be1527ae3e4ab435 : SUCCESS in 10m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/75e3f87df7e54414ba89da2b58bcf03c : SUCCESS in 7m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c2590d7d6420477f9a9749341c80c994 : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e8393f7b27e84c3184d21cf01f34b8bd : SUCCESS in 11m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8c340c6429e487684d6263fd63681cf : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b7ba261cd6d4f21a8555d39f0134de2 : SUCCESS in 10m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ab39a2bccc2f4e618e12367a33d00bee : SUCCESS in 31m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3d5bb58acb94c60b9129c9a0b7fb0e8 : SUCCESS in 28m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5db4a4a0398b4b01a050bf6381c006dd : SUCCESS in 22m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce421fe9833f42af8305a5a6b21896d9 : SUCCESS in 21m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6aa1bd198bf6424c9475032869a99d6c : SUCCESS in 15m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8c17f13be25944c39a94bf4465d08378 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b555d61526ab4f4b85b990fea4024a75 : SUCCESS in 18m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18d1d85474564fb2b16ccaa5c07a44ea : SUCCESS in 4m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0fc5d9498faf4c6f914e10ae9415021b : SUCCESS in 7m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1e09eeeb6ad49eeb77b10dbf4463103 : SUCCESS in 10m 05s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/eec17a79d3a54a46b1ac3118e67830ac : SUCCESS in 1h 06m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1388d3bd49c046a38a8d8f7f2a66e1e7 : SUCCESS in 1h 09m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1b5e8b0b52c44338cf81144b8461ba1 : SUCCESS in 1h 34m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/979369fa27374cd987ea709efefd4224 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4b6b74dc177611317184c06609240eedbb248b45","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 18:11:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29a7b837832c2284b36e7dc128eb924ea55c1d93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 18:12:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f239b575477878878015db9453608657b20661d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:44:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e835bbfb7d04435791fcbddacd9de24f : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41060a566d7445ee91a76c16196a32bb : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f2c6535a4b54bfc81598220cd2de0d4 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1819971697384be3babfd120ed279483 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fcb2d325c6546e1876f44788c8b5685 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d51b1c4c6a664cc8aa102e3d0a99bf50 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a3545bc31904ccfbceabff07754d275 : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b6df42565b0f4b47b8d569848ad5d76d : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d67ba29c7c945d7bb3e66a0add61832 : SUCCESS in 6m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e50f1fd29827405d89a37901c50b1717 : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a1c5829121b14d1da8c8c1633935b5e0 : SUCCESS in 12m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/76ee6d3fe55c4cdfb48f67eed3ffa99b : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d3428aa16fc495399c582d25b308f07 : SUCCESS in 5m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2ce6de04bee5461c8f1e223f73f1e495 : SUCCESS in 12m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/13a8e6bb11d6489e9eca1eb8477a703a : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2232bfe78b7546a9be3e1351b4a17af4 : SUCCESS in 8m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/830ecb2f5e784cad907a6b180a7df3a4 : SUCCESS in 35m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/09e993dbe44d4ef3aa9f3f6863dfcdb3 : SUCCESS in 30m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a7fbd674327e4048b2e09287f4df6cec : SUCCESS in 23m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/859938f4938142488e209e8171ac0948 : SUCCESS in 29m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/540084c61fa54efeabbd53b5205eb589 : SUCCESS in 17m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/edd3b8933f30463fad5fae1fbd342f39 : SUCCESS in 18m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/28bc84b71bda429bbc2bb43b8a8fd4c2 : SUCCESS in 12m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80d36e8e6fdf4a18842e4b7e49302170 : SUCCESS in 19m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08e01765abc04f9eac80d37716132cea : SUCCESS in 5m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc26880522594a1b9a030e36505d9433 : SUCCESS in 10m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c4b92fd817f64a17b6f930777a8872f3 : SUCCESS in 10m 19s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d6dc1b9fbee24b3e9c59498112bb4995 : SUCCESS in 1h 17m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6956e314191f45388dd16b083bc85e09 : SUCCESS in 1h 15m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cdf886469b504d53b22d2bac5f650e49 : SUCCESS in 1h 49m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d66941596a8841f094405d8d652cb8b8 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":1},{"id":"18d66e09d193dfdf560e6048c0bb1671925f8dbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:44:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8be7dc5f0524458da9165016af1d407911d8fa3","revisions":{"c8be7dc5f0524458da9165016af1d407911d8fa3":{"kind":"REWORK","_number":1,"created":"2019-08-19 15:16:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/677235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/677235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/677235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/677235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/677235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/677235/1"}}},"commit":{"parents":[{"commit":"a81aa355d78054c96568de284a34394e4056097f","subject":"Blacklist kombu-4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a81aa355d78054c96568de284a34394e4056097f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 15:16:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 15:16:27.000000000","tz":0},"subject":"update constraint for python-freezerclient to new release 2.2.0","message":"update constraint for python-freezerclient to new release 2.2.0\n\nChange-Id: Ib225076f859c10cbcbc7a636c2381a51feed307f\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I57b24c372e6fe6fe7f2023225f9f08766f73740a\nmeta:release:Code-Review+1: gengchc2 \u003cgeng.changcai2@zte.com.cn\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8be7dc5f0524458da9165016af1d407911d8fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8be7dc5f0524458da9165016af1d407911d8fa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
