)]}'
{"id":"openstack%2Frequirements~677243","triplet_id":"openstack%2Frequirements~master~Ib1ed44f79b0c42346667039493f6b61303b83074","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib1ed44f79b0c42346667039493f6b61303b83074","subject":"update constraint for python-zunclient to new release 3.4.0","status":"MERGED","created":"2019-08-19 15:53:12.000000000","updated":"2019-08-20 17:58:42.000000000","submitted":"2019-08-20 17:57:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677243-1566323828323-462ff7de","meta_rev_id":"36f684cc57617e22468e9570c8ee5e28b0f7be27","_number":677243,"virtual_id_number":677243,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-20 17:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-19 22:26:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-19 22:26:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 22:26:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-20 17:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d63d849fdde8e08ec9586f8bafccc9bdfca1546b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-19 15:53:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fca03f8c76e013cbb68d1144e46bf8ca71137f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 18:25:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca2c4fa528b49838272432eb2b01d7d : SUCCESS in 3m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/87bda7c0b955460f8341f94c7d4c42b3 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264b3ad5cfa94d78a02ab2533223320f : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8fe00fa0bf640749c913f410a24c466 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/746fb56f54e7482eb27b30326b4370b6 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c36d31de94b947ef81758cbdc81ea1bc : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb05c173cf254d38a3a57f8061477a85 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92c5b2918cd1473d92f212d0284f7812 : SUCCESS in 3m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e000745cddc249a389cbba9417cf5a53 : SUCCESS in 21m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/174afd83e72c4bbdbc941aa15107615f : SUCCESS in 9m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a82d90df653496c96982c74f1522844 : SUCCESS in 11m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ddca02253d2c4a6fb633c25a081e4279 : SUCCESS in 8m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/db42b5f9327f473fa4bbc961a79ff4e0 : SUCCESS in 6m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68deb6b6dc864b1d8a71f47c62cbc910 : FAILURE in 6m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/107efa7f9415438e880a68733a6e5683 : SUCCESS in 10m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6c673cad48d94f2eb7691376ecb5e562 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0f9b4c21b804fdeb1cbb25a75ca0c49 : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0c63d219f53f4cc9bc77493fad810c1d : TIMED_OUT in 41m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/019da594baf74c789d46150aace0aa31 : TIMED_OUT in 42m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/92d4fca44963452a98c9fff23f59d461 : SUCCESS in 23m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc4077040a854853a17c1f5713b36f37 : SUCCESS in 29m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/65a2f0e6f0084bc1a063ed628e373e02 : SUCCESS in 18m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f202ab85630e43aaae7934a2475a20a4 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3f772338f7e4515b0a29c99347f4f91 : SUCCESS in 21m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2dea1e5b2294426b652b1e1e33ad39f : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/63dc2ac4e6c94001acaa18b2a3fd63b5 : SUCCESS in 10m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5230098736bf4b48abc66bc907806039 : SUCCESS in 9m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/89895a7dca764ff7930a73e1c17c9a31 : SUCCESS in 54m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddd7af19f1144d06ba87bf75528478da : SUCCESS in 1h 15m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/451fa5602ac4450cbadd99d9fdd11886 : SUCCESS in 1h 48m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6b48b4117264c64b2af53e2d4cc2030 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"375f5edd78f305e0c7972202b719a5f6aab7ff4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 18:39:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e04f36d06e939ab1fdd9a2561cf92dce43ed0762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:13:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea9240c554447cabe88908972c7c233 : SUCCESS in 4m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2ecf65c925be45309faf1bd06d402b1f : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec9c40294394368a4f76a6adaa2aa15 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a92ef5e50534cc6acd57ce8d67dff75 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc4f75e58d744008b1e070bdad320b24 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07735195d45a4383bd2f07a0e64d410b : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5f7bf4e73a3449dbe016a09694299e1 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fccb9db12d1843a586e89f9eefa6f023 : SUCCESS in 5m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6dfaee973ab74725baa56a9ba7d3b60f : SUCCESS in 11m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8cd65a55378a4ddf9cfdf7ea84693246 : SUCCESS in 6m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/88287f39f325433cacdd85ad08338d9a : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9942b63cf48a4455af1f4169a58cf2be : SUCCESS in 8m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6540e84a471a4e3f94d92536a11c71dc : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b60678e5397a46d1b9264a7530e89385 : SUCCESS in 7m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/31ed88a3984447d5bd36a80893fec967 : SUCCESS in 11m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eb686260d2e64452b04f8eceb0b1361b : SUCCESS in 8m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc5c51b5827b46978359195166a598c9 : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/96288caaf577438cbed22c141c53f92f : SUCCESS in 31m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/71da23616e7a41b0b3e2c3e12413f049 : SUCCESS in 26m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c07f10ae477c400c918330ff35d97bca : SUCCESS in 23m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7aaaa4a659f24a3aba6bb794142138db : SUCCESS in 22m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b868831859db4479b067a1c43e4941d7 : SUCCESS in 14m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bd2de23c875f4e969ff57c3b8d0a2b12 : SUCCESS in 12m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bd34188d971b4c39b5796ed32c62967c : SUCCESS in 19m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d1be3f375f847e9869ca97475a6a1b6 : SUCCESS in 20m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6075a59356fa4dd38c4e573237475d56 : SUCCESS in 4m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6de96527351841d5921a22d8060b7a2d : SUCCESS in 9m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e97804324ce44c99b0c073e7338f4cc0 : SUCCESS in 8m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/653a05ba5a0b4cb9abe107bf61829ed3 : SUCCESS in 1h 00m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4853ec82efef4fd994fe8b381f4eb4ff : SUCCESS in 1h 16m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f848a3c8e12c44128f76f16cc6106a24 : SUCCESS in 1h 21m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b7d8f7867c34081824f747cb105082c : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8311090cd442f469393a9d6d11be6097d0bde285","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-19 22:26:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4b8a44a06685ab793fa9d7b1b12c01c303cce8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:27:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5aa3e5853b27dfef7af87baa5ce095486a116b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 03:24:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c20be6c7d046e582baf724291dcbab : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d3bcb4bf9d4a73bc7dd5ef18bfd308 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d52ff9e3074f4f2b8a2a519055bd6bf4 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f4cbb6307bc4612a7d81a4951982d31 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6707bddf3826494ba352445b5085895f : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e00dcf3d0d924db788dc9b7c425e4483 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd3c1623429d41388ad9d33fe03c45a3 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a5f2903d06804410801e70155816fb2b : SUCCESS in 8m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bcd70121567a46c09d8f0f6406c5340b : SUCCESS in 7m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/742ae9f1a0634d56800afb1b045927c0 : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c22cc729ddc64a25977a2d267819054d : SUCCESS in 9m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/98c9a26687fa4c58a5a604082577b52b : SUCCESS in 5m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec698cbea5844f338e5cdc7aac611c13 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fc353423c03d439699dbd800715cb575 : SUCCESS in 10m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b3dfa73e3ff488d99b24a5744c10f93 : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/951c8b1c00474d5aa33454adaf847374 : SUCCESS in 8m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/65d3e29497c74e0abe81b1246fcbb51c : SUCCESS in 31m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/369978bd49cb47e5aae1a6fff26774f4 : SUCCESS in 30m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3dd1039486934dd0ba2731e1619f89c0 : SUCCESS in 21m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e512b7b31cd441ed9dac30b42fcbff9f : SUCCESS in 23m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5ccb3e7b6e134cfcaa30be17b1b7eb1b : SUCCESS in 17m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/209981b1502e42ecb3e5157be5a3854e : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d30f3ab56d14ad6b84a9f562fba9eab : SUCCESS in 16m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43f2c6fe9a9c494fa65925db2cec9238 : SUCCESS in 19m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fcb7f00dd6747d7938cb49270f5cf4c : SUCCESS in 5m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef5c8c72b4f9443dadeec90a39a1f1ba : SUCCESS in 8m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7da043adda974bf79dc3d140bf5bbdcd : SUCCESS in 10m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1c0200460a4c49d8999f3e05326aab65 : FAILURE in 37m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78fd5c59c7de4f94a1093ff7b18a462d : SUCCESS in 1h 08m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/600bcd4f32de4c09884473aeeb74e73a : SUCCESS in 1h 39m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2a1be656d18f4bda975217b63f2974e1 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7e62183ed9c8e665c8136d56a6b0ceb346f2d9c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-20 03:32:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d56086464270eacef89186c5c4793e836d47889e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce0f159beb042549572e4299c562344 : SUCCESS in 3m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bba8ccfd030b477d85464612b469d5ea : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a4c98d13f5437488d77fe1926ebf2e : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5d53d2afc53432793827e6a682650bb : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3596d08f1d6a4e0a8856baac0c76f07a : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffebc6e8086444d99b6cdb2fb703f50d : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8bec27933194e4a803928cc6e0fd440 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/117d86fb26c0452c9173d2bd39c38fe0 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a1d2633509d440039c80de0bc639acdc : SUCCESS in 9m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e8fdd3ce66142769423f3508d786e6b : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ffcd7d5bf4c14e768083a071233b7a63 : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e42d61b99534ef69192da74d0b678d8 : SUCCESS in 8m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e9f3c5ae07044ebaa2e3dc93dff228f6 : SUCCESS in 4m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a890c46309f64bcdae3fe48edec62c1c : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bf4d4dd567d944aeae65c257a1731f76 : SUCCESS in 11m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b4dcc8882b34ffdb37f0f84b5b6bbfa : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6b1161de22c411ea021dbecdaf1abc7 : SUCCESS in 8m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/70186a82c3ba4089ac578c9aaf6aca56 : SUCCESS in 32m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d9cfdb57e4f045669bd1fd54a2208a2f : SUCCESS in 25m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2c2db5fd6bb24f8da3c537addb4bfa3a : SUCCESS in 23m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e57d5f703b83403bab41d3631faf6aba : SUCCESS in 23m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6b69e6ae101f49949e104846fd59a7e3 : SUCCESS in 12m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f36f36d0d6ee4f1e90750f151e0ac30f : SUCCESS in 14m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/852944593f2d4b63b7541a484cbc8412 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/951ebac112684585ab8c9719069d84ce : SUCCESS in 21m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af7ef2374afa4ea5b15186228aa07048 : SUCCESS in 4m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3dfbfe720e76473dae28edbd87d0c92c : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7e030c947b04fd39e4e56e145d92744 : SUCCESS in 8m 05s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b3c9b08355344d6ab1108995586ab2c8 : SUCCESS in 54m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f74329834cd4014b3ee9e527b7f0acc : SUCCESS in 1h 16m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/495cdd67f26245a289daf09165bc7b2e : SUCCESS in 1h 41m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f1498aa108d24b31b2e09d51152bbbc7 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1be3c335e1a9c64c5e91741d943434d27357ed6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a6a3c188e1b0ae4b64cf58d97a3bcea1ea6d8c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 07:03:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328e921b1e8a47bd916996330c4763bd : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc33e5ac45af4275afa69213f6c54c49 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbdcbb019df149f7a5f419741f369477 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da82d4a7e7aa42d1a2e534034d02264c : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d4a778bc22a4d5ab679b9599cf82e23 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2aa5bc737e39488c8789f0962e33469a : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2dbcbfa7a9af4ea39089f766a59c9a7d : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c6735e5222f344a4ba3fc0071f9cfd17 : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7ab2e467f334a86a1aa58c21f9050ec : SUCCESS in 6m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ee88abd9d34f449080f9a42df7c0f43e : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3980feb09ee4144a22b2fce93722b59 : SUCCESS in 9m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3ed66aadfc8741e7b29e1a7aa8a200a3 : SUCCESS in 4m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f2e5520bf71145b1b33ef83b2e3fae88 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8fef6ae01539416982b328b1eb21e65c : SUCCESS in 11m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/544c7062e6aa474ea1272acaa4c9713a : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/478918d7727949ae862955dfd9566950 : SUCCESS in 8m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/765afeef783241c6904d928786f32b81 : SUCCESS in 23m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1597acb7505c42d993d95a32835b853e : SUCCESS in 28m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5badaec2cd7c4fcdb01096adb16579a1 : SUCCESS in 23m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dfa961e9b6f84574b08a45c54f6c2fde : SUCCESS in 22m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/414427edff4746b38d3aabf0233c474a : SUCCESS in 16m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c471295faa554d10ab976aa7cc17dc09 : SUCCESS in 12m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8c967c51cfca46468642549ad11a798c : SUCCESS in 11m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cef376bd252c4fdfaa4bf53f26755b78 : SUCCESS in 21m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c05d0cdc4a6467a808b6e9d51f0805d : SUCCESS in 5m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2add05fb7c464361b889b6937774f77f : SUCCESS in 9m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a01aa1d38514812b90f4dcd5780fdef : SUCCESS in 7m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6863518bf3a447978a6a6ac7f997b224 : SUCCESS in 55m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e35d18bc3b004ef9bd6ec859bc05e147 : SUCCESS in 56m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/389b1fe500bd400bbe2e183722234782 : FAILURE in 1h 37m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/48eecddaf29249c39b81d46173a4fd0c : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"28170404ec521851b77a8e53294e65e4bc97e669","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-20 14:32:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2760bd870f851c675503153d19a2ff5c06fe3ad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:11:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29c49ffeb7d41ad8012200e97c425f9 : SUCCESS in 4m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ec749cbd1898474da8ff4a91beb24f90 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3c5ee94f454e628d51b1b5558b2dfc : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26d6c9e87bc44fb29a0188985ead180d : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2062a73b070f44b2b952b7fca3d2511c : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/760489353bc349ee81cee80acade288a : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/561b5547296a4280919d3b52ea35f31c : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a23fa65c2ea4c979c9271aa7f37737e : SUCCESS in 3m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ad5cf94bc3a74c61b85df444a9d1b6fd : SUCCESS in 15m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4952b085ab5848dda3be8d9c447f80e4 : SUCCESS in 15m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f5312a3d755e449c88886c96aee7cde0 : SUCCESS in 10m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc8508da500743db81cb448856ff0e86 : SUCCESS in 9m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cc5e067a4570449eb042d9cce95d1f2c : SUCCESS in 5m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67d30654b22541febc112f1730f7bb77 : SUCCESS in 6m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7975576bfe8f4dd288a0a5a9680bb74c : SUCCESS in 13m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cd854728f63d4e5eb58a31009943c324 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddc674ac4b33467bb686f5e5a0f910fe : SUCCESS in 6m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/452f44b7f88f4f97b8ed7d04579d1e68 : SUCCESS in 31m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/11d62442ac104af984e0c43722e5d767 : SUCCESS in 30m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ea3331cbbbdb414997b4764059b024d5 : SUCCESS in 26m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e53958ac358e48dbaaec6177519d02d6 : SUCCESS in 23m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9f0c402edc3d4aa493748a85dfdc872d : SUCCESS in 14m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8ea85feb4cb44236a50a56d566321d07 : SUCCESS in 14m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/06f8a48ac3e146a89e126f784f165517 : SUCCESS in 12m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a581deda8364019b68fc74181f2e05a : SUCCESS in 20m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5feff9220cd4cffa4a70717f415d168 : SUCCESS in 7m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0e1920a2b0564e4498a610e155db44f3 : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b6270dc33c3427eabf698f0dd00204b : SUCCESS in 10m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3a8a7abfdcf746cab0fd67a25ed30c20 : SUCCESS in 56m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3b5758621f646d2b161212f62d86d30 : SUCCESS in 1h 09m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/92cc28b4861042cda572b3256495da11 : SUCCESS in 1h 31m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/53a0d414f00349139364b525b623fac8 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0acc0e818aa83fc8b6fb65c177886cb8a18531ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:11:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26a4de18bad5dfdcfa4eaa6f12c0c02ffea64a27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:57:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5f034c16feb229b7966fb12b6cc51af507836187","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:57:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915305f7034b401e836c9bbc2ea630ee : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd732fc570f4814bb8d2e3c5b692b69 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2e39ea7e90e4b0a82fe2684d6ba5005 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f192cc02bd42424f855c2cfbc786c200 : SUCCESS in 8m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a77a2a381de24253ab88d66f198520a7 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0b56ab5a6224024a177094ab1f7b170 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d1fb574edf848cb84ee61d7cc6b8a48 : SUCCESS in 6m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/78ed03f28b584d4c9209ac30d7fce165 : SUCCESS in 7m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65e7613ee9d243869f7fe2863b5d5cfc : SUCCESS in 9m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/74b4c07407764f41b9a05f77e1a51ba3 : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/44b16218e8984a6b8d539487f2f5bcfc : SUCCESS in 10m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4efa9705101845458898bf7221be87c8 : SUCCESS in 5m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f8aa7aee78641e38815f8f548ba9054 : SUCCESS in 6m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2cae692514a34533b42d23c50186885d : SUCCESS in 9m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4dc11962be464482965da5202b743047 : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65b1d710613444a2aebf8db1664f02ca : SUCCESS in 8m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a8ee834055ad4cfc8768a07bb0914896 : SUCCESS in 28m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e29cb1afb3604a1f81dfaca0b917e861 : SUCCESS in 30m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/68fd12f8c2f0467e8f52380256c1ce4e : SUCCESS in 25m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bb4000cc99c4aa89c020e5aba4cf232 : SUCCESS in 22m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0a89049291574b44ac197ee97ed5cbf2 : SUCCESS in 16m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0848a8248366482597db5dcc10b97418 : SUCCESS in 13m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/87ed18b17c9f49e9bd01fe5a2183077e : SUCCESS in 16m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24df5657df684090a978cc5927cf2ed3 : SUCCESS in 20m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdb4a5fd79a04eb2ae37b402187f6ad1 : SUCCESS in 7m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4829cc82e8694fed8227855c2eced53c : SUCCESS in 10m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f784fa31e144c37b651d6e4dfd8fbb1 : SUCCESS in 9m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a15f0722521241f6a1d3fac2024fce58 : SUCCESS in 1h 00m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51ecf9c3d9ea41f498b33f02f208ee41 : SUCCESS in 1h 16m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/996fbb25e4d54dac9c659d530f221b0e : SUCCESS in 1h 35m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c792e98e33714705a9f90de3bdfdb219 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2738cfde221f12158bd9b13a957f45c8e1b082c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:58:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36cb03f0b9c4f72a93e6175c6d803ef : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e3d66ee6471afd42a36f77707526900c1551959","revisions":{"3e3d66ee6471afd42a36f77707526900c1551959":{"kind":"REWORK","_number":1,"created":"2019-08-19 15:53:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/677243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/677243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/677243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/677243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/677243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/677243/1"}}},"commit":{"parents":[{"commit":"a81aa355d78054c96568de284a34394e4056097f","subject":"Blacklist kombu-4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a81aa355d78054c96568de284a34394e4056097f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 15:53:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 15:53:09.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 3.4.0","message":"update constraint for python-zunclient to new release 3.4.0\n\nChange-Id: Ib1ed44f79b0c42346667039493f6b61303b83074\nmeta:version: 3.4.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: Ib098cc2cfad404f9ec0f60efd9e223e83067522d\nmeta:release:Code-Review+1: feng.shengqin \u003cfeng.shengqin@zte.com.cn\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Hongbin Lu \u003chongbin.lu@huawei.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e3d66ee6471afd42a36f77707526900c1551959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e3d66ee6471afd42a36f77707526900c1551959"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
