)]}'
{"id":"openstack%2Frequirements~677479","triplet_id":"openstack%2Frequirements~master~I3853803780197c419cde0c87447f795edc6d7382","project":"openstack/requirements","branch":"master","topic":"DNM-websockify","hashtags":[],"change_id":"I3853803780197c419cde0c87447f795edc6d7382","subject":"Update to websockify-0.9.0","status":"MERGED","created":"2019-08-20 14:57:53.000000000","updated":"2019-08-23 21:40:42.000000000","submitted":"2019-08-23 21:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677479-1566596316647-b3de4fea","meta_rev_id":"0c1851d61dc1c9157083c8915be3e68be3d3b0e6","_number":677479,"virtual_id_number":677479,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-23 21:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-23 19:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2019-08-23 19:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-23 19:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 02:29:11.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-08-23 19:36:12.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-23 21:38:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"723f3fffee45465f0a4c22ad0156c05781cda456","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-20 14:57:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8efa0142246c0d80af2391ed3bff09035f0133d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-20 15:13:02.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"934d7285c5ae54bca533fb62dbc10d33020cb185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:54:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8619b8b71bd04374a35afedf6fccd90d : SUCCESS in 4m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6b3d38f6c4c142708b66b1c146db1abf : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f1511052584b71ba5b3ccb60412d9b : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbda48c1dd0e4c21bd26cff9273dbb64 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/644cf26081064e558859c3f12505f89b : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f03948a641940d097019dc0bbb9286b : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/320a017698fb4448ad83d09651c2eaac : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96bc173967d64420b7cff6c4d9649c3c : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/61335c4e42ef4609b235845c326f2eae : SUCCESS in 9m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0aabdc96f40042c586c915e79d4c5a3f : SUCCESS in 13m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/168a5827f0fc4e1096e84d2118dff966 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f06292f88919494483ba3f3b3a6b1cdf : SUCCESS in 8m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f34107c0cf3f4dfc82a251ddbc707f17 : SUCCESS in 6m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/78ede4d6015142b7a4064be6720a2b32 : SUCCESS in 4m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e80b22305eb143c6b22a2928f0648a69 : SUCCESS in 11m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c10a1e1cf4c342e8b34e1982d786554d : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da1d19765b4440ed87e3ace5c5bb6a24 : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6e5c1835fad24a5099b2471dbcd31b51 : SUCCESS in 22m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d9b6937b7eb40d09728c9dc1bbb30b3 : SUCCESS in 27m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/97ecb1a3a5e348b5b04d7fc6a56ac12b : SUCCESS in 25m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/32edeba7f53340df8dcb92b3b5cdca4f : SUCCESS in 21m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ce1ac01b5f224368b2c3524eb75c7d63 : SUCCESS in 13m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ba36cdc34c014a72ae962957a35fb335 : SUCCESS in 12m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0c7024d953e3426e99ea73f0545f7bf2 : SUCCESS in 14m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7965114f32f46eeb3d4269713d16e85 : SUCCESS in 19m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdbda45b0c3e46b381408675cfce4d39 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d88524c7d144e0491cdac48391daf8a : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/574fff6d599a441a8557e00e26cf911f : SUCCESS in 8m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/45126a63bd7b4bf8aa12cd6b1f767594 : SUCCESS in 1h 03m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/37e8348d92dc44bfa960e63b3da2e159 : SUCCESS in 1h 01m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d1eb8b1ffc484ae580b8e4b75979782c : FAILURE in 1h 50m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c28a4a9614944aee922d7e547f636018 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"61f4f19a2310a9f519cd4c84666570bef07e3af7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-20 17:52:53.000000000","message":"Patch Set 1:\n\nThe failure [1] is happening here [2], where we are scraping the textual response payload for a particular string (\"server: websockify\"/i from what I can tell).\n\n[1] https://object-storage-ca-ymq-1.vexxhost.net/v1/86bbbcfa8ad043109d2d7af530225c72/logs_79/677479/1/check/tempest-full/d1eb8b1/testr_results.html.gz\n[2] https://opendev.org/openstack/tempest/src/branch/master/tempest/api/compute/servers/test_novnc.py#L167-L177","accounts_in_message":[],"_revision_number":1},{"id":"e3b88e69a58d61e527d60a8e96ed3501d4f4786f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 18:42:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a94144afbdcd898aa94f1bacd06985c107bb1fc8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 18:42:17.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"22d2cd1f3506d29530043d496c963ebf59778c83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 23:25:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/894fe113e21d426c8d80ab753cad7b3b : SUCCESS in 3m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7b345a9daeec4be8afd95b4bd4ad7ef6 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3ad707808c403abf28c4455c62b97b : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13c42d166be8486fa32dd343690ca6fe : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8de452734c44db6926cd23c4ab45505 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a09a742ed1a4cbf8c394673b62de749 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/209b0ee67df3433cbb5ca44571af3825 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd43533f6c624e31b4cf05dd7b41cac0 : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7002a4a650be47dbb0cedc7497364c8d : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c527e86fcf27467daf7e5132bc53ff44 : SUCCESS in 8m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1461382e01e24cfaa1ce5ada69817a49 : SUCCESS in 10m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18af94be81c34f04b5428b58a9f4a0ef : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/63d0c12a80a44bf98c92a759b0cec2f5 : SUCCESS in 4m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/15c9105da4bb49b0906210cdd20999bc : SUCCESS in 4m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1f608ccaa7894cf2b8363ada801dd123 : SUCCESS in 12m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/70a9d938d2284f7c8d2af55dacb9e599 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e39ca2af4ee4c49abe1a65e9bc1e459 : SUCCESS in 8m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/183ef97e145d4a928cb9ea1bd18d4ff0 : TIMED_OUT in 41m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c5c0793c9874ead865c949117bb70a8 : SUCCESS in 32m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/34ca6a5e7e384da7bd83e25c1bf242ca : SUCCESS in 23m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02dccbd74d1649c1872d69746cec5846 : SUCCESS in 25m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8533f23010de48cc9842eaf15c601adc : SUCCESS in 12m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28cffc9a5bb4425d9fddd904dace207d : SUCCESS in 13m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8492a344323c455c88efb245437dc6f9 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b758dc5c162a4a728bf7887e24ed19e1 : SUCCESS in 19m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d5864f99b9b43cb8af5da25f4c50f1f : SUCCESS in 4m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f44bb38d73934bb8936fd902a1d75860 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/01936841bbb444fe977a65d139c099dd : SUCCESS in 8m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/28854364e5d64052a8b7a1834e80180d : SUCCESS in 56m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e5687f458984793ba5a650b3f6701a0 : SUCCESS in 1h 18m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d27d0a35d3a4a449a03f9b730921caa : FAILURE in 1h 24m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/50120681ffe246b9b3a82d63ab9e2cb9 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0372962912b81d6cfbda71a5fdafd72d91c54a93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 23:41:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"58c4b8501014b2c005b542c3735458230bc22cba","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-22 23:50:40.000000000","message":"Patch Set 2:\n\nNo point rechecking this until the tempest fix merges.","accounts_in_message":[],"_revision_number":2},{"id":"79f671bb8704edebb9c963eaefb3783fd0be967b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:29:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e781077d5eb4219978d7f02457d4b5c : SUCCESS in 3m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/71ef4466d2764e51a092fe542df3d41d : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc48b7d0f23469ab5aa45646747bc4c : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fc174e705f84ca6975c5d822e103f3a : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c179224d364a508ebb5b2e37d8120c : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45c85ad6b53949b48d2b4007d2fbfeaf : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aab28ec4c15a4d9fa98e4bb1d1aa9696 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/284a0c8b26344393a9f7e07b84faff2e : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5c1d3328e7d941f29f8001f5273b6398 : SUCCESS in 7m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/232babc5d1fc47f7b781e86ed7e8947a : SUCCESS in 7m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/119c0e6cf407432eb1790cdeb0c5f3af : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c91eccb40ca5446c813dc1583ea675cb : SUCCESS in 9m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f27a1da291124283ba71714bdc1dbe79 : SUCCESS in 5m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/77e1c95ebe1d44a284ab8d590ac4c968 : SUCCESS in 5m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1508fa7b7f86438cb7ea307b794f9d38 : SUCCESS in 11m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/165607cc192a45ab940fea40eab711a7 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba9956cf1b434ef68b7d254b7e439e0a : SUCCESS in 6m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e86914fe754d48909418083b9e0ab140 : SUCCESS in 27m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/46ea2e3103be441eb3b109bf3854fe7c : SUCCESS in 24m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/16ce289609554a70af16f644987ec13a : SUCCESS in 24m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c6120786ec14687bec615b663263460 : SUCCESS in 23m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b0b21034caaf41a4aeddb6238861a872 : SUCCESS in 13m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ef35005f4ba4045bdef7577deb287a6 : SUCCESS in 12m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a3a9a894a5e49c09c72b71f4e79d66a : SUCCESS in 12m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d2f79e5bf6c497693310458ccde0618 : SUCCESS in 19m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35c2a001a7b144cc91727d36356c387e : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e84def944abd4b158d9fa78f0eb8369f : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/851e7cce2f424cfdac841d7ac4288f7b : SUCCESS in 8m 34s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/409665baf0f245c2a72c7410c21a4adc : SUCCESS in 54m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8baf2fa174c346d396710a9f0ae78d29 : SUCCESS in 1h 09m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/01177019131e41109a18f42bef764b6b : FAILURE in 1h 29m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b2d7267e75ca4721bbdcbe893f7d5136 : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":2},{"id":"139b8f89a946cdf5bdc14627bbee50d1ffa108bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 02:08:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"440b3f39f134b765b5a1a5eda24e17725edae043","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 02:09:09.000000000","message":"Patch Set 2:\n\ntempest, not nova, ah well...","accounts_in_message":[],"_revision_number":2},{"id":"7e7240a8b146b4d8a79fd1c4a9249d6b5a3c58b3","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-08-23 02:29:11.000000000","message":"Patch Set 2:\n\n\u003e tempest, not nova, ah well...\n\nUnfortunately we need two fixes to get this working, one in nova (merged) [1] and one in tempest (needs core review) [2]. I pinged gmann in #openstack-qa for the latter.\n\n[1] https://review.opendev.org/677856\n[2] https://review.opendev.org/674364","accounts_in_message":[],"_revision_number":2},{"id":"eef5dbfc2d6eba7139b1c36b6c7f9bc3c618043b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 03:41:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea8ee6ba55f46d6968e04d29d221019 : SUCCESS in 3m 23s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/391caced934145688399c860aa5cbe7f : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4714486dfecf409d93283b337d4bbb51 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96c350e27e1a48af94a09221d26431a0 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4372dfc558bc4e66838a57b0d21ea391 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4086d09c99641d9afba23b26ea632cd : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4fa39a552d34b6b8c9d56d959b1c950 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0da479495504df4b24c3a6759d7e463 : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/335babbdcedf45729753d96f2a1cbabe : SUCCESS in 12m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/897e84695ff4451a960bb7b3ad2a6d91 : SUCCESS in 8m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/92bad118f7744f6f9a6d636c0dda1e1b : SUCCESS in 14m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a735a6d8f22c49fe8385dbeca543b8a3 : SUCCESS in 13m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5612e8f0bed04959aa2e771d9adef51f : SUCCESS in 4m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/475038c52d8547b8bb169692cfca293b : SUCCESS in 8m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/86de7cd5e926479abd6eeda8172153a2 : SUCCESS in 10m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4a47fd36ed8431487547a21a69a8208 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba2022c4a68a4eef9965e6a655764610 : SUCCESS in 9m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/07de73b47ae84404b4f2b4ce4f0f2c04 : SUCCESS in 36m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac6bfcfc2e4746988c39db4528842265 : SUCCESS in 28m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fb767650d6534d17a3856860cebc2994 : SUCCESS in 23m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/62d4f54afb4446cf9304ae5e18cf98ae : SUCCESS in 21m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e8c7f4bf91f240c68ca58eaac8e9511e : SUCCESS in 17m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1aff7ec48aad4632b9e56390bc529754 : SUCCESS in 13m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ab52ae0620d04e7292b5dee842d7ff69 : SUCCESS in 15m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1955deadb9b1414d950ae370fba12f27 : SUCCESS in 25m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65f90e8983ea4c05af789bc54fb41a30 : SUCCESS in 8m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ca8bbf2bb83543d79e0c6cc728f57713 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d9020e2bbfce4c5199e0dc92d06219c6 : SUCCESS in 12m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/56a0a5fdece9483d866e49bd39152eb5 : SUCCESS in 54m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2115df83d0f04c4fbda48ffcd82eb257 : SUCCESS in 1h 11m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/affe05888f6e4e22ad3139e7a64cbd15 : FAILURE in 1h 27m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1e5b488513654f56bdd31e364f1d85ad : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a73bf554e6097db7c599a14c0b6b0c547efaf9cd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-23 12:15:37.000000000","message":"Patch Set 2:\n\nLet\u0027s do it this way...","accounts_in_message":[],"_revision_number":2},{"id":"da9f185f8608335a0102ac15a01536e7ea273bbd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-23 12:15:50.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3d6ec5e3935a08ac349de5195ff516913f1c54c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-23 12:16:22.000000000","message":"Patch Set 3:\n\nNow we can +A it and stop babysitting","accounts_in_message":[],"_revision_number":3},{"id":"75a85e5830e1830ab4171753ccc063cb39b09289","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:51:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e56348fbf7b486aa46080aabff22859 : SUCCESS in 3m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e3aafaa33d9c4a0680d537cc3d8ad734 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d16d6aaa41b4fb6a2b1205f1363cfbd : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3868054849d34b1c8e982307d8c66ed5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2f1ed4fc474eb48edc04a7ad1c9da6 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/beab5de64abf438181cffecf0f136996 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0effebcbac6c4f8a94c04b0ff75c7de1 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/494a76303d844589a21e70bbfb28c31f : SUCCESS in 4m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/66832fcec3b744438f0a310012d2675d : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9f451a650a14997ab6aba50e6fb4971 : SUCCESS in 8m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7ba61211eab24457ac6d070a2d7f10d4 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dcf5ae71215d41cb964ba067c1d235ec : SUCCESS in 10m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a2c1370289104e8a9c8d50e94b3207ae : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f47805bad684634a1773f5a9b33a37f : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ab54054dcfb0486e84edc1bc73f337fc : SUCCESS in 14m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/50282c7e18614521a647520e21ec5378 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2eb6968adc5495d8278318a40d0aef6 : SUCCESS in 9m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/83b6d2a964534a3ebaf994e9ee06b0d1 : SUCCESS in 36m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e0b3f899402a4df9aea5d3c4a4f7da22 : SUCCESS in 37m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ebac4e7aa7554b668c3a9daae2c3a883 : SUCCESS in 24m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1105e112097f48149cc8abd052181b5b : SUCCESS in 26m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b91c587354d046d5946413f30f1cc038 : SUCCESS in 13m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/159edc8d217546faba9937dbf5d33ec7 : SUCCESS in 13m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0e3b15415acc452691e60f29274e6c53 : SUCCESS in 11m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3bc4715797bc4a40a163f525f0783ed7 : SUCCESS in 20m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a941edc6d60446ee938a9b4498a316e6 : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2f7c374041bc473f81269f5975999225 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c7d0c5f282d341bba9c8e414ea63d49d : SUCCESS in 8m 24s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0aed474d06a644f395c2d7975a7f3d81 : SUCCESS in 53m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ab227200ed143f8af936944a07c862e : SUCCESS in 1h 05m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d620f712b6441b7ba1801ee26ce92e8 : SUCCESS in 1h 30m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a8e2c62d3dca475bb430718c817fee6b : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a79790565e7358ecc41ee96bcfea2eaf4166ffbf","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-23 19:36:12.000000000","message":"Patch Set 3: Code-Review+1\n\nDeps have merged and this is passing, should be good to go now.","accounts_in_message":[],"_revision_number":3},{"id":"03672ba61844b04d63cf236408455bd2bcc40470","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 19:50:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d9b0fdde8b3333007619e2de9265b5a565f4994e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:50:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"16bb85e91aeb39d75628370c0352db74bdb846ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:38:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf7fb6bcfec4e538fff3639311380c8 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5efa122bc9fc4d859e2c95fa5337b583 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/962e7745d5004bab9783c70abbc4534e : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd905f04bcba4855aeec9db90f674a8c : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cea512e45b0f4691bc80dfe3eb993e1d : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8af53e31166843f3bfd171875104a731 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49aa1d54aaa547c7bf07deb8ef786340 : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f26c5587fe0c49528a4f47bb9d49942c : SUCCESS in 7m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88be7ac4824846fc892398ff18a56b4f : SUCCESS in 19m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5e3ec09a4d9c4b2f86b1ad8b85bee442 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba57a6eb4ec14476ad9b0888e9cac3ba : SUCCESS in 9m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5b0df781b90a4828901aca12256fda52 : SUCCESS in 5m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4d4021e0973440eb225d9e05b9f1683 : SUCCESS in 5m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cbe02271095343258fd7cd68f2a85147 : SUCCESS in 10m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9af950630c754bf19fba0c7475a2394c : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e196a3a06c44abd84f1a40af6096b13 : SUCCESS in 7m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ad57060dbb9b41e78646fa3db0148636 : SUCCESS in 29m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ef5ac6bfbfd44b39420c3bbe8a72cfa : SUCCESS in 34m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/19ebd68869c547ed8eb9c3f31d940004 : SUCCESS in 21m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1f6c22d1f8e43ddaacd77f99426f542 : SUCCESS in 25m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ac629494959d4dc7b5f2af93659eb0b1 : SUCCESS in 13m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/31b2e26271cf433f863d723e0538d0f0 : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/deeefebb434a4afa96ecd1cd51b94301 : SUCCESS in 11m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd461cebf9fe4e96aff1e9da6e24c6b5 : SUCCESS in 16m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1013e401597c4d66b6e74fa0bfa84acd : SUCCESS in 5m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7d964403c9674ca995b15569c8b66e0c : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19d353702aab41d3b6fd1222e46a2449 : SUCCESS in 11m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0936153fbc4548459e6e6ab91285cc70 : SUCCESS in 1h 00m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fbb75f278cc480ca34eb9ced610b9f5 : SUCCESS in 1h 15m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3e09452d1a2b443b96a0fd25c5f83a3e : SUCCESS in 1h 43m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1ddb1e3b7d1d4543a6342c8578134cfb : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":3},{"id":"53137c2e77868db1398fff0eb7a8d7c127617e90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:38:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bc0c09d89968157d5734bfc98e0a9969ef7f53b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:40:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd46bd9349a544e5b329338d7abe9e0c : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"196d188206c82786a15a366202625b12a6fde3a1","revisions":{"d515b9aa62ee0f30d91b5832cc2c6cd47568771b":{"kind":"REWORK","_number":1,"created":"2019-08-20 14:57:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/677479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/677479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/677479/1"}}},"commit":{"parents":[{"commit":"778fae625fd7d1a89b3810704d4407f4bf56a80d","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/778fae625fd7d1a89b3810704d4407f4bf56a80d"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-20 14:57:27.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-20 14:57:27.000000000","tz":-300},"subject":"DNM: test websockify-0.9.0","message":"DNM: test websockify-0.9.0\n\nChange-Id: I3853803780197c419cde0c87447f795edc6d7382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d515b9aa62ee0f30d91b5832cc2c6cd47568771b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d515b9aa62ee0f30d91b5832cc2c6cd47568771b"}]},"branch":"refs/heads/master"},"ad7aba56dda17569fe0205450db64839625f886d":{"kind":"REWORK","_number":2,"created":"2019-08-22 18:42:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/79/677479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/677479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/677479/2"}}},"commit":{"parents":[{"commit":"28875ef9ca02f1c62b6af10ce04907a926e44bbf","subject":"update constraint for mistral-lib to new release 1.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28875ef9ca02f1c62b6af10ce04907a926e44bbf"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-20 14:57:27.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-22 18:41:57.000000000","tz":-300},"subject":"Update to websockify-0.9.0","message":"Update to websockify-0.9.0\n\nChange-Id: I3853803780197c419cde0c87447f795edc6d7382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad7aba56dda17569fe0205450db64839625f886d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad7aba56dda17569fe0205450db64839625f886d"}]},"branch":"refs/heads/master"},"196d188206c82786a15a366202625b12a6fde3a1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-23 12:15:50.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/79/677479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/677479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/677479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/677479/3"}}},"commit":{"parents":[{"commit":"28875ef9ca02f1c62b6af10ce04907a926e44bbf","subject":"update constraint for mistral-lib to new release 1.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28875ef9ca02f1c62b6af10ce04907a926e44bbf"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-20 14:57:27.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-23 12:15:48.000000000","tz":0},"subject":"Update to websockify-0.9.0","message":"Update to websockify-0.9.0\n\nDepends-On: https://review.opendev.org/674364\nChange-Id: I3853803780197c419cde0c87447f795edc6d7382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/196d188206c82786a15a366202625b12a6fde3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/196d188206c82786a15a366202625b12a6fde3a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
