)]}'
{"id":"openstack%2Frequirements~677727","triplet_id":"openstack%2Frequirements~master~Ib96161ed15546e05f98b33499ed12e1f1ad449c7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib96161ed15546e05f98b33499ed12e1f1ad449c7","subject":"update constraint for python-neutronclient to new release 6.13.0","status":"MERGED","created":"2019-08-21 13:25:56.000000000","updated":"2019-08-22 03:45:19.000000000","submitted":"2019-08-22 03:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677727-1566445350108-b79180e0","meta_rev_id":"a29ab250aa3fae74572d5cca22b15d48e481e62b","_number":677727,"virtual_id_number":677727,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-22 03:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-21 18:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-21 18:03:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 18:03:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-22 03:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4387df1a43ed21bcdfb3567e86a3f289b7eb94a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-21 13:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70835d5b3eff77d598a646c4a46566cf3cb1291f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:16:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e91b41ec3c4b6b8014c333e7f4671d : SUCCESS in 4m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8f9ba466ebb848f79abe01dc7c1881b9 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986f6aa8656247e9ae179f3331f4ef91 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/908b60afdfc245adb39e1efa85de0298 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03598d8dc00f4a468ecadbc70407fc57 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd20ae34d12d4fc4919b0e4fd068991a : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f38a5e64a7a940fc8059987fa2565199 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccbbf779bae74b55b50b4d6549f7495c : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/363137f9438b44babf30fbc59fc038b2 : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d30d13e4b5354f1f85b21a0767051d63 : SUCCESS in 7m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bc6336bafdc54daeab85282eb0fd85a1 : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d1f3b2b3e6344bec96e79ee951e60b64 : SUCCESS in 10m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e914f20d329049a890a3a1dd42f804b0 : SUCCESS in 5m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14f9252d870142ca8cae12ec383a2f8e : SUCCESS in 6m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d0a3e79a772d41bc8c7d1a215fc4ffec : SUCCESS in 10m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f1b617efcc8c4d6fb7d0c9f8078a6cc2 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82f33804a90d46a6816fc674967ddbee : SUCCESS in 10m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e731d44ef76d4d4e8c43bd2a8ac2d023 : TIMED_OUT in 42m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a313acfc59434c1b898fe86fc527ed05 : SUCCESS in 33m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9baacf5eec794091a13335a4efed16e8 : SUCCESS in 26m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2a877d7d576c4792bcf8fed529121fb8 : SUCCESS in 20m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/abc7bb0ff2bf481f92f15eaf61a43473 : SUCCESS in 14m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aaaee381d91f4e0d94f9bbfadc4fb73c : SUCCESS in 15m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dc29f9fe8d214a04a25b888d72ee31c2 : SUCCESS in 17m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7829a1335c64faf9bd18327fef32857 : SUCCESS in 20m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/523e57dd1c484bcd8c1606ec50758b77 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/56d91c8a30f74f859b80e3776a49afcc : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85a801fe50c943cba711bc18f8acf83e : SUCCESS in 9m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c5238563b04f4c579fb1687bb66ba2a5 : SUCCESS in 59m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c511c80c7d44e97907280b83559a9c9 : SUCCESS in 1h 14m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d0b82d910bd4ada9e856703aea48934 : SUCCESS in 1h 44m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a6d8c0429e524db3b6ec2648a4930953 : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"053ae414bf75ff4fdd59f368912bc095e293ade2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-21 18:03:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c434f63796903343dfb77df6dcda18abefeec9ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:00:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09c906e59304e589a05ab1cf13a7be5 : SUCCESS in 5m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/70785afee4704f12860eddfd55303f1a : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f982f20445b842b7bbda04feb1fc6fae : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce0bc335083e4ccb9a87c5944a32bca5 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/730a2e577ce842b48243eef1ac893703 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e65940e2490454daedbd16ad45a2597 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b491cccf143b4a27819c68abed9c3e87 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83fa9a2f4b6041fea6746aa7c34d885f : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/838ddb8efbbe4f139894ca352f586635 : SUCCESS in 15m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb7d25375d1f4b4eba3b6446de82e721 : SUCCESS in 14m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/96c3083430cf47559767ec55b0519566 : FAILURE in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/276047bd4c624d008717a8c9e63c48ea : SUCCESS in 11m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa3722f8f83c47efb22b9534fae748ff : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/55fe9451752349bcbdb5e222676f677d : SUCCESS in 7m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ef4604cfe515486bb80ffa3f829e3471 : SUCCESS in 10m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/02fefbc9650e4296b7db6f61d0ef80b2 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb8824a2dc314c69a03e1a48127dca9c : SUCCESS in 9m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fa2d85efb0694c7d81a3af32b6d6a699 : SUCCESS in 27m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c70ea23ecbc64ad488372b7fba49886a : TIMED_OUT in 42m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/61b6919dcff6495aa822acf5079876d2 : SUCCESS in 28m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e77bd53527a14127920679dd2ab1526c : SUCCESS in 23m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f08c507ca8c24dda8c9e8efd2beb5305 : SUCCESS in 12m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68487e44d8224b928a3db5b0b3509f3c : SUCCESS in 13m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc76f388d24b4992b9229a3b5a1ae2a4 : SUCCESS in 12m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa3fe0eb2f174dcaa50e87434c6b5397 : SUCCESS in 19m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37f5dcb424564ea68c2349a39aadf2a1 : SUCCESS in 5m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7cbab163d1c14252a7bb7bdabcee397b : SUCCESS in 8m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b49dd0248434bf9b69b7e60d9f21023 : SUCCESS in 8m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e66d53b812ee4f3b91441fc98efccb15 : SUCCESS in 57m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/384060b6c27d4eeaba3b1a4738511ca3 : SUCCESS in 1h 19m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ab1671240b174209a35c8156f79792ca : SUCCESS in 1h 35m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b380053ba8244191a8da2851f43dd8e6 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":1},{"id":"fcfcbfe0bf67126e03105e66a10fafca5e6f41fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-21 20:16:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"644e20914b6b70eb5c07b504c9da7f3e20dea0b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 22:02:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d19a8471e7d4c468e1237e5e7d43336 : SUCCESS in 4m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/55cf4362d205484494c0db1389294ed7 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e76b4d8b624a08ba20498dabda29f4 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1261415464eb4ca89534c021a6293cfd : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4a8b61db85542228745cc9753666793 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c8b5302cf454b4286df68c3a0c5acbe : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5148eee1cc14bd29e4d50a11516c9f9 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b1634d05fb94c3e9e6db4030db72562 : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6a22250f84654f56aabd08b3d2598f53 : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/878873432d464259baf70b79a90145c5 : SUCCESS in 20m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c12f24998f48401a82508d8ec0e8bbd8 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9106e73866314acda4b4bbb5d64c6382 : SUCCESS in 8m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3241b8f8d72f47ff82e7355dd4b06a5f : SUCCESS in 6m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1ac99a71f6d4d609acbe03bbf5003d3 : SUCCESS in 5m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d62b5ef00b794daf9b5e1a69019fc570 : SUCCESS in 11m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b53d671e494a4459bd0eb92d97b58d0c : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96dacb66233e40d18840043576044913 : SUCCESS in 7m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e43da5d71d954b9680bd6ad2c0c2b90e : SUCCESS in 25m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ebfd661072f74d5cb8d850010b76270c : SUCCESS in 30m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0cce45342a064619aa788f3334e2e401 : SUCCESS in 21m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f40eb00472d942598f1bbcae74f7cc63 : SUCCESS in 22m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ad5a3b9fa7c747668d4191a35eb99a2a : SUCCESS in 14m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b0764fad84cd41318e505edb94e68e16 : SUCCESS in 11m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e5b2b574bc5542a9a7c48d674e8a4d53 : SUCCESS in 11m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31bdf5263c0c482a9d4a223ef5cb2149 : SUCCESS in 21m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d7d0e7ddaeb4743afaefc1b41903865 : SUCCESS in 7m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/748a94cf0b394df2b1b19aca1f220fea : SUCCESS in 14m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c0a30e0f8bdb4ca49c875b85f760c562 : SUCCESS in 9m 16s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/387dffc8d43a4c4cb105a6a5382d8ca1 : SUCCESS in 58m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5d61e34c73b407e9645b800ef1300b7 : SUCCESS in 1h 01m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4c9fda312c1e419ab385642b3480880c : SUCCESS in 1h 44m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b2220b80d3324cd9ad3b5fed2db61699 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c27dfa01d5b517060877ba3d5f095b0fa08a2c9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 22:03:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dbd766caab85812b88516abbd16d7abcdd45912","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:42:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d09ee0b2a34ecc98f1f8e712c9c701 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4e43274ee64889be3ea16dd3a61acc : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a25bb78ff0744a38a1d8257bc2a6efe1 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8da08a1d28ac4111a3c7c34afc1cf12f : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88397aa6bef74227b445b714ca48b851 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6361d9c2395498b8bf6a2ed79daab57 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3fed89adae2b4ee9afde062d1765a755 : SUCCESS in 5m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0789956d3b114498afca5eee6ba170a4 : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e56b964b5544db78dcdb7084cc99e81 : SUCCESS in 8m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/90187332d217444d89a49b48a7ca057a : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa5aeef5937b4f71ac5d1ce20bd16335 : SUCCESS in 11m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0b22e14ee31c4cff9997edd66a826937 : SUCCESS in 4m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ae7d674489d84614be6358fae5aec877 : SUCCESS in 5m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/06e73d967ed741d2b3b5c1a73d164376 : SUCCESS in 11m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/be0eb2a0ffe3410db0e273584b67534a : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a88b577cfae2421994911855bbc9c45e : SUCCESS in 8m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7be04791f5df413190fd5b83a72c8bc6 : SUCCESS in 35m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aac05752c2b94293a0bdbdb2b7246958 : SUCCESS in 31m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/83cf056ded264fc48a38f2109005af3d : SUCCESS in 22m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d737bb9b15994de98852f3a41b5b40cd : SUCCESS in 24m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ad36e0d0207f4c3686f1e1eae232fea3 : SUCCESS in 15m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3294908cc3f9415e9bed9c777a24c29e : SUCCESS in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/666a04848c9646b4b7c31e7c29c1e823 : SUCCESS in 12m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7d3c1e897bd47a090ab18a2fff889f7 : SUCCESS in 20m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb4daa5257a84003b924174bf769a004 : SUCCESS in 7m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/51dffb1f58794565a08a9f5b5ae1d7f0 : SUCCESS in 9m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ca03034e1cda4b2eb3f321ea5e8ae4b1 : SUCCESS in 11m 08s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/39ac902be2d34700876fd5d3fba1b322 : SUCCESS in 56m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb1247a06ca3479d9cfa69230229575f : SUCCESS in 1h 12m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6ce5f16e48e249df80bc899209c97661 : SUCCESS in 1h 38m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9adf93c9a17e4628ab71dcaf3715163d : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c807636264b780e26187570cce78dc58d4546531","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:42:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"553dd22c6801db4ce0c9155be7e5ea2564e027aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:45:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba27d94f9034dd0960878dcdbebbd7e : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba3ec0e76cceb2e3b758bd4ed36636c2f61b4fae","revisions":{"ba3ec0e76cceb2e3b758bd4ed36636c2f61b4fae":{"kind":"REWORK","_number":1,"created":"2019-08-21 13:25:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/677727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/677727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/677727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/677727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/677727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/677727/1"}}},"commit":{"parents":[{"commit":"150d8ecd0d8bdfa36354bbe23ccd0291cf9967a1","subject":"Merge \"Adding sphinxcontrib-svg2pdfconverter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/150d8ecd0d8bdfa36354bbe23ccd0291cf9967a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-21 13:25:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-21 13:25:54.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 6.13.0","message":"update constraint for python-neutronclient to new release 6.13.0\n\nChange-Id: Ib96161ed15546e05f98b33499ed12e1f1ad449c7\nmeta:version: 6.13.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I21a6981a70ba5290fe951f36ddb3800ab25632cd\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba3ec0e76cceb2e3b758bd4ed36636c2f61b4fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba3ec0e76cceb2e3b758bd4ed36636c2f61b4fae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
