)]}'
{"id":"openstack%2Frequirements~677732","triplet_id":"openstack%2Frequirements~master~I0944224a65ede02154dc555d2a11f52f1507217c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0944224a65ede02154dc555d2a11f52f1507217c","subject":"update constraint for python-qinlingclient to new release 4.0.0","status":"MERGED","created":"2019-08-21 13:38:02.000000000","updated":"2019-08-21 20:36:55.000000000","submitted":"2019-08-21 20:36:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677732-1566419815001-b6a2cfe9","meta_rev_id":"6090ddb07ca722491b731f2813dc99d1648b48ec","_number":677732,"virtual_id_number":677732,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-21 20:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-21 18:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-21 18:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 18:04:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-21 20:36:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de095c936ea5550cbe28cd6bb96aa3d3fc444397","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-21 13:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9242564bed1515cdb65afc5776746326a397a23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:42:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792297d269fe4c848805a9aa139380b0 : SUCCESS in 5m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d55db9b862bd4836b134627f2bcc24bf : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2bf10327da44b1080d24e72a789a22c : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c5e304257d4a85ab283fafa433cb51 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b316e272434440adbecf4e8b712a32 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d5a64db745c4dd9bb4c39bc4bf83360 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cbf56b3a91a4cf795a84e2a5fcf5bd3 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/477c105c781f49f4950d8ce19f61a171 : SUCCESS in 4m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/30d9b278a07d4b98822bad698cc4ee60 : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a21b3d610fa41709199b8e9320e15ab : SUCCESS in 6m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bb86d9efb4984f959b318d8affdd60e1 : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/131c03f740a14ec9be69490d735fd39d : SUCCESS in 8m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/80849875d9e24b10b8b0944e14f3e0d3 : SUCCESS in 6m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/364e9e5aeeb4405ca632105579ef2a8b : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d74bbc67585a4786a051e3c9ce84ec22 : SUCCESS in 11m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a076808699744f45976bbec3a52ebb76 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14a349c328744a76b6207ef29e9b354d : SUCCESS in 7m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c40600e70e8f4331800fa0da2a0845af : SUCCESS in 28m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2337cc971b874d24bd9329e85a4a936a : SUCCESS in 26m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6c9793ccbc514b3bad2f919cfd793702 : SUCCESS in 25m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bac06b703b764eaaa57a6a5952ffeb9d : SUCCESS in 23m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c135f4c312f746d6aec0d1ccfd837119 : SUCCESS in 14m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/954e66db6d5e4bac8033451f72cbb2fe : SUCCESS in 14m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9dbfb393dc8c45a785961b67a1bd5540 : SUCCESS in 14m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67646ccdc52844f3a8d5f00db86ad919 : SUCCESS in 18m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b3efee880e74a67a96dd857d48115f7 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a4ad72aab5444e6ab1db597575ff4919 : SUCCESS in 10m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/10be92b993c342cb9fd96c34d31f8a61 : SUCCESS in 9m 21s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/278ddc7fad264f578786790f13455295 : SUCCESS in 1h 02m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb7d954f688d46c4a4886998b5a7ef3a : SUCCESS in 1h 11m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2f11bf35adc845edb84fbfcc7e11cd24 : SUCCESS in 1h 52m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1ae2c7f618fd42aa953e7f0ad7f88459 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b172d5852326e59c99b09494861abc5688b2f57c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-21 18:04:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2412edd46f97febb2e10de75e9ed7beffe6cc0ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 18:04:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"647fc375930a99782bdb8cf520ab12c1e05bf0e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:21:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c88e73bd5a4733bd76dfd201960f2e : SUCCESS in 4m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9a1fcf7334b940949641e11adfc38e10 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65a2123fb2a4dd5aaa93fc08711d870 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/268fc8d0268a41d9a122f07e17dc9494 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c256454f1f414fb4ae78f1e8d9ad9feb : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6db727121f11441c86f40e2e16c4da05 : SUCCESS in 9m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2684dcfc4ca44d91a92e91eeab8ca121 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2813c433ac445d5a01936c6c2b551b4 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0d2c42a62f5c46e28fe76a00f1888aaa : SUCCESS in 13m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0efd37be9ce4f8babe682e2d87fa185 : SUCCESS in 12m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/40bfd61632d84bed8ebfd92147e3f435 : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/611d2c81bec0452083cf358621d1c390 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/304b95c1862f4e9b9c87d6cf2c56e6d7 : SUCCESS in 5m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/960129a8a67d4d1d9751a907dd85aebe : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/12e3dfd678994d158002220c424b6b7c : SUCCESS in 12m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad6cfccc7ab946d899dda0b793ff37a8 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3098e519547a45c59b920887db60079c : SUCCESS in 8m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/58afbddf27084eaf9ab454ebc54791dc : SUCCESS in 26m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a1acbe5879c41fb845ac4686e1ab803 : SUCCESS in 32m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0a287bec7e1945b7bb9bfdf31b3135d3 : SUCCESS in 21m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7d8e94d1e53b43cab1751ab7bbf6b7b4 : SUCCESS in 22m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/da914122f0c340adab7280416dfd4862 : SUCCESS in 12m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e94d226a4d714ef595ba18a5f2551e26 : SUCCESS in 13m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/01cc51158ed0444ba8502d17bca561d7 : SUCCESS in 13m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b38c10423ed045c69e7f3be4c2d05f12 : SUCCESS in 22m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d5ef2a04f4e453a830f26de8e5bc979 : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/842373a919da446b9074f038b48afb57 : SUCCESS in 10m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/279e974d2c38463796707f552bde5366 : SUCCESS in 9m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9e7da827ce224c7582225c97bf361c38 : SUCCESS in 57m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbd2069396fb4524b7425dd054962e86 : SUCCESS in 1h 09m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ece7e98fef6644cca2c89c11119f67f2 : SUCCESS in 1h 50m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0279a0f5e21642e0bd433b94a12b9541 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"460e4773fb5edf09e6cf03d7fc46b069baa4ce1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:36:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b298768fad034a438adb183f97c26a3d : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17895015106d4527b111e6fc6d90565d : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3927522737dc46148738d64e9a5d00d7 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ddbc54a0ab84ee1ab5761c23d5bdc69 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a266506857d43328e411d7d14755278 : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ac020ed5e0d43b09f0d0961d89fd915 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1a57f0214d1401c8a70b2ce4d714e66 : SUCCESS in 5m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3caa2f3c40b24ccda380194c318f583b : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/591e51ee74e240bebdcb98ff1a56f46a : SUCCESS in 7m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d584f7a1a9d94dfea7cb943211f23bac : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f1886f1a9c224e18b2f8718116fff9cb : SUCCESS in 9m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4039d3dca4ec42e1ac2a611736ec6025 : SUCCESS in 5m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/873f75885b28411494a1d5f62b05d9e9 : SUCCESS in 5m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c0dbceff39af46bcaf5b1e4f0c09fb12 : SUCCESS in 11m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c9e9bbd20a14155951d75f7e73adb1a : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99a01b12c62a4913bf84ad63a17aa0fe : SUCCESS in 8m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7cb34ae191a24b8e9595ffa47cdd3957 : SUCCESS in 27m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d885958e8014477998c1923917183233 : SUCCESS in 27m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/833de6cdad4143d19340b1c32667540b : SUCCESS in 23m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5eed49688a43411f80e508f8ed53e271 : SUCCESS in 32m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/24d6b582a3c540ffa44128d8ea46a8a4 : SUCCESS in 12m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dcd59e3c16794f12ab7ad30eb7737f2d : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fffdb5bb36144c1da45a2cf5a3fcbaa5 : SUCCESS in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/378acb9202674528a64281279712addd : SUCCESS in 20m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3462ad453d9483e83ee03e7b5fbf70d : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bdbd8eb5d87947b889b91dca3c9a8985 : SUCCESS in 17m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/34ee8d6387614f77bc812af0773d2e0d : SUCCESS in 10m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/729d7b2709bc4d5cb106f36116c680e4 : SUCCESS in 48m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/97670123fe874d1596909e71f58b17a2 : SUCCESS in 1h 02m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/95fa954f65114605bee439ddd6599e25 : SUCCESS in 1h 32m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/364128790d3642aa9f3f440a067e2a1d : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0758ce99da0ec8262d254114c247a14d88d2cecd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:36:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f4df621e10f13943562895817fdadaa1cf3c3dc","revisions":{"8f4df621e10f13943562895817fdadaa1cf3c3dc":{"kind":"REWORK","_number":1,"created":"2019-08-21 13:38:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/677732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/677732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/677732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/677732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/677732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/677732/1"}}},"commit":{"parents":[{"commit":"150d8ecd0d8bdfa36354bbe23ccd0291cf9967a1","subject":"Merge \"Adding sphinxcontrib-svg2pdfconverter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/150d8ecd0d8bdfa36354bbe23ccd0291cf9967a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-21 13:37:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-21 13:37:57.000000000","tz":0},"subject":"update constraint for python-qinlingclient to new release 4.0.0","message":"update constraint for python-qinlingclient to new release 4.0.0\n\nChange-Id: I0944224a65ede02154dc555d2a11f52f1507217c\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Gaëtan Trellu \u003cgaetan.trellu@incloudus.com\u003e\nmeta:release:Commit: Gaëtan Trellu \u003cgaetan.trellu@incloudus.com\u003e\nmeta:release:Change-Id: I8bec038f8e745746dd4b9809f299c324b861d990\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f4df621e10f13943562895817fdadaa1cf3c3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f4df621e10f13943562895817fdadaa1cf3c3dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
