)]}'
{"id":"openstack%2Frequirements~677941","triplet_id":"openstack%2Frequirements~master~I9d634e7cf569c07764defe46b6c34ad2df041006","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9d634e7cf569c07764defe46b6c34ad2df041006","subject":"update constraint for mistral-lib to new release 1.2.0","status":"MERGED","created":"2019-08-22 08:51:01.000000000","updated":"2019-08-22 15:42:34.000000000","submitted":"2019-08-22 15:41:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677941-1566488477253-3b288848","meta_rev_id":"40c8844424a3d244cdd30906b438fcf59d38f02e","_number":677941,"virtual_id_number":677941,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-22 15:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-22 13:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-22 13:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 13:13:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-22 15:41:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"806bd9d0feebe60796980dfa6ae8c7748be52616","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-22 08:51:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6edbee59a32f120d1b25ea2e0fc135a0870bfba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcda21c00fce4df5b1358fbf9f6e46ce : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a4b5c4df5e344f3b82f7adee9e58aeb5 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22e3a42d3bb4b63817a91f23b413009 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/087853fab0bf49a1a3e53c9bec8392a5 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c63a52f8c7b64a3487ba3e93b932faab : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0889417e72e644c59409341eeb2839cc : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c9d8dde7be04a05beeb80982b1c79fc : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2ffc8d530bb409cb2a86430694ef6e6 : SUCCESS in 2m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2a1b71d893a74fba82bdddc60476a93b : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/671808208b2b40b89371d511877b515d : SUCCESS in 10m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fc2fe1ad6bd54152a95671a03dd69fa2 : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c6bbaf18733b43289c6c8aa28b665a5a : SUCCESS in 9m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7a9a57fd7c824b4681916b6c8c3d1fc8 : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fa1da70879594173aa2b811d7e49f6d4 : SUCCESS in 5m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d3b56be6c069481d80b235c27484effd : SUCCESS in 12m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7100d6ddae89471fa5b5a95e387689c3 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/543e8930239e4e69941a8910cfbf2ad4 : SUCCESS in 10m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c58cbeeedb1049f1b80a2bf16b67e464 : SUCCESS in 27m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c0f3ff0d7ba94f9d9d934c5f67a33c82 : SUCCESS in 26m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b8fb2ce0dbc9459e89a2f259bc37b5dc : SUCCESS in 25m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/85662955697440d8ac4aa4d04ffa4277 : SUCCESS in 26m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/92fb45208c534272b3cf2fbb285b5542 : SUCCESS in 16m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/12753661d51940c999742d29a0a1eeba : SUCCESS in 14m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/29c6cb4f807d4708bf9f11bb49fb81ae : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c9e69edfe73435a802690e6c4b7a170 : SUCCESS in 19m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe3f3559ab7e46d1897c077615f6867c : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b68fdf31663c4244adf1b445b24a384e : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5a757ade46c14718bf64e85d556ea0d5 : SUCCESS in 9m 05s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bb1c9bc9834842b18d8f536c81443699 : SUCCESS in 59m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04141540896b43be8067d64df0bb40e6 : SUCCESS in 1h 13m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/590eef88a33143d0beddfb54b510b409 : SUCCESS in 1h 58m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2b22c10016b741318441e3d7ec0ece03 : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c8cb0fa22bd5e2497edc097cb1fde6ecc2c07086","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 13:13:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dec968a651e02cca081fe156021a09ce6fc000ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:14:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc0381bd74c59d29924ccea8e30738322a8db974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 15:41:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7207565977b64e92b63ca39d36e8f45a : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17df4a143b32434e93f5b33f420feba9 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbac5688e5be4b0cbb7c71a5fa82ac42 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c5cf90a6d7248a5bffa716e56fb13df : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c80105f653942fc8360d20f3214108f : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dcb36930fdb44053888b34448b0dbacc : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b15a9414b6ac490db95f14afbd485fe6 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9e387826e60f4776a00ee15b348194b8 : SUCCESS in 8m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3adccd32a60649aa8d8eb732a1427791 : SUCCESS in 8m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1a81cd2db6654a2682894f5253b6dfc3 : SUCCESS in 9m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e8f96336b06420aa469bb24688241d5 : SUCCESS in 8m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/90fbe0389fa5417fa3a60c0822fd9936 : SUCCESS in 6m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2ed2fb4deb546b7898b644e127e75b8 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fb52965778e94d00b0cc068b2ad54a1b : SUCCESS in 15m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b935d19bc0f24fbc83ebf94f2320c44b : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd252eabe07447968eb4a3857e2fd4b7 : SUCCESS in 9m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7a4b686e6ab64eeaae2bc31a417ab394 : SUCCESS in 33m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ba1db40745e14082ac77c2477a3a6094 : SUCCESS in 31m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/25fca9449f4a40859304b3a86674a0df : SUCCESS in 23m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/035d91324e3e49ef9ecdde8c5aa0dfca : SUCCESS in 26m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cd22ca9cfafd41d59720f0bb696d255f : SUCCESS in 13m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1db169c9afc74486b089b721044d30e4 : SUCCESS in 12m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7e66f0e3195a47f89d74605a971fc4c4 : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e63d277bf5d2498abf1c388d225a5e9b : SUCCESS in 23m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdcc11eec7a04da48da2195451c06458 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2f0a9a8dd6904082b930ae44c2f3a6d9 : SUCCESS in 8m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e0cd333d0f484a5d8ed6738cce3ffc67 : SUCCESS in 11m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9ff3da7f515445ffb1dddd485683a0ab : SUCCESS in 58m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d707b72e81841e1a142cf7e551683d6 : SUCCESS in 1h 11m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/790ca59e0c8f46e3b553caad1606ab44 : SUCCESS in 1h 36m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/117b276ee94240d1a31fda9782f845ab : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"10d1cf0b9b89e2ffdb54fbb7a804f12a5be6cf1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 15:41:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b6827b70b476c084623e8711d37aa8446cae42c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 15:42:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1561feca566a429e81c3911a9ce4e71c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28875ef9ca02f1c62b6af10ce04907a926e44bbf","revisions":{"28875ef9ca02f1c62b6af10ce04907a926e44bbf":{"kind":"REWORK","_number":1,"created":"2019-08-22 08:51:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/677941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/677941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/677941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/677941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/677941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/677941/1"}}},"commit":{"parents":[{"commit":"1e13a8e6cf9bbb97fe5e9093ea3dcec4fb01f79b","subject":"Merge \"update constraint for openstackdocstheme to new release 1.31.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e13a8e6cf9bbb97fe5e9093ea3dcec4fb01f79b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-22 08:50:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-22 08:50:59.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 1.2.0","message":"update constraint for mistral-lib to new release 1.2.0\n\nChange-Id: I9d634e7cf569c07764defe46b6c34ad2df041006\nmeta:version: 1.2.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: ali \u003cali.abdelal@nokia.com\u003e\nmeta:release:Commit: Eyal \u003ceyal.bar-ilan@nokia.com\u003e\nmeta:release:Change-Id: I070996739b2bbff701b7fa1acaa0f2cff07914be\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28875ef9ca02f1c62b6af10ce04907a926e44bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28875ef9ca02f1c62b6af10ce04907a926e44bbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
