)]}'
{"id":"openstack%2Frequirements~677942","triplet_id":"openstack%2Frequirements~master~Iddf926d52b61527ea2438691192ca26120136587","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iddf926d52b61527ea2438691192ca26120136587","subject":"update constraint for ironic-lib to new release 2.20.0","status":"MERGED","created":"2019-08-22 08:51:59.000000000","updated":"2019-08-23 04:54:59.000000000","submitted":"2019-08-23 04:53:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677942-1566535999335-9d7068c6","meta_rev_id":"3f0ee75397bb18533bde518268253c202f3b20cd","_number":677942,"virtual_id_number":677942,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-08-23 04:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-22 22:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-08-22 19:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-22 22:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 19:04:50.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-22 22:30:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-23 04:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bc8448f096774890da81cf9c21868130dd23cea","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-22 08:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46957734863612ed65a8eed0a68c9bf902fea2fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:08:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5704babfd9841b4aec165d757bb9dd7 : SUCCESS in 4m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a355037018c246f6a65e3645ac1c3080 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e397de8151a94d659f1e4bf70fb246aa : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13e6e29133b54e8caf50215e6a4aee76 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1268550f5a466388a4adcd8acf4e0e : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fcf773ec39a43899d48856fd7c0e6c9 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ead880c0a254d2cb850f483ea57d203 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9050d73fecc47e3ac7ccb444e861651 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1ba31e2f73ca43bc8df5cd37d3d9fc78 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b719c9f5b0e7447eb61dca4fbbda3b3a : SUCCESS in 7m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/567ce793430d4b08a6790d6758e68b27 : SUCCESS in 11m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a84392243de74ac88a84d8a06b74d9f2 : SUCCESS in 10m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b79572a64ef8413684fb1dc93f475f80 : SUCCESS in 6m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32f8d348598346d3a09f645b2e5bc7f6 : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/110c9341dc0441979b5511f04a66cbfb : SUCCESS in 13m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/46d47d82bd1b44399610c6db85180fbc : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83459f5cfeaa433799c50267036d1874 : SUCCESS in 8m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/36f1182a4f31400180e7c1cd78983632 : SUCCESS in 31m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5adefaa1ab0143d28e1cf98e3c7888cb : SUCCESS in 29m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/03d39f2b7d104134bbc4694eb4363ecc : SUCCESS in 26m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/68b084c440f349469b9b861f38d9e387 : SUCCESS in 25m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d1525df58fad470686621c68ee096ce5 : SUCCESS in 14m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/390a4d1bcfbd44aa957ae8b43052d3dc : SUCCESS in 20m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/95fee1087e7746cc8131ffe592222b96 : FAILURE in 24m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39d5413c511348e0a58d6331686f70c3 : SUCCESS in 19m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a11c5238003458d81294de74c9f9386 : SUCCESS in 5m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6a9858b5effe4856bd33b1ab65604bfe : SUCCESS in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/93361135f1824d09be356f29625bf8c1 : SUCCESS in 9m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/78fd043a62114a34a5ce41ed6c216ac4 : SUCCESS in 1h 17m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/051262b93572482eba113947311cab92 : SUCCESS in 1h 15m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f1d09676b6746e4812e77aeabe89ac3 : FAILURE in 1h 33m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4f3e8e83158845db8bdda162d28c6a3c : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7d42f1a79d983dc5b42fa7a547aee31deafdea50","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-22 12:22:57.000000000","message":"Patch Set 1:\n\nrecheck unrelated failures","accounts_in_message":[],"_revision_number":1},{"id":"001b4c5d7fa63a7fefc05246b56f7465270a887c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 13:14:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd469d10499084b09e650cdf9480865e17937b6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 15:38:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633bf579f4ee4d9f94e195e8024f390a : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/67156f65d1fd459e8c53b2999a64345b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e893a8e38e4a928e48ac3b95ccf84e : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/597090df72f645d9b9b0a03999f5f30a : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68ea4f6a14f1487bb6e5c15ac68e8f36 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2e19b643f274f24af5e782a94321e98 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07b454a622984a28b72c4cba55076092 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07901eebfab44a939f384340bea0fc7a : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/15948f1dd36945acb146ae6646efd9d6 : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12030b5296584e7e8765833792aa4a7f : SUCCESS in 6m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0eafbabc50cc4206a547bc1214702bca : SUCCESS in 12m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a558d18f29fd4dea90b017c8f1850e13 : SUCCESS in 9m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2e5f753967824eb691f04b8ce91c8b92 : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3a45ce405a84ae493d6c35e1197f2ad : SUCCESS in 4m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/180f2e6be37643b49361aca43614347f : SUCCESS in 12m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/05ea9ecf56b84e9b8038e5f896aa55c0 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b99994933ab7413abdf8c3ac190099e5 : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7121615002c247009e9c11781eff6a64 : SUCCESS in 36m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a27fa5b65cd8488b8bcb88b209385565 : SUCCESS in 39m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/dae5bdba062b44b8a3995499d86ba67e : SUCCESS in 26m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/535eec321f874b679589a732c3dca9c7 : SUCCESS in 25m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/be2d68a150854bd6b5743c30e8b639fe : SUCCESS in 18m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0cd61edf24e0451ca5f279c3b0328379 : SUCCESS in 13m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/55194afd639e41009f0d20fd7698c3af : SUCCESS in 14m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df1b2bc4496840049a02ef26cb2ddc1a : SUCCESS in 19m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1773d34ebb72433b83858e6442afdc6c : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4cc96747d3664cadbc506e519909f36c : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db5ce508959249258a0d1ee45d8104f7 : SUCCESS in 8m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4a0e0fcb3c9740e9a7929e5cd2337a40 : SUCCESS in 59m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdd61fba627d4b6eb821918fdaaee4a8 : SUCCESS in 1h 20m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d6d7125dd2d545049790ab96bf90fc12 : SUCCESS in 2h 02m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bc83f171f9124a829d411400c7994820 : SUCCESS in 1h 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"61b3f2c4e2513b0a2995c14bb48831bb5b281860","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-22 19:04:50.000000000","message":"Patch Set 1: Code-Review+1\n\nYes please! This should remedy an intermittent failure issue we\u0027ve been seeing in ironic\u0027s CI, so we would greatly appreciate this merging sooner rather than later.","accounts_in_message":[],"_revision_number":1},{"id":"a8998debfdd52f499bcc188db2d0a1e1302f46e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 22:30:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7dfa67c11f10918c538e6aba698896474af213db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:30:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aa281c0a60bedb827da8414340b034caf371380","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 00:11:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b51a03a6eda4de6ae800dc5f2cb380f : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4176a241b94eb0ac92ebe46ac225ef : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e8b0e448ece414cb5c7eaf2a7a49f31 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2634c74fd4c4492a65074aa2291a713 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a68212ea41040cc8a9a01d043b7f01c : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a5ad525f92545b3b7ffadbe2ab18195 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7aab1ff5e8704d0797f0d0b6a3f6dab6 : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a6692a1e334c45bb97e3e82a6290d25f : SUCCESS in 10m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/17ba10c5f9074e7e81ea6f7592c02eb0 : SUCCESS in 8m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ce0975ac4b1146ac86d21ebf891d70ec : FAILURE in 10m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/934c6466d06744a7bfe416ed992028d2 : SUCCESS in 8m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8aee579d133c405b80cd9f8a96f5762b : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e488e1e0968d480a9e52ff11920ef9fd : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/acd8ef1ecd96476a83e13f12f23cb0f6 : SUCCESS in 11m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dc3603bd3a6d42c49b2b9226578fa53a : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f61e830f53274b129711a34984d84fbb : SUCCESS in 7m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ae85ce56223f4eeba78f1a723605f9fe : SUCCESS in 29m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ed9f73c24164953beabaf927bafffdd : SUCCESS in 27m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8a0f8e37fd92473db1b562f705e1f18a : SUCCESS in 29m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c7f6137814694cd1b68739e999b0604f : SUCCESS in 23m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/03d7168204e048319fe9532d3037eaeb : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c24502d353d94a25a5c3b409a589768a : SUCCESS in 12m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/229bbe079c594fb380b85aeb919cb99a : SUCCESS in 17m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e8ff2c07ac84cf0bfd77354efa5c3c9 : SUCCESS in 19m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b2b82fe4d114753bb9f47ddee3b3b69 : SUCCESS in 7m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a24efc6f27304bfcb6f5ffb543bbdb22 : SUCCESS in 9m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12c0d9a1d541426aafe89dc21a923088 : SUCCESS in 9m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/201c97026e0348ad962018fbc03744c9 : SUCCESS in 1h 05m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bd783e4fa4b4f93a648ea13bbd1cd82 : SUCCESS in 1h 19m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/52c1063032f842ab9b4ee79306013c1f : SUCCESS in 1h 36m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b6f865c39ffe4bd8bd7b4d24da01a851 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a5255223a2f5fd9f3be1406b9b7b31a9a805394b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-23 01:22:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4ca874d5c9eeea9dc28005ae8fbd128d39b7867","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 03:13:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167aa285edfa4349b9296bc66cc0f758 : SUCCESS in 3m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a35ed7657f9b47608c01db00238b095d : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9d42f4b37c4838ac21ebcee4cc08d2 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02d1fb15e3dd450bb74bfd14c3f5fc64 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b6cc385be30475c81d6d5cdeeaf0266 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0062dd57934444a892aedfb93a27417d : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f225d59b33094367888d401414dbba5b : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39714c27694d478688bc4b4d44375842 : SUCCESS in 4m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2eae0cd2e7d943939393fa115fe10c60 : SUCCESS in 7m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f75853d5ac3b4e87b3f9349fce8fd329 : SUCCESS in 8m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4794ec9df5b04dc1ac51a46314ddbbc2 : SUCCESS in 10m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/489a73a07bf943bb90f81d7e8f59787b : SUCCESS in 9m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ea03044bff49488a8e0681da8e077110 : SUCCESS in 5m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f3eb65dc09e4611b0d39f1a72edf536 : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/71a5c76e60a845858c485cfd1b494ae7 : SUCCESS in 9m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9d1c2f51b1f42c9befccb1d7e1fa396 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/598e7090cd97497aaf2d4f7cffae249c : SUCCESS in 8m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/277fd5a570314c5199f1c784e38dd692 : SUCCESS in 30m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/649f679d594c443aa2359894acf1ddf4 : SUCCESS in 39m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/48f0b09572974c25bb2a399c8eda382a : SUCCESS in 32m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c44a5f0205514732837fd1965a780737 : SUCCESS in 24m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/03e6856eb9804ed688cffda593452820 : SUCCESS in 12m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0e9fb9b85a54540a40d7c09f12d1332 : SUCCESS in 15m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/51f07f16bde2499c9863034e31ad34e1 : SUCCESS in 16m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8c8d6f2020a4fd7947c40f274420a31 : SUCCESS in 18m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6fb06004a11479fa1d98d433bc0f45b : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/20427563c50e443a86cc5f4ae887511e : SUCCESS in 10m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19666d14d42147cfab216e71dc2d327a : SUCCESS in 8m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a8a060a557434b23abb9c039871892c3 : SUCCESS in 1h 12m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/076b7424087147abaf473beec12f1e01 : SUCCESS in 1h 05m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a64b6b44b847401580004589d2aef115 : SUCCESS in 1h 39m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7e40f085d5f844e6ada186fceef3715d : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e49efaed39f2a82ceb48071147f63adb022800d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 03:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6bbc7d3afded20dafef0013c42440784c22193a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:53:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ddb4513d5c45e8b33e36a1e3cedd7f : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044fcfb29a1d43df95c0810003a44112 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28300b844472494695012d5a60cfb76a : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86ce4bef134548e0a30d26779f516335 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfafb6cd3d654300889698c9d67affe8 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21877e395aa14af594259bc316f07837 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06ac49d49c3b4a66a65f615c74e05893 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0f7d51cbdec648a9957c8bc192f129b4 : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6a4c2b5e61f647da8f9fac60e0a5a9fc : SUCCESS in 6m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/74807c92f20d4db789cf5de566fb3741 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae3e26b68c494b6a810f710f5fdb40e2 : SUCCESS in 8m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/20b1603ddd5549e6ad10e9b4e0571a78 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc7ed6a6998648ab93a7598ba0dd1c2a : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9e57620e33d34ace923736f2b26de109 : SUCCESS in 11m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/741d85997ab64eff9cabeae9d47a9a06 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65789fc090224e19b11da6fd790ee10c : SUCCESS in 8m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f9dee5a3b76439393acb6177a8ddc9d : SUCCESS in 33m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28b2506037634df083f0845438ecaaf4 : SUCCESS in 30m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8ac0e442f04d497aaf9fa678f6560f96 : SUCCESS in 25m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18d4cdaaddab4f498da4e724e381367a : SUCCESS in 26m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/33dd2d90382949eca88feedb6d166f5d : SUCCESS in 13m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3878cd2b3a8a488db6773b2cc6bc1681 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8d374ad2e503486aa79b5f943eb85d63 : SUCCESS in 11m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01db1c15ab37412ea0a3aaf9fe928c2f : SUCCESS in 19m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91a820d8d89c44e3ba6f451078848d85 : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0cac2ac92e7d44cba97982a7a07a38f0 : SUCCESS in 8m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/569494350f154e609e20ce94516e9d94 : SUCCESS in 9m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5e85c9255d1740caba6f16850dfe8cc3 : SUCCESS in 57m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d65f26a427f447d8ce1e85f1cd7df37 : SUCCESS in 1h 10m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4fa05c978214f0da6a03655bf2a24ce : SUCCESS in 1h 33m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e4f585a0dbf468c99f9d675a82e1432 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":1},{"id":"379a6a8f4a6defd8455daf1ed445d5b4d0d293da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:53:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c9895cd8134d7180620263210c7909a5deae884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0cde4f39b554ff8bd02b1b2bf6a7718 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"617511a2798e62394550b48461849f4b2b799477","revisions":{"617511a2798e62394550b48461849f4b2b799477":{"kind":"REWORK","_number":1,"created":"2019-08-22 08:51:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/677942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/677942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/677942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/677942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/677942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/677942/1"}}},"commit":{"parents":[{"commit":"1e13a8e6cf9bbb97fe5e9093ea3dcec4fb01f79b","subject":"Merge \"update constraint for openstackdocstheme to new release 1.31.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e13a8e6cf9bbb97fe5e9093ea3dcec4fb01f79b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-22 08:51:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-22 08:51:57.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.20.0","message":"update constraint for ironic-lib to new release 2.20.0\n\nChange-Id: Iddf926d52b61527ea2438691192ca26120136587\nmeta:version: 2.20.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Commit: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Change-Id: I5a152eaf2d0a61bc66b4b1de1ea7bc0a0bfa8133\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/617511a2798e62394550b48461849f4b2b799477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/617511a2798e62394550b48461849f4b2b799477"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
