)]}'
{"id":"openstack%2Frequirements~678078","triplet_id":"openstack%2Frequirements~master~I39ed2c57015fcd291868e20fe2f3b657aa8f4eb3","project":"openstack/requirements","branch":"master","topic":"bug/1840551","hashtags":[],"change_id":"I39ed2c57015fcd291868e20fe2f3b657aa8f4eb3","subject":"Blacklist eventlet 0.25.0","status":"MERGED","created":"2019-08-22 17:55:38.000000000","updated":"2019-08-27 23:33:30.000000000","submitted":"2019-08-27 23:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678078-1566948690663-1d8aaa17","meta_rev_id":"f40ca74768902706e3849d9f3c58d529106052f4","_number":678078,"virtual_id_number":678078,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-27 23:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-26 15:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-27 20:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-08-27 20:42:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 15:47:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-27 20:42:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-27 23:31:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf71292c4e3ca2849593039794b456ea0c638cc3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-22 17:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0015f64236ad953fc257179ff535b09abd6d231e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:39:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a2f485c4504e8fbaf70f142cd8ad8f : SUCCESS in 4m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5ada547f38fc472eb1bb98590a4b7bd7 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1488917c51254bf69bc3e18539a0bc39 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5614a0ddb5eb47b4933aa98c23d4045c : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/539db395eb9d44278e0d2d52c71a8440 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27fb65d576ce414397871667d4d8b0b5 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b838c353fe1045ef83f4418857ff8b6d : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2157ced5ab049bba04da51bb9091b14 : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7ad90423557b49df89a6bf60443b0be5 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49c072af68674f96bf51d9642de17e9c : SUCCESS in 9m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f2ea02f8b2d04476835e54aed639d0ce : SUCCESS in 11m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e5bb3dbe7e34f3ca656905bb9e3f6ac : SUCCESS in 10m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d22a99b8d09f4ab2afdc88e82050366c : SUCCESS in 5m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/384318c098c1472898ea56f99b8950bb : SUCCESS in 5m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3aa2acdb927a490794e6d0570684c4ff : SUCCESS in 10m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63175883d1c64ed3a6bb831d29350e9d : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0fc34640e7aa4d63bbe5435560e02625 : SUCCESS in 8m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bc05bc5d24f6455ea84b041aa0ac2776 : SUCCESS in 34m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ec18b0f02b548618845edb7f55f16bb : SUCCESS in 33m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/87d07a17fe994ff98957021117dfcf92 : SUCCESS in 42m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37abadf3efdc432fb9ca45e895191062 : SUCCESS in 25m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/afe78db20f7544109f590dc552510374 : SUCCESS in 18m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/99737d662a16410995d1a58b68ee1975 : SUCCESS in 15m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/298b734de0b84a949101fffa4d4108a8 : SUCCESS in 14m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b613d9d45855457cb1ebecf180a02aff : SUCCESS in 22m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6a0b2a623db4bbab0bf12bc447a458f : SUCCESS in 5m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3836ba52ff034ba891fc8d4af49f2459 : FAILURE in 12m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/590df1b33d6e4947bbdb3459ed8241d7 : SUCCESS in 11m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0d2687011d454f84a6e4585b049d3614 : SUCCESS in 51m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/073ab3b5cd9540a0ace9302239b9d48c : SUCCESS in 1h 16m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/42ebe7d9746c49b1a30df716c0d2fd5a : SUCCESS in 1h 45m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/559e2bc782b0444bbdc54e8d90e27111 : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"366d406d94401bb4cad528f25e6545e84bbadf17","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-22 21:43:09.000000000","message":"Patch Set 1:\n\nrecheck\n\nThis passed cross-nova-py37, so it works; just need to give swift a kick in the zuul.","accounts_in_message":[],"_revision_number":1},{"id":"a0fd713463289fdf7e1fff7434f00147efdc3d22","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-22 22:37:56.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027d rather this be two separate patches, one to update eventlet and one to change kombu","accounts_in_message":[],"_revision_number":1},{"id":"85d2458f2ad91784e4aa7a1c5c815daf22862711","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:15:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfbcfdfe92241689c9610965bf8c11d : SUCCESS in 3m 37s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/80a6087f98d847ee8056227fc2dcba41 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5060b62c53c4d4e97c8c86d13b64769 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70c4caad4d0d45bbb545b32464749d0a : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d7ab3a0d084fcb804291d1c18f32ed : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6fcbbc5b6e64ffe9cf3d107169400cf : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6aedc6c8fc81431d8abe43d225d665fa : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08c7c90b4d744523846ee7c101a83bc0 : SUCCESS in 6m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bb31b0794fcd41c8a103d0c0cf71ba36 : SUCCESS in 12m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dfa09b005fa349889d702cb076739d60 : SUCCESS in 9m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f6cc8fe12ce644b18af2c74e84fdd7a9 : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9fe6b2a070934bde83612bb5a9cad3c7 : SUCCESS in 11m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/66830f8ed4aa4be3929dec1b734f8aa8 : SUCCESS in 5m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5236a2fe6c674e9398a85cccf511fae9 : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3350a58abddb4ac8a207e4ad88ee6a22 : SUCCESS in 10m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b36416e383ed428ba6bf38e7570f0109 : SUCCESS in 8m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f86fd197af943c493be4a999fc418ee : SUCCESS in 8m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7c07b6780d2b4d6cae4b06305f1116f1 : SUCCESS in 31m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5df44aa93e0e47b186e8d33dc6db6c40 : SUCCESS in 35m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f8ba8508b6c94a5ea00574f5c2c2769c : SUCCESS in 22m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6953cf5c053647cca8a7538cb56bbba3 : SUCCESS in 23m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d81c4631e67d4be3ab39faa2f4f19bb8 : SUCCESS in 14m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7fe2df8c36842488913fac83d1386d1 : SUCCESS in 11m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/17f5e511e84d4586a70357fa8166bf92 : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb5eda88512c441bacd3c085b6a405a6 : SUCCESS in 19m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da20a30b1fa04abe8e6443bd3e0b9846 : SUCCESS in 5m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/507f36eff4734fd0a7a1cdf484903581 : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8fc3e158c3694d0c89aa7bee79fb1123 : SUCCESS in 8m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/788f278a93174bd8bb593f4ac0acff05 : SUCCESS in 56m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7f1fd2d34e2425badb73dd62035cbfb : SUCCESS in 1h 16m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c2168e73dbc24cb0aae0b3e4baa2b8bd : SUCCESS in 1h 57m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c3d3d84a9f846c08efa994d82f75be7 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9d4ec1f2664d01413924b586aefa54e8020425d0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-23 14:44:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"580f14d809179426835a161370bfc1277dcede0e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-23 14:44:43.000000000","message":"Patch Set 2:\n\n\u003e I\u0027d rather this be two separate patches, one to update eventlet and one to change kombu\n\nDone: eventlet here, kombu in https://review.opendev.org/#/c/678245/","accounts_in_message":[],"_revision_number":2},{"id":"2ba0b0488703d26c986cb45d84ba704d76c35989","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 14:45:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6717583481b660d013c5715a6852a34f172a0635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:18:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c3ef8aae684b7a8e3547a3a96c2c96 : SUCCESS in 4m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0a7ba3db21aa4dbba1977d0cf0dcc3d7 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bfe44eb0f3646c9a774e268481c2ead : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d98496c681fb464497ddbc9e35e33a77 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/121dfe519ef447b8a4f7a14529f78ee8 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b26998304087467ea2edabe4c6b44cbf : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc5e0e4d27754498ace046116eafaa6f : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac0f2a5c609d487fbfc9d85d50408c73 : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fad5fa1591c74c0bb137eebd8dedfbfc : SUCCESS in 8m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/748472133bb044708a8b7ba93aa6386f : SUCCESS in 7m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/54d835ec013248d983d334df6775f774 : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f4fa965b87b248508bc212e9f66f9bf9 : SUCCESS in 8m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8ce320feb9084191b24d9a97981454c0 : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dfcfd680b2f14deeb89dc9819ddc2327 : SUCCESS in 6m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5a906c111e79407d83d1f08360f24b0e : SUCCESS in 10m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b06689ff459c4ebfb1cd41886799da33 : SUCCESS in 7m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08f76fcfb577437a86b8fedf11b8a805 : SUCCESS in 7m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b1da5a52c4c04dd9ad4672eaaaad5e7d : SUCCESS in 32m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8ff5d8f118024ceabfa8dd065e376442 : SUCCESS in 31m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4b80cc8ee35244148e7ad06bd142233e : SUCCESS in 24m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2aa7ff245e749afabeac9788e572844 : SUCCESS in 27m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8dfed99031ce4ee6b4dccaaf0bf12915 : SUCCESS in 14m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2f45e77bd9914b3491a83db5adfc97eb : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/33e260559e1348e8b3187418cbcf0041 : SUCCESS in 14m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/daf348011caf461cb9d0198c4d44df60 : SUCCESS in 21m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d69d41efe9634486bb27a6fac22d0028 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/486ea897f44f4424a831548a50f86764 : SUCCESS in 8m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abff1b23e8c845339b1b609f8a90a766 : SUCCESS in 8m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/797c2593b5e84bfe93370e5cfe8fc295 : SUCCESS in 1h 00m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a1680865cd34d2baadf6f8d34aca0a2 : SUCCESS in 1h 16m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff68bb6d412c431ab7d98adccf2527e8 : SUCCESS in 1h 47m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b5d90fea829403d990b4b050a7ed58d : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":2},{"id":"57acdab4f5afd1057ea7c9dcc09eb2399185369b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-26 14:35:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cc39d55ddea83dcf020f2ac743db95abca59f7f1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-26 15:21:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7cc0b8a0925831ecd3b7775814725a42fd363353","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-26 15:22:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3eaee4bfd8266d98f654801d52f1b2216abcc95f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-26 15:47:16.000000000","message":"Patch Set 3: Code-Review+2\n\nwfm","accounts_in_message":[],"_revision_number":3},{"id":"72b1d868b39d1c50d389b63b19945a463e694a0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:22:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7857069612946b5aea622ddb3701d5b : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9123b6c9614d41109637338680331018 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc20ac3e6de4524a2019f08a68cdcc7 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59b80b969db742a19199b7f6c1a9f427 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c48e919e21b4104bfd12c2227f4a8e2 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90e03d78d9604c60aa123922e0b30d4c : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba2e16c04ece4116bca87703a773acc0 : SUCCESS in 7m 15s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fb20a808084a43229aee031247805785 : SUCCESS in 1h 09m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1c466ac9ddc49b9a0bb916f293c66c2 : SUCCESS in 1h 12m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f918e7e4c6f4886aabb0ddc5efa0f76 : SUCCESS in 1h 47m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3dfb0646297545d88c2cf35bab7a88ef : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":3},{"id":"76d6d60a072b78f8223f336e4374bc94cd2c3e0f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-27 20:42:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"49aa742ac14abc2a8e5b5fc512382ca90f4c7e6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 20:42:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d075bcc35c035b7fa347660281946b005a31fb9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 23:31:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e14cb320e5413ebdd246af4a68f717 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7575531c8c946f5a234ba4315dfa389 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f93a8e2bb0f240e8929b2058c87deb6f : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be60e557b0744fdd95f22b37fdda78e8 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecc22a16f5114ac595e3117d77460453 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bba848aa702b4941bef2f4fdf17c50aa : SUCCESS in 3m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3496400e48274ae2aa57efe01368b290 : SUCCESS in 1h 16m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a841071116824dcda54dd24b17ff15e0 : SUCCESS in 1h 11m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f1d0b3e21674889843b36dc8a58ad91 : SUCCESS in 1h 50m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/613519e0870d475f83d19d02666fbe4b : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":3},{"id":"015d0437e97728a97adc8bf478e163da8a92dc8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 23:31:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"21e259ab78cb16b6aa647541809a1578d0255b45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 23:33:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34672e859ae14fa59d4f07cf311146bc : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb6baea86732ad4d73f6734cd1db5495aeff99ce","revisions":{"ae57c7afe7b0631e32bbf79d7f6ce66906446a69":{"kind":"REWORK","_number":1,"created":"2019-08-22 17:55:38.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/78/678078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/678078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/678078/1"}}},"commit":{"parents":[{"commit":"28875ef9ca02f1c62b6af10ce04907a926e44bbf","subject":"update constraint for mistral-lib to new release 1.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28875ef9ca02f1c62b6af10ce04907a926e44bbf"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-22 17:52:24.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-22 17:52:24.000000000","tz":-300},"subject":"Uncap kombu, blacklist eventlet 0.25.0","message":"Uncap kombu, blacklist eventlet 0.25.0\n\nThe related bug was fixed in eventlet 0.25.1, so we can now uncap kombu,\nreverting [1], and instead blacklist eventlet 0.25.0.\n\n[1] https://review.opendev.org/#/c/677070/\n\nChange-Id: I39ed2c57015fcd291868e20fe2f3b657aa8f4eb3\nRelated-Bug: #1840551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae57c7afe7b0631e32bbf79d7f6ce66906446a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae57c7afe7b0631e32bbf79d7f6ce66906446a69"}]},"branch":"refs/heads/master"},"0d09d58704a403a529a3b6bfbb1b0d478572ccd9":{"kind":"REWORK","_number":2,"created":"2019-08-23 14:44:01.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/78/678078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/678078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/678078/2"}}},"commit":{"parents":[{"commit":"28875ef9ca02f1c62b6af10ce04907a926e44bbf","subject":"update constraint for mistral-lib to new release 1.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28875ef9ca02f1c62b6af10ce04907a926e44bbf"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-22 17:52:24.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-23 14:40:52.000000000","tz":-300},"subject":"Blacklist eventlet 0.25.0","message":"Blacklist eventlet 0.25.0\n\nThe related bug was fixed in eventlet 0.25.1, so we can now uncap kombu,\nreverting [1], and instead blacklist eventlet 0.25.0. The latter must be\ndone first, which is what this commit does.\n\n[1] https://review.opendev.org/#/c/677070/\n\nChange-Id: I39ed2c57015fcd291868e20fe2f3b657aa8f4eb3\nRelated-Bug: #1840551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d09d58704a403a529a3b6bfbb1b0d478572ccd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d09d58704a403a529a3b6bfbb1b0d478572ccd9"}]},"branch":"refs/heads/master"},"fb6baea86732ad4d73f6734cd1db5495aeff99ce":{"kind":"REWORK","_number":3,"created":"2019-08-26 15:21:35.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/78/678078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/678078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/678078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/678078/3"}}},"commit":{"parents":[{"commit":"9dcdbd62dbcfc53d62c4cc86b2d3fa86cb371e13","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcdbd62dbcfc53d62c4cc86b2d3fa86cb371e13"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-22 17:52:24.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-26 15:21:20.000000000","tz":-300},"subject":"Blacklist eventlet 0.25.0","message":"Blacklist eventlet 0.25.0\n\nThe related bug was fixed in eventlet 0.25.1, so we can now uncap kombu,\nreverting [1], and instead blacklist eventlet 0.25.0. The latter must be\ndone first, which is what this commit does.\n\n[1] https://review.opendev.org/#/c/677070/\n\nChange-Id: I39ed2c57015fcd291868e20fe2f3b657aa8f4eb3\nRelated-Bug: #1840551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb6baea86732ad4d73f6734cd1db5495aeff99ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb6baea86732ad4d73f6734cd1db5495aeff99ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
