)]}'
{"id":"openstack%2Frequirements~678115","triplet_id":"openstack%2Frequirements~master~I806911f090c1bf5f8cd99e03d9dd9334c5387147","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I806911f090c1bf5f8cd99e03d9dd9334c5387147","subject":"update constraint for tosca-parser to new release 1.5.0","status":"MERGED","created":"2019-08-22 23:58:09.000000000","updated":"2019-08-24 05:13:35.000000000","submitted":"2019-08-24 05:12:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678115-1566623528204-1b5df26a","meta_rev_id":"2fd6f8c8b6e6a51c6131c9281c857a6104ef3a2b","_number":678115,"virtual_id_number":678115,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-24 05:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-23 02:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-23 02:10:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 02:10:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-24 05:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7c3fca5dcba7bea4a536da973354cde82f7d2b6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-22 23:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"490d61e1681eb33ef3e0b850bfff7c5e0d44aed0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 01:34:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e6be4efe1249478e81f008a4641ba8 : SUCCESS in 4m 48s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0bbdf482352149f6a640a0bf64ac3cab : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bf77ce6b5b4886ab59fa8d23c50a1e : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f49e6e0692954c1cbd2f373fa02e8c98 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c0332855f03431fafa97c09efcaa0ef : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/142f2a1a60334d89a262ed73d4709941 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/475d6c4aec594a9286e0a6dabd7928bb : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3370c362cf0479d894eae429a539b0b : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f9bb0189f29c417ba7b7654e885836d7 : SUCCESS in 8m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c08397907264b97b3fd44d0102c5ff5 : SUCCESS in 8m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ac01ec1315fb4a7ab88625c8335fcfe0 : SUCCESS in 10m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ea2576ba38b4cdea3548ab3524f685d : SUCCESS in 8m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/01796391ed2540ea8df8ea18501041c8 : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13f85712edb2439d82cd08fbd61cc588 : SUCCESS in 5m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6d3d442bd6f142ca8445199fc1a1a4c9 : SUCCESS in 11m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65521c7911b84b20bfa0b23078c0948f : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1af4aa5f792494089e512a004dbf43f : SUCCESS in 7m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8610a58005d0499e978fb9d6f942dd02 : TIMED_OUT in 41m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5a7e6c9aaef8479dad6073b3bb985b9a : SUCCESS in 34m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1234d9756a4749f38abc990caf06e883 : SUCCESS in 23m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/40326c7025314feeb8ca81597d730f05 : SUCCESS in 20m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c37983d27f6d4d45810a16f5774f1b83 : SUCCESS in 13m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1a0b206d635540cdb0c9961a9036e1d3 : SUCCESS in 18m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b8a0776510c54fdd91865cf8864afeca : SUCCESS in 18m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3c2cdd87ea9425a8cfa7a9fb8c81eb4 : SUCCESS in 19m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/184932003f4e4fe5b992aa51196e9e7c : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/acb23fe2d9ea4d078b968ed94e578e56 : SUCCESS in 10m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96df06421b104e33b92d42b401d9640e : SUCCESS in 9m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/15662d03da33458796a2b7bbf636e624 : SUCCESS in 51m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4054925c26aa40a5973932efb4106f41 : SUCCESS in 1h 12m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc6c43e2740b4936b549cc7f2bef81d4 : SUCCESS in 1h 24m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/334e8ff5a9de43e48af9e0cbf43cefec : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"eb12261c89892de6e8d0c756a2ec8787a80865a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 02:10:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7374725ead79666cb6e78b4dd4dd17c67a4d5f28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:11:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd07d73d0a2c43868fbd56062e6bd698 : SUCCESS in 4m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/19871d0a6f1440f4968958e4ca32516f : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102d7be34b094ae8aa7407bfc3b2cfae : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13499746bd0d49d390e346566a7dfa75 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/096157cea5be4d7ca074ee5b6e97d4bb : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/296d7701603a409780338658b29b83d0 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6faaeacacea042859d5713c7bd419acc : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcc8b13d50dc4ab6931ec2673858ced2 : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/37b5c99d4183460d9a1d65608f316bf6 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0948ccc74c63486aa59b138b76002fab : SUCCESS in 8m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/988c47477d3244d1ae0f0a3919886c63 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/95c8d70296394c3fa7f02fe29e6fe2cc : SUCCESS in 10m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6f50b8a5d4104e87991ffa6651df55d7 : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d0f636289c794844924fbaac90fccf05 : SUCCESS in 6m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bddb8dbe56b74b5b83f749a0fbe713f7 : SUCCESS in 13m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0eb5b3287d2c405f9e32c405036ffab5 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d504a6bafc94543ab71272faea15075 : SUCCESS in 7m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/30c77471bd3c4eb69ef6a73b3fcb9fe7 : SUCCESS in 30m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c02f3dd16a0f41c8a8da61b7e27a40fb : SUCCESS in 31m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2059d9c9e05d4a54879bb94977e9b0cb : SUCCESS in 22m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/88ef5ff3ebd3407285bfa25b046aa9cb : SUCCESS in 21m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0abcad55649941fd888e1cb64c12f128 : SUCCESS in 14m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5f947d17253846c6ab47fb662d18eb87 : SUCCESS in 13m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e558359a69104676802880f74b334ad4 : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e917a6c17be04aa88264247a43a62f37 : SUCCESS in 21m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28d5bee265bf43b9a08ed73f35f2a65d : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/28a9a8e3f41944a7b7caf06906f5b8bf : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/327282064dcc4f11ae1f5e724549eb78 : SUCCESS in 8m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/b47f656f53864a67ba6fb4d3af6401ac : SUCCESS in 50m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b593564e652042ec9b39ba83768eac7b : SUCCESS in 1h 14m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f2489536f7274bf886deda4a6b1e4a2e : SUCCESS in 1h 55m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/803982e276e843d4995a3f07b93a9303 : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bc45b7f19d91d732d45dd822dbfde7e0781c75b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:12:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac2736cf02bc34e976be7f551290173147036afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 06:53:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae18ed431d4415a866acddaff1b6d8c : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb4a412f792e4b6d8ceb01ff85410fbe : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6072539ebcd54cdd8477217ef23c1c13 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76aa4e87e8744ae9ad6fb976681bef37 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/914d20681b1e428789d33ff1a1f9dfee : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/682abcf6318e4f85be6a43013d069fa3 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e1dbc8257f94bf19a3886df6be97efb : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0625cb55c9db4fbb98923f4b340adc5f : SUCCESS in 6m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f9b43d57f184e5e8a7bee819abfe60f : SUCCESS in 7m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/97275c8d0fb24a4c85c9b96c64fa8d18 : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5afb9124f40f4e03928ded7f56ad9c3e : SUCCESS in 8m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/371452ae314444feafec8f65effb84d7 : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec1467fee7b84f208a4c4aaf65eb3949 : SUCCESS in 4m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/14ba18cbbca94302bdf2b7248c34f25a : SUCCESS in 10m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c8121978f99d405f80dceeba725d297c : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6ed3850251b45418457a9f4ce0c2ab5 : SUCCESS in 7m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/95877573fd88471b94c20b36cdafb8fe : SUCCESS in 28m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6968b4b8f16b4b56a7326b4ad198fcab : TIMED_OUT in 41m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7b83a98ee1864dedbf9bcf71cd9f1d73 : SUCCESS in 25m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b87ca2aed98f44759ed4f918202c0e2d : SUCCESS in 21m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/30d7498963d74904bb866e949ff983c6 : SUCCESS in 15m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ddbd7e01f9f417990f578f07d604491 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bc683f49894d485581a78847d254ddf3 : SUCCESS in 12m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2e759f7e9d047219781207ddd7de305 : SUCCESS in 20m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b2a6977d8c24f68af3aa1df03a1c263 : SUCCESS in 5m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/223e5bf5bea440aeb966fb8ee345403e : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f6b279cdef54b7fa3c42271e743d165 : SUCCESS in 8m 41s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d6f8b4f6da564eb281ef46766f5b4ad0 : SUCCESS in 57m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99bd5c93a9ce4a1cba16a5085c6a06e4 : SUCCESS in 1h 09m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b3f7641a2f73482facebfd867869d05d : SUCCESS in 1h 33m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/70b5c4d895094e91ab520c403c2c9c18 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a0020aa56ed4de5924b91f04050911d25d14d27b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 14:05:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06603b9034104b048321efeb49c927242a2d6a0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 16:17:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d75737f01814ca4aea20cebbc9ba7e5 : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c772a9618b3b4d29ab828bb9305b2e28 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40694fa32b11462f894bd4b2d9e36f85 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38759e7e181941b684ba0d2bcb9ded07 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f723203436543288b448bc3235212d6 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e85b6c712da4317acc3e91d3cab29fa : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9da4ec568924405a91f3a641ede04a7 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03726e354c0d410b94dac38eb556e63b : SUCCESS in 4m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/519f32998893401b92f4408901375efb : SUCCESS in 8m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f18e6ac245be4538bfb3733a074d3da7 : SUCCESS in 10m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5918e189b57e4429a710f0c0b82d8e05 : SUCCESS in 10m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d756181db5d4482eb3686e880b273a53 : SUCCESS in 9m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3538658a38714be8bd3db5f429b20637 : SUCCESS in 4m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1252c7471ed448638f3b857c8a156be7 : SUCCESS in 5m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3b613523cfc24f6bad55094d781935f1 : SUCCESS in 12m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a236d21892442568422c81fb16790ee : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8731fcd46b741608b38005b5ee0f29f : SUCCESS in 10m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b860413d74fc487e823738f057971445 : SUCCESS in 31m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/98483736e9ba4908901d6cc0be404b25 : TIMED_OUT in 41m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1b46b71d28724c35b9a4f19bfffe4388 : SUCCESS in 41m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4491a67e6161441f95f76e2bb77037dd : SUCCESS in 22m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8c29046e8873417fa6cdc51a563ce170 : SUCCESS in 12m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09f42c818b7c47d181c563db6fb8a349 : SUCCESS in 14m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/150302a43fa14f95a90d9138b9e7c402 : SUCCESS in 25m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aff6cfa0737b4632813058ad26ecc61d : SUCCESS in 19m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4de6fd877e1a4cdb93145974289399f2 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d8fd31d166e8459cb8006f835fc813b7 : SUCCESS in 9m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a56f9b02b17347d6906572434502cbbe : SUCCESS in 10m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/caaa1bfc792b44419abe19373aa5103a : SUCCESS in 57m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f9e46df8e924a06b040214dc13de7a7 : SUCCESS in 1h 14m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba1228f9491f4a3db56beb9a02d11ae9 : SUCCESS in 1h 36m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/158420605a3a4401b72b83559df42946 : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b04ece232b2d03e091c23385444ec6e429dbe572","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 16:26:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d4bcb523d9c227d995c9eed0e128a21894bebf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 18:58:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287b14d2edad4a4bbcbef50967d914e9 : SUCCESS in 3m 47s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a16c12c975e6411eb77a2e5f0845d06b : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9100fbf1204f3fa350b6505e10b40b : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/065c8f0fa4ce490d816e0f8a995a1707 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b32bbb04f0f42b7bec594d793442195 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9e5c9e2d2004a8db4e59eb318d25005 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6391a898ee554555a9f8386648aa364c : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36baa509129d4a5fa19c4ea01b0c30cc : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b2bbb2f714db4e0483ce086a2a84b70d : SUCCESS in 8m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cbfd2b3e18dc4ee4a07959c473a1c829 : SUCCESS in 7m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/055889c213b94e17b8b48310a64d0af0 : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2bee4306a6614fc08564934f36faa527 : SUCCESS in 9m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b24a2450c7c8453188bc6a52619ac659 : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/916bc8e27e9e4176afbf1b197ec12b94 : SUCCESS in 5m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/716c0a269ec344e0a7f8ce33f0e91697 : SUCCESS in 12m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5c46aad832f3465199945642d2020055 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f774aec11c094481a48c65a0fa56be6f : SUCCESS in 8m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d7f2f7e9256649b78dfbb6da4300688c : SUCCESS in 34m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a2f9deeda8e447b8f52648d6e4c63c0 : SUCCESS in 30m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d635d60cdc604af88f730970480f4d4a : SUCCESS in 25m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6860d0d0cce040f1b631389a48b00164 : SUCCESS in 22m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/72c62b16f74246e18703905c579eeeee : SUCCESS in 22m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/54a8ee179d154794b744373db5656b45 : SUCCESS in 16m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aca28d71b7704ea88960c7ae5ecfec8d : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d25f6406625646959d5a302b6b1135d2 : SUCCESS in 20m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d3d89894c1442b1bac39d3ccc16c7b1 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/31eabe2da1c744ff9c86a00bcbaeef9b : SUCCESS in 11m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3364e74f0e0e44299c3ae7e01a670592 : SUCCESS in 8m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/85df457f61c84894a2a58ca07f82c1d8 : SUCCESS in 1h 07m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af64010863474bf39808816e166eafdd : SUCCESS in 1h 16m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/219eafcdb579421da4e18894e23af651 : SUCCESS in 1h 30m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/32b87730db9e4a8ca1271c971a7e2a17 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dadc7784d3c253bf90777bb51fb74cf28a94aa1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 18:58:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be7739356e51f9d4dad67fb4984e586daa7fdc2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:13:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add9f4dc10a54a16847876bd86853cfa : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d4611674294d379423e07c58cd1f47 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2956a7e875c4a5c8c67d8c4056c5dad : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e05458adf6c43af9f63501e34c74ebc : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/495c09c1245f4cda912fca5d543694df : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a2cdaa530174dd191a6e58a45bc523e : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e4f3c6282f3460ebba76673b46ed2a5 : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/95170b1a9bd845098a90f76646cc1d1c : SUCCESS in 9m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4cefd8f3aca8409db5419c79ff9e9e88 : SUCCESS in 7m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/53d95ed131474c598055e6cbcbb33a31 : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a5d17e360b9943b291dcdf1655bd5084 : SUCCESS in 9m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9d64bb7c9414488a84624aac5661755e : SUCCESS in 5m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/addc0b37012047bca7cb92740e579104 : SUCCESS in 6m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c903b90f683d47ac99bcbc3b4fa3635e : SUCCESS in 11m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d72d0a25e88f4a319565c7bc971552bc : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6475e9eef12d4d598f798c768cbb1f67 : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b2253e57c96c49f6a730d7b2e59ca2e2 : SUCCESS in 30m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a3a4e7afebbb4672a4e05854fd5c8838 : SUCCESS in 28m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0742523a1ca24946a75a05edb36e03ab : SUCCESS in 24m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d01354bbe0484d71ab448e1ff91618ca : SUCCESS in 26m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/55b27e218abc4dee9f6cfe5a66e9daba : SUCCESS in 16m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec0c07bda88842568c95f0b5703a076a : FAILURE in 26m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a24d2f95adab4fea8e07534142d343ee : SUCCESS in 14m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd12c079f021411da597bf93a03fb226 : SUCCESS in 25m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/861c2bb88ed74dfd940f09e17c68c239 : SUCCESS in 6m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1672c4fcb2294233a30dc9e2acf80177 : SUCCESS in 8m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/439b8eab0c7e4330a65367a23f0152d3 : SUCCESS in 9m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2a6c620c0ab7438297e076a3d279ec02 : SUCCESS in 1h 19m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/430b503ff3614c129ca15e7e0e27b081 : SUCCESS in 1h 14m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e409712c524f4cf6bf4f542ff6f8c0fa : TIMED_OUT in 2h 07m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6fd0fe05b43e42f29303cd6d44cbc606 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c5e55a8c2216b776055fc8831674fd304049b1b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-23 21:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c63da66029cb1b2cad9a1828fa79726b510d9867","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 23:10:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4afc1752190940f0a70cc57e3f4b1c28 : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2848800b757b4675aa4baa8ff5f71f57 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c09e0c00b14f0db61426252e1c1c66 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d414f396a064258ab25b36d9d2d2c3b : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b48e95515eb4b2894c23c2b09afe678 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/484a60b778dd42feb2122494a3621475 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ef06d879da1498596ea1775cd782c58 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e5dc07c19a34577be473ad18dfe3114 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ca969399edee447bbc7768c9ddc122fc : SUCCESS in 6m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/03d18123329b41a4b9e91e1e27940d48 : SUCCESS in 8m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5b5460c8e7234e629336e016dc58d6d2 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/373b05e2cd8949ad8b2f46a8de810fdb : SUCCESS in 9m 57s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2645e9cdfd1f4acdb13fc77745fa270c : SUCCESS in 7m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40b3c14a73d841139676e792eae9967e : SUCCESS in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2ebb4f7588e9477e9431997a1f2427cb : SUCCESS in 11m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/691cfc42f2684d94ba6a5efc2dca7bff : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/806c9e44cbb446e5b409c50603754c7c : SUCCESS in 9m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b8bb5d5d1460456ca4698b4ab05d5316 : SUCCESS in 33m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b6767d9ed6764cd897e6edc952c03463 : SUCCESS in 28m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cc2bc1efadf74913bee43e2f037ddd56 : SUCCESS in 23m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/966d37ddbc4b470c9644a6d90d57c2f6 : SUCCESS in 27m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5fdb2551bc3c4c9cb2f5b01ad3f33ca3 : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/959b94bc5b6848ba80ffee6271a0a81d : SUCCESS in 13m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/55aa920cca304d06b4cc03f582f036fd : SUCCESS in 14m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70001c19138a45a1874897878d698637 : SUCCESS in 20m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f1f15ba38bc4e6d99ce3c60c8282341 : SUCCESS in 7m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c4229c7082e34754a8d479fc1ad31d41 : SUCCESS in 8m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/afe79de9652545ccbe4a863f01cd26b0 : SUCCESS in 8m 16s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/44fa9a196d0c49268b6248f035906663 : SUCCESS in 1h 24m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcf69161c734402fa3ad06df8f7b5d58 : SUCCESS in 1h 18m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e33c9200291a43d1b77fbc2b43744bd8 : SUCCESS in 1h 39m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d990202edd2d4f33a7dec5ecee7c6491 : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c60aab89e09ac9a115e22edcfdfa2f6e1df87eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 23:12:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"945df501177c11ebbb817d089f088a43b7726fa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 00:58:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd3acce234542cba196603af3f67cd1 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/307eb6a88b9a4e828b04f1f2db5c0b38 : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ceb73cf9d7d8451e840e80f61d406d53 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93311259014b4dc68b9504905820804b : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a5dc635d85a4de5bc9403289611b830 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ded5f6b6b579426081207dcaf5184f26 : SUCCESS in 6m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d1ae78137b34b1e85081b50840359d1 : SUCCESS in 4m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8d29503badf54b50b13a9bef305994ea : SUCCESS in 18m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea351f9d083b4c10981baa5047b8ca5f : SUCCESS in 9m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/299753113625467a913475a3122db81f : SUCCESS in 14m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/63a45e07d90b42f796181f3ab68bef06 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e043a5b735f5482bbf33facea695fe46 : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a83e60e4d014f5db095ec59c069314c : SUCCESS in 6m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4b30e43b4be3403b8df65e4fd3f920da : SUCCESS in 10m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d4bbba80e00b4f49988baa0499bd5339 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28bbc51ac9ad4a6d88ca22dbcb5e0233 : SUCCESS in 9m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4493c897fbf84f5e85765850df3e31d7 : SUCCESS in 27m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ea8bad525aca4ef5a3c302535c9fd8a7 : SUCCESS in 30m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4ee1ed084c7c484b8b355c1b606a2417 : SUCCESS in 21m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/233d73f9a04d45938b400632abf02943 : SUCCESS in 22m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/99fe74147fc046f89a38db40ec512f28 : SUCCESS in 16m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f64663cdd744b6c9c7a9bda03ce2ca1 : SUCCESS in 16m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc0fac9a00bf432f9109db12a1fc8d4b : SUCCESS in 21m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9490376e70e440fdbef47ef8fa8cb80d : SUCCESS in 20m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/829e3c7cedf94938b3c083c3eae0d146 : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a51e3f4ac9b044c8901ebd447e4deb3e : SUCCESS in 14m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/623b2351d4ce48bb9b3288ff79894070 : SUCCESS in 8m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7b0ad16515644384830a6c66e0e6a119 : SUCCESS in 1h 19m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/644cab2ecdc44bda9304c2e6b9250d00 : FAILURE in 1h 20m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2712194666864a95a2a7bd64068468df : SUCCESS in 1h 42m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8fac011335a44522903315c60a89d664 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7df5a8b4604e833bf4a8dc2495869ec36f050b70","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-24 01:54:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aafc2659e0ba558ca09f47d8f89dfe6e13431304","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 03:29:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c04920c98f49fa97c683a176c3f461 : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cb2cb4d718854589bee0d493f23f8e34 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d853eb89fdf8448e93bade0d968b2eb7 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14cf603a392c4c8eac68282b50bf2265 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4a4812639a341c8b85c999a45829359 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d0ce871a2634fb2b47a970159fb3797 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f52eb8aa9d254502998086fc12cacd4a : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b161eb990bba4c15b86d9baedd501895 : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/909487da25e0404d9400f52a15b05aef : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/607d81d2bd364f96ada1a5e8321e9f7a : SUCCESS in 6m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2f0a0f6f999846f3bc43fb10b0ae5fb7 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e45d9af82fb48cbb90a8c3dddba9d29 : SUCCESS in 9m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4ab1cb82b24b454090cb9c12cedc5deb : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ae5b6cb782db40e291f113ad1179e17b : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/666ec24063ba4459ae9051f6ab891a40 : SUCCESS in 9m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/03748dc222804f4ba00ec48f957faa2b : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e964807d8ae42639aaa025eaafa3147 : SUCCESS in 7m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3648e51017ae46ebade88049e041cbda : SUCCESS in 30m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ccf3a2ead71c4207a87315c3b6fdf3c1 : SUCCESS in 27m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e7701e221c704016950d7ba03a1680bc : SUCCESS in 22m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d0502aadd6454268b4da34c0d8842143 : SUCCESS in 20m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a7950c8c8e4f43b1b8601479fb3b7a65 : SUCCESS in 12m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/91efc799ece6426593f9b34c45942a26 : SUCCESS in 12m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ceef22b7f1bc4e89a1fab4c41f9cc531 : SUCCESS in 15m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25b072e61f994c56899e4a23de085a1b : SUCCESS in 20m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e85ec96e549941248973826be5788f15 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f9d3eee264ae4ea490e7d3f9704bd5ca : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ccd8914804b4f289c527e06fc5073e7 : SUCCESS in 7m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/22abeff131c84827b5c223ac3a5efdc3 : SUCCESS in 56m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ca1d505a6494dfd87746a27c2bb9fb9 : SUCCESS in 1h 14m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d88325eb60e45568dcd5f6b4b779870 : SUCCESS in 1h 30m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7eb32f698b8943d9ba09c4f3da792f42 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"94891d38106a3118a183fe6c9809b48d2e7a1857","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 03:29:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b81911ff74407baaf1b0aa11397844c5cac10a22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 05:12:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3f7d72f651426a8dc6931b0b2470ac : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3c91fdf31846b9b2e47d739fc3ed27 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/828dc2631a8a49228f14555834471059 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97ef7a49c14e47eb95d2aa8635414e7d : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6513d663b45646aca8da53b4c6d5adf1 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e11f31618ac434ab9602e35753db9da : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86fe01a87cb1498fbda4b118eace4e55 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f477bda8080f4b498d7ffdb85fa62aef : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ccd65c5a677e4ab3a561fda990d1860d : SUCCESS in 7m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ddfb1e80e18d48bd82559c22655777c4 : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e9e7c60141844a538c3fcb2e5aab9341 : SUCCESS in 9m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a2f2a71c3ed6452f9885914a74b34f1b : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00357d78d2994ab2ab2865b61ac0fce3 : SUCCESS in 4m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d163b52b70704d199117287c3cf2bf72 : SUCCESS in 9m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/684e03773dfd4bdd9f6b6865a404cb30 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/417c1b8277ec482ab2f5d41432e8e220 : SUCCESS in 8m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4f063b3783184cb19d33e001f79873d2 : SUCCESS in 36m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd5c33158f2d4fb7a21f6b58e130dc4b : SUCCESS in 37m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8ec8b91d02c8496997ebe99752343559 : SUCCESS in 20m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1944bc0f0e3e43dd9fb8e248fe4c3fb1 : SUCCESS in 20m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/795c0857e25f4db0af54cf0187bd1551 : SUCCESS in 15m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1dd1841bb0d940e982151b3f20e94e12 : SUCCESS in 18m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9568bed1a8154f3798be78495d7fe7f5 : SUCCESS in 19m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd8ed49aef614f418b69ffe3d982d647 : SUCCESS in 19m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/253eddbca9e74ad5b889c11ba6319d86 : SUCCESS in 5m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a197b18fd97341258bb7b2996add19ea : SUCCESS in 13m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7cb35735c0fb4f53b5ac611581b7a8f2 : SUCCESS in 9m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e4033362d58840378ef30a5e953689b2 : SUCCESS in 51m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c73cd87c9b864dc4b039097b8abc1bb7 : SUCCESS in 1h 01m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5d40cb0867684e1a89cd5144d3e2107b : SUCCESS in 1h 39m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/29a248618c1243a0b42291d5f253f33f : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3b3b0328dc621322347220aad40f843d427d36c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 05:12:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d3f0de5887958080ee3ab0c84282dc190576cf4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 05:13:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f7e5617a4143e297c8d79fc24f7bf4 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c9659450a3b1bd6822781acabfb2d51b3cab7f9","revisions":{"9c9659450a3b1bd6822781acabfb2d51b3cab7f9":{"kind":"REWORK","_number":1,"created":"2019-08-22 23:58:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/678115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/678115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/678115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/678115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/678115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/678115/1"}}},"commit":{"parents":[{"commit":"fad8a25ad1475f2a7f56903864b0fcdb6356742b","subject":"Merge \"new version of network-runner used by networking-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fad8a25ad1475f2a7f56903864b0fcdb6356742b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-22 23:58:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-22 23:58:07.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 1.5.0","message":"update constraint for tosca-parser to new release 1.5.0\n\nChange-Id: I806911f090c1bf5f8cd99e03d9dd9334c5387147\nmeta:version: 1.5.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I4757970bba1f5274a870be90fefec8c57aa6ca95\nmeta:release:Code-Review+1: Rico Lin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c9659450a3b1bd6822781acabfb2d51b3cab7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c9659450a3b1bd6822781acabfb2d51b3cab7f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
