)]}'
{"id":"openstack%2Frequirements~678367","triplet_id":"openstack%2Frequirements~master~I16d96a01dbf6b5e592bf5f447e5f21f790ce8e11","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I16d96a01dbf6b5e592bf5f447e5f21f790ce8e11","subject":"Retire swauth","status":"MERGED","created":"2019-08-24 17:00:59.000000000","updated":"2019-08-26 18:48:36.000000000","submitted":"2019-08-26 18:46:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678367-1566845213175-e91fffcc","meta_rev_id":"d86e8f7fa8190566b4a8a9beebfc89543a2a9333","_number":678367,"virtual_id_number":678367,"owner":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-26 18:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-24 20:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-26 14:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-08-26 14:06:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-24 20:00:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-26 14:06:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-26 18:46:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe301f92f077c9dfeb29a9578f52d379ae73a1e4","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 17:00:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8081902f3d4efc8583e2b7f927f7fa2c1b5e3ce","author":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"date":"2019-08-24 17:26:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e2be7185e95c210af4ea35ec91b555ae2061778d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 19:18:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ff88993a3c4d049a0bcf9bcb698c1a : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8f3af8e9e7da46c7ba63f47cc5d5ebea : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a2d2c11d86a4e4a953ec754826f6fa9 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5c4496cb49a4db28c0312a560889ffa : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c540f84fbd14b2f98241cb1f50abd15 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87e1d14ab42042ff8cc7c4716ad27f56 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33f41736e5074fc29728d9b4c286df19 : SUCCESS in 3m 52s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/ac35d3460c354875b03b574335b64c12 : FAILURE in 3m 34s (non-voting)\n- requirements-integration https://zuul.opendev.org/t/openstack/build/09d788dff0604bafacba986140110330 : SUCCESS in 1h 01m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9187fc9c8a04e908f030b635eb04f79 : SUCCESS in 1h 10m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0ed5e52283e84ec3a5a9812b8e39199f : SUCCESS in 1h 43m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f71091f337684aadb406da5c98b0bbab : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8e42c3ec4cd1ebea9b4b945f318eb716e3708ec8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-24 20:00:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"68adc149e7363c707ac4e297eb392252d4dca366","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-26 14:06:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97dc0928ba06f7c59f935b09396d57195e7d3992","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 14:06:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f901dd5c77a95bb56fec0308f8664d334d00d74f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:46:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db103586e1cc4caf9a2f00508c9e899b : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408e394c69314ed19a75e5ad92d9392f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c6324fb7cd4481d82fc2276607a652a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f75d4a7e46354431b437d758c08f2250 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4284113bcec4f18bd4f1caa9211771a : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e240467084364870a8513dcb0670b224 : SUCCESS in 4m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/eda2b299726344d187b8541c69d8ebe2 : SUCCESS in 57m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17f03f20cd964f41be7dfa0ada0a9dd6 : SUCCESS in 1h 13m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d5f62cc1a87470ea1457a323b0fe005 : SUCCESS in 1h 45m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d8610b0889ac43d99a2cfe7496028bcc : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6ff676cb95baf822832199bd79e683f2c8da873e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:46:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d86e8f7fa8190566b4a8a9beebfc89543a2a9333","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:48:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae30282c6c94bbfb3e91217b28371e3 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"826bc0787e901d014139fce7d70e411d5a3acf3b","revisions":{"04d4bb363099551015444ca74a1d57031043f6b8":{"kind":"REWORK","_number":1,"created":"2019-08-24 17:00:59.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/67/678367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/678367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/678367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/678367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/678367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/678367/1"}}},"commit":{"parents":[{"commit":"7c7b321e3f1a76f35169b3d4c9662db8a6821bb2","subject":"Merge \"update constraint for tosca-parser to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c7b321e3f1a76f35169b3d4c9662db8a6821bb2"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 17:00:07.000000000","tz":120},"committer":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 17:00:58.000000000","tz":120},"subject":"Retire swauth.","message":"Retire swauth.\n\nChange-Id: I16d96a01dbf6b5e592bf5f447e5f21f790ce8e11\nSee: http://lists.openstack.org/pipermail/openstack-discuss/2019-August/008416.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04d4bb363099551015444ca74a1d57031043f6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04d4bb363099551015444ca74a1d57031043f6b8"}]},"branch":"refs/heads/master"},"826bc0787e901d014139fce7d70e411d5a3acf3b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-24 17:26:41.000000000","uploader":{"_account_id":16896,"name":"Ondřej Nový","email":"novy@ondrej.org","username":"onovy"},"ref":"refs/changes/67/678367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/678367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/678367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/678367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/678367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/678367/2"}}},"commit":{"parents":[{"commit":"7c7b321e3f1a76f35169b3d4c9662db8a6821bb2","subject":"Merge \"update constraint for tosca-parser to new release 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c7b321e3f1a76f35169b3d4c9662db8a6821bb2"}]}],"author":{"name":"Ondřej Nový","email":"ondrej.novy@firma.seznam.cz","date":"2019-08-24 17:00:07.000000000","tz":120},"committer":{"name":"Ondřej Nový","email":"novy@ondrej.org","date":"2019-08-24 17:26:38.000000000","tz":0},"subject":"Retire swauth","message":"Retire swauth\n\nChange-Id: I16d96a01dbf6b5e592bf5f447e5f21f790ce8e11\nSee: http://lists.openstack.org/pipermail/openstack-discuss/2019-August/008416.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/826bc0787e901d014139fce7d70e411d5a3acf3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/826bc0787e901d014139fce7d70e411d5a3acf3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
