)]}'
{"id":"openstack%2Frequirements~678464","triplet_id":"openstack%2Frequirements~master~I1ffb83a9cff46e0307d2ce8b2b90596e7984f8eb","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I1ffb83a9cff46e0307d2ce8b2b90596e7984f8eb","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-08-26 06:19:11.000000000","updated":"2019-08-26 14:25:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66d8e3bf3776b37fb5cfb2c692154b7fac6c9866","_number":678464,"virtual_id_number":678464,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-26 08:28:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 08:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0950659e96b1b96580f7bdd9f95ebe0172925e5c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-26 06:19:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0eb614d5aa9b8670f49b2a119d7a40093a59a7c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 08:28:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43584f710ae54187b128bcb6c0dc2dfd : SUCCESS in 4m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d8ab71604deb43cfb5e0b34e1a1ceb15 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f66b9f8dbff429b964561843a9690f7 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6102b8897514f65900f18413fa971f1 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e14054d2655b4987bf2e985cbadaedc0 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d04a4a7b8f954a48b21035ce08ecdadf : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88f27d04351b43a1bc19a6b0c07b4f95 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ecc3f33f17b4523aa37c53e95320558 : SUCCESS in 5m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1c6a7deda0ed43ef8f978d842b10a51e : FAILURE in 7m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a35992b832814303b64d034eced9fce5 : FAILURE in 6m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/23e6bc7e911047a1b73c2188f63c3017 : SUCCESS in 13m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64b2541244e94cd9a6489e8dcb005329 : SUCCESS in 8m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6f5a931de2cf493e94730c832d3fd84c : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6a91e7225a804757938db2a9cf3ec45e : SUCCESS in 4m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/12ce1894cb1f4b08bfeb3460193d8d3e : SUCCESS in 9m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8affdd4b3a2d4f6ab1d64df92f058984 : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8e547e5f76b485a9233152b7062a1c3 : SUCCESS in 8m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f986279b2f3346f581d13e424fe840bd : FAILURE in 35m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3b61b6f8d84e4a17bbc9c6a09624abc6 : TIMED_OUT in 41m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/97788df29b4d4f7bad966ff4464d7eeb : SUCCESS in 26m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c189f0953cf64f709881b6308c4d105e : SUCCESS in 21m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/86928c64ea2e463b80053e60a9223a0b : FAILURE in 12m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/933bb69c0c02411dbf9f675d883f37b8 : FAILURE in 12m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c9c7c0d025dd4f12918ef7c74a3e1021 : FAILURE in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/341364c095c148e6ab0256415a928d31 : FAILURE in 10m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a15ae120e394cc696b29c261929aef5 : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c370e5ffa3224494998f4b1f408487c8 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb8977ef35164706b9169d7d63c11653 : SUCCESS in 7m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d9b2fee96af24df29701f37447fb2da3 : SUCCESS in 56m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/439db2536e0a453bb335af0d89369922 : FAILURE in 1h 14m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff729a18915641c1b8c579983ec3d6be : SUCCESS in 1h 36m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/efa1d3afce3243b4a28e8dd2d1ea3e92 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3990f6c4f071f73a2ea5f721d37121bdb3cc27cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-26 14:25:51.000000000","message":"Abandoned\n\npaste can wait","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"626ca163ff41246b85e976ea91bce328927466a3","revisions":{"626ca163ff41246b85e976ea91bce328927466a3":{"kind":"REWORK","_number":1,"created":"2019-08-26 06:19:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/678464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/678464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/678464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/678464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/678464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/678464/1"}}},"commit":{"parents":[{"commit":"9dcdbd62dbcfc53d62c4cc86b2d3fa86cb371e13","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcdbd62dbcfc53d62c4cc86b2d3fa86cb371e13"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-26 06:19:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-26 06:19:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1ffb83a9cff46e0307d2ce8b2b90596e7984f8eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/626ca163ff41246b85e976ea91bce328927466a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/626ca163ff41246b85e976ea91bce328927466a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
