)]}'
{"id":"openstack%2Frequirements~678934","triplet_id":"openstack%2Frequirements~master~I4705d0a9771926abb3560a6204ee26169bf47f52","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4705d0a9771926abb3560a6204ee26169bf47f52","subject":"update constraint for osc-lib to new release 1.14.0","status":"MERGED","created":"2019-08-27 18:17:43.000000000","updated":"2019-08-28 21:42:41.000000000","submitted":"2019-08-28 21:40:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678934-1567028451844-1d71bfeb","meta_rev_id":"425c857514310ea22afcbc4b99338730cf1d9b68","_number":678934,"virtual_id_number":678934,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-28 21:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-28 12:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-28 17:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-28 17:30:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-28 12:48:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-28 17:30:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-28 21:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8166cfc1f2ee735ceb5a3908a7654dcbd22ad682","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-27 18:17:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3988b0cb82974e8667aa5f20f991faf19a40668a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 22:12:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4786583ea6624225ae707eb088b99944 : SUCCESS in 4m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c95175ad0acb4456ad0c2193c02968f3 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93b256768584fac9d9ec08496f81823 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2a17a81e9e940b69990c8300d9dbed5 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f9663cf255472f8e2d36574ef3f582 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/676826603f3e4afe96e69b786305c626 : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1512e02ec70456f85bc3e726b579a80 : SUCCESS in 6m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ae13a5f5fe7413abf1270eae486d94e : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f7df9581848847bc93f818cedd406078 : SUCCESS in 12m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d4396e1a98584239919e6d05a197681d : SUCCESS in 11m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/80793b2d6e25491b83454dbf78a18dd1 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2479c2e7b44e4cfd834bbed75741804d : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1a57482dbee84316bca963ac57b85454 : SUCCESS in 6m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab3c26a9593a4edb83e7ca9426fb4dfb : SUCCESS in 5m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c6f8dd632ac0452caf13be73a786804d : SUCCESS in 10m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/808b913a568b4054af55603a4f58322e : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b8de646a07f4aa5b97128057bb90afc : SUCCESS in 10m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/16763bc1f62041b2a4211048bfcaea82 : SUCCESS in 28m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/17453aea72a54d13b215f1f49d3df794 : SUCCESS in 34m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/42cad4e18b914ac9b7250d7eb52bd564 : SUCCESS in 28m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/482c268a93b940c4821a9989a6bcb073 : SUCCESS in 22m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4d6b12edf7ab469ea4ffecbe4e59074f : SUCCESS in 16m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c86c0ec1da204757845aa1fae906a660 : FAILURE in 18m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5f7e6c4c78d044bcb9768eaa4eb2fbcb : FAILURE in 29m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c15161e0974c407aae010947d049a453 : SUCCESS in 21m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52321728bb4744bbbe66f7e36f9585d1 : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bd734b923ac24af799fc595ef166b4ba : SUCCESS in 10m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/631ccc0c24114d31bf8134b3ddae8813 : SUCCESS in 8m 32s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e67a4520ff934d5b86051ae0c4dc7c92 : SUCCESS in 1h 13m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf2af8d1e71c4370928b0c33cc324cb4 : SUCCESS in 1h 21m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f90d15180a134a929b9162f117bcff1a : SUCCESS in 1h 34m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/55edc45a4e6b4178a452c78fa411c95f : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ab418a749c1306b50278ae3afbbdb61d481c95f6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-27 22:53:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d85e5d0b853649b77fcf956afd8282e14ea4dbcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 01:59:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e3fa9f1405424c94d15ea79f82da51 : SUCCESS in 3m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/30f0359112ee49789ba679f8153ae9d3 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc268a4498646a08ee481c77fc31741 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/212a4f691d424abe9c8ee44f058fa265 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e5fb425582645d395dfee344aa30ad8 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/020ebde9b4c64257be71c5fc5d7d2e9c : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76ee9b747f5e453898ae17e963ec6cdf : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5634776a83c844d786b6d83c0be73082 : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3b76cc5a80db47c7bff55a8d6c91ddf1 : SUCCESS in 8m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d00bf9c3a5854f80a86b3cabc2d5aa40 : SUCCESS in 9m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e41a8165fe97486499a44070c8dc5bc6 : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f57c0351cebd48af85d71fba2a8b884c : SUCCESS in 8m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fe3c7a9fd07942dea9ab05f8dad48136 : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1a0d633c552a49569969c88371a10edb : SUCCESS in 4m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/40fd64f875d241e1acebebed87810389 : SUCCESS in 10m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe531fe2d44d44c7b013a0a3e74bdd44 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b85c4fbf149b4ba095a358e819a430f1 : SUCCESS in 8m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e63823e8498a449da834cbbb2ae7e66c : SUCCESS in 33m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/88c0ec61c1c74ae19a101f95f96c420d : TIMED_OUT in 41m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b60cbaee439e4a5891f86f87d39d9ac0 : SUCCESS in 23m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f37df59fbab740b0b36c9acfae16d959 : SUCCESS in 27m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/13d0e8b546444e9a9b9befe4f5901402 : SUCCESS in 17m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8a68f40b409044299b01244df4932b79 : SUCCESS in 12m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0b0b8a3c8963493d97ddeb3e1c5b98c6 : SUCCESS in 17m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2276980fbc0a4e47a4ee99e5547748e2 : SUCCESS in 18m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53cdb24761054ffdbf1bf336acfa46e0 : SUCCESS in 6m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4be80d5f81f84b3bb69ad666d4bb585e : SUCCESS in 9m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/641d58d6023949a0aa306a900aad4b8c : SUCCESS in 8m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0b05a6ea63894ca197d41a4846e7bda2 : SUCCESS in 1h 09m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80a903ce1d924f5686ae6bdb753be7b0 : SUCCESS in 1h 11m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/56f61b085e324d53b182173a0e646a89 : SUCCESS in 1h 51m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fb1ef7476bf04b9db73e6f2cb86f14b3 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"de887d6ef755472a460ae24f3aefd5f0e0b1f953","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-28 10:57:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5dfba31fc69cfa9359edcaf9a719dce408b7d891","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-28 12:48:26.000000000","message":"Patch Set 1: Code-Review+2\n\nDifferent failures between rechecks, so the update looks good if we can get past gate issues.","accounts_in_message":[],"_revision_number":1},{"id":"e00bef01edb8fa16e832fe8fde738d92e0ba8dc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:34:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5421520988c4f389d6eaae2fef45da3 : SUCCESS in 5m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a49d5621c44e4572ac2ae71c0909a1f0 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e5ed8074184f31aad11a43e791cf4c : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4409cc1c3207405cb15cd0d1b36ece29 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b09acbd13614e08bb8c6142e9083d5f : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef7dd72a52c843caa82dd5c544bfc547 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14ecd4439275498c86682bdc0aabd8cb : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/205a819781554da39d854e9275a469d7 : SUCCESS in 4m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2bfb7e2577b342eeb9b3e4daa435d53d : SUCCESS in 10m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0a41b2dfacf4be09578ad6fb57689f8 : SUCCESS in 10m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/73cc3a34a0d74ddf8a710e8c2f430fc5 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4606b84dfda6481ebe1baae7d030b0b3 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/feb4ab845f744b82a0891d5109e91322 : SUCCESS in 4m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/de3730b80b62456aa6a725d3fd066e25 : SUCCESS in 7m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8dee55385d324f70851d67fa89291ea1 : SUCCESS in 13m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66752da830ce46c9b31efb974ae69e15 : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aafb0446d0a04c90b8fc0f4022a34057 : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/30c96ada974b402b9fa9ddcebf7c69d0 : SUCCESS in 32m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ba53dc6d19842b1bd779b050169f342 : SUCCESS in 30m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ccbf8517f4514e32916982949ba3762c : SUCCESS in 27m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f8524e0223640bea39b8d14a6e7ba96 : SUCCESS in 24m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/53a60094d32b4916b7a44980a11926dd : SUCCESS in 18m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a71c8f3f60fb42728a0bd7913ed91b39 : SUCCESS in 16m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3d704e2f667a4baa88106de2cda66cc1 : SUCCESS in 15m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50bad4ad454d45df91afbe7319b31ec4 : SUCCESS in 22m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/568fd00c9c844884a665026b7d92b136 : SUCCESS in 8m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e99b328fed704af0b1d50805f021f357 : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5a060e81c55e4eb7a063aeedfd3b174e : SUCCESS in 12m 32s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9d9ef8ed383b4037abf3f6896ad63c0b : SUCCESS in 1h 25m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dad707e566714eb8acba0b9f7b3e6f95 : SUCCESS in 1h 19m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c936eec364d347389dd64b28db1477d4 : SUCCESS in 1h 53m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dedfa430dbbe4ab6bfa9699ca5597da3 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a8f8e577f855710ceaa32f805b6784b041066a42","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-28 17:30:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5f0c0699303124d77d0ad78686b899ff39fb900","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 17:31:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22437ac6967a872e2f003e44b09731739cc7c235","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 21:40:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab72bba19041321cbdb360f8ae72a7b3bc9b2675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 21:40:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccbdfa168ab4ff1afb98d508582c1b4 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1baee162125c41eaa3958b24f380e1c7 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69ba5356f14f42239dd2c08766550375 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1324fc49ac2b4f02964eee9b4a7767cf : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2b4b1b9599b402f851fe0a8e845f2a3 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9549f62a7f0f4587b9b027f20a046eb5 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d6f061c96b54b208ea39db6f8b9773d : SUCCESS in 4m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/658e748f797e451185f420512d608b07 : SUCCESS in 8m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64481859681c407ab595f16760e04ce1 : SUCCESS in 8m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/de87170ab6d24b2a920ed1f1fd95642a : SUCCESS in 10m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f9439909c3cf4137b504d8fffea7657a : SUCCESS in 9m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/01745c0f197645b2845ca752b0d1eca9 : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ff49e768b584b17a40ea9852d0136fc : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/731273d7f0344babacdc70804c9f2e89 : SUCCESS in 11m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3a31f68e09504f838844f736518a131b : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd726bab07c346c6a6f35257dea6c64e : SUCCESS in 7m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/225d9e8d5efc42688acc5f02d311c7e7 : SUCCESS in 33m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6407b0a7729642ee9fc06c86b0270817 : SUCCESS in 30m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d1d77e5849a947679da677e42eaebecd : SUCCESS in 24m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d161a0d90c384c8b8e2547d36465bd36 : SUCCESS in 21m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6a368069529f42d08847d33e358d5b43 : SUCCESS in 15m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd63e1a5f5f04dd49b44beb303134478 : SUCCESS in 14m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/642a1d4c074f43c8bfa652f229506ff4 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22c8f63542b6428c9250f3c7043854ff : SUCCESS in 19m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6c3f3754a2a4bf496eceb8d61108a2a : SUCCESS in 8m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c121585dca104fcdb37296b435f98c69 : SUCCESS in 10m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/02ea4f0f94ac4d8ab053f48ad004b2c3 : SUCCESS in 10m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ba9ebf7dc4374cfb9c6cee98dc5852c3 : SUCCESS in 56m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90cfd1eba1944e6baaacc9fd754f943a : SUCCESS in 1h 14m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/976129e4d76f4c6b92a327fb04da00eb : SUCCESS in 1h 48m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/557c916804de47e9ae12ad5683e3a267 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b2bc9d370e10f397ce9508d9f8e240d2f15f6a31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 21:42:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7256491f5cb24556882f811e0e089d68 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b35360be875591afa08b847375fe1d4461a9869","revisions":{"8b35360be875591afa08b847375fe1d4461a9869":{"kind":"REWORK","_number":1,"created":"2019-08-27 18:17:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/678934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/678934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/678934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/678934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/678934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/678934/1"}}},"commit":{"parents":[{"commit":"52cd6b7b9686e8f8156257f3cc77d8cae836555a","subject":"Merge \"ansible-runner release 1.3.5 was pulled. 1.3.6 will be released next\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6b7b9686e8f8156257f3cc77d8cae836555a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-27 18:17:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-27 18:17:41.000000000","tz":0},"subject":"update constraint for osc-lib to new release 1.14.0","message":"update constraint for osc-lib to new release 1.14.0\n\nChange-Id: I4705d0a9771926abb3560a6204ee26169bf47f52\nmeta:version: 1.14.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Commit: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Change-Id: Ic2fd4535ba2a0d37573ee252f45e126245cd3248\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b35360be875591afa08b847375fe1d4461a9869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b35360be875591afa08b847375fe1d4461a9869"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
