)]}'
{"id":"openstack%2Frequirements~679000","triplet_id":"openstack%2Frequirements~master~I3b423f1c64fc652c27b5e63038f373ef015d1e7f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I3b423f1c64fc652c27b5e63038f373ef015d1e7f","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-08-28 06:39:53.000000000","updated":"2019-08-30 05:40:13.000000000","submitted":"2019-08-30 05:37:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679000-1567143470760-727564a4","meta_rev_id":"3f5fa99e402c23887ea9f97ee19fbab235a23ff7","_number":679000,"virtual_id_number":679000,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-30 05:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-30 03:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-30 03:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 03:01:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-30 05:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0cb13a47fbd64823921530fe3bd0fb578f66002","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-28 06:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5f1d68645014a38b94bf565fd61c4e93bb0c25f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:07:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0088d8a1cde4dff871898fb42505440 : SUCCESS in 3m 58s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bd884599f0cc4890b7f298a2b9cfa3f7 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b4157f7c8242af82c9e7c90676e9af : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7393dbae93d7426aa4e9e2992a5d929a : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/463e3c7f6f6742d8aa3d472c0dfec334 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cfd0f67f4f84928bae94546c45c0f18 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c04b89387ddc451799031c053dc90eda : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/087bae60158741b9876030bafcfb6840 : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a637761cb51b4412ad94bea95a92b473 : FAILURE in 10m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/20b407c42ea541108cb8d80adbe1a4c9 : FAILURE in 13m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f7496b24a5eb4430b75cae16347d88c8 : SUCCESS in 13m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9aad93f9102742a19f0a9c6087672828 : SUCCESS in 11m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c01b33138ed8448ca76f0f5e98eaeaab : SUCCESS in 6m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8604739ce98643ffa685b5f83e9f5e0f : SUCCESS in 7m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4d4a02ac2c194b6ba8b8079261e5303d : SUCCESS in 11m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d5b899b5a2f40aa9850eef49f1aafe7 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/445c837bcbf64cc7bc5b6b10c8fd0796 : SUCCESS in 12m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0a075f3aea174d2996a3131d4b10b800 : FAILURE in 33m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/753e76d08cb64096bebd6b1d5daf551e : FAILURE in 34m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/19f80cb006374214aa4451c2d8988062 : SUCCESS in 25m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/753c9c106dc547c0a11c77882146b92c : SUCCESS in 23m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/eda62fb0467b42639fd6c68601cbceb1 : SUCCESS in 14m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/703b28eed6474c169e652d4ed70174be : FAILURE in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d471c91244554219a89f8b89973b913d : FAILURE in 19m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d8eb9bc235a4464926e91583c3a57a3 : SUCCESS in 22m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63ef637acace49fcab417c8f6ea3e131 : SUCCESS in 7m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f6ae21287c634a47a2a14aeb4cd91e3a : SUCCESS in 10m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/469b2717147d42bd8b7bc51c4711fd57 : SUCCESS in 10m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d16dc470bf524fef97a77297a078e4b2 : SUCCESS in 1h 10m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/65cd9d3a3f2f470a9cf6d4260e5273c1 : FAILURE in 1h 11m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/97747998b061432a8b309edab1790998 : SUCCESS in 1h 51m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a31101b3a3114f078c433c945fea8c7a : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ce6fe65f2363a5be1f92f9e84a3c789ef850e548","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:41:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74aa8a2b360a71e19846e32ca6f746f96b9bdd7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:42:32.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"abc2ea29afbfe5102832c50dd08ac4745ce0d883","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:45:45.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2cb4a56355ee24f8ad9b38f5e1bca3550da20902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 00:37:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e18e9334f44475995fa8b952d71692 : SUCCESS in 5m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9679d758cf9f4a7f8c7209d8196727a3 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a5b48944154d019ea2b9f3b3d65cf2 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40046ea9f77f424ba9be143605f0ea17 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72882083b74c41109bf20b480c3dd11b : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcbfbe2e60fd4139a2653bb3bb0bd87d : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63506d8be37841b286db1bf457750d89 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d0938dd06df44bb86b42182ca614ded : SUCCESS in 5m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/65ebe90fd70545beb97bc59e90440ca3 : FAILURE in 13m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a86306e2ec04aea8e14f3c8201f08da : FAILURE in 13m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0bb7595d144f46c7977357ac2420534f : SUCCESS in 11m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/51f9075d04ab4e51b802a03f16bfa829 : SUCCESS in 10m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4876f393d9214717b81a8f3a5d9f1550 : SUCCESS in 6m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa1110a4ede34bf8b04f00d025161316 : SUCCESS in 5m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/31cb37a6142b48698374be1e9b2708a7 : SUCCESS in 11m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/43636d6625bb43c9be8dd77de30588f4 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95984366588a40848eabdd85a7582490 : SUCCESS in 10m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7cc55cae908f4c4a8afdc914955438d1 : SUCCESS in 31m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adbe6c16129d43ad8fa8117993f8d6b1 : SUCCESS in 32m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/053c24d18c63462398fb4cbca354d253 : SUCCESS in 25m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c7094f9018df4a3196f25f96da127a2c : SUCCESS in 29m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a6740fdf39b849af9a16f24aeb404fcd : SUCCESS in 13m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3b17670fbc5847e29bc77abf3dc5af9d : SUCCESS in 12m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/644793cbae984081a341f09d47b0532f : SUCCESS in 13m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/637d56d52b714889ad2d08cee9f3252e : SUCCESS in 20m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46689b596b57469aa944b58473d2c09e : SUCCESS in 6m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/54f6b340eed1439a877a7c0911bbf7f7 : SUCCESS in 8m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6592b5ed686d4e118c43b6240bd3a089 : SUCCESS in 9m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/456a6aec95454f6da83154309fb0409f : SUCCESS in 2h 01m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c874c735db9e48c5a58445f1cac25e0c : SUCCESS in 1h 22m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2cd5d097e18744fbbca0bc17c0acec0f : SUCCESS in 1h 42m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b867408f5fe42d8bc2182fed95b8142 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":3},{"id":"fd4288004654321708cfd9c197f46f542e0685a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 00:46:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"25153f16aad8b60ad3ef2f1e2910a111bf741d46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 02:27:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0aaca0a48d94b4a88d910574ae5f2c5 : SUCCESS in 5m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b415eb51cbd9416b97c4ffc681259fc6 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083d9be511604cb1969899a721a10e66 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ac98dbc7ea64752878ca420c1b1d5a5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dfa51fb5e7c4896b8effe7062598559 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/985478c03cc34c0cb3fbcc3173df2710 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7abc62bca5143aeb69922e21ed63b3b : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/291fc6192908408a9a4cfa872673944d : SUCCESS in 3m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6a96fa170417492583d6f0f0a054d7ff : SUCCESS in 12m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/21a2b92eabc44d14adac6b270072a09d : SUCCESS in 13m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1c28a97be45f4a368dce5ea4a00268a1 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ce15e21fc5594df4b8acaa9b369a84e3 : SUCCESS in 9m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3aa313f7ac6f4ecab24c890bcf3dd7a0 : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/944db06a203641e593a371b3490ad8ff : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/08b36632bd974bba88317f76e2dee37c : SUCCESS in 11m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5c1b3a424f3144cabe9c1931b1459dd6 : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0525e97e2343470fbd5c9c7070b255ca : SUCCESS in 7m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ec8a802b0d624123881713ed9faeda9f : SUCCESS in 28m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fe004fe6727d4c6a82b6ce00b28069b5 : SUCCESS in 32m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7f8dd88c733d4d078a334b4f2aef2b79 : SUCCESS in 22m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f1c255c1769423b923cc7877abe526a : SUCCESS in 25m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c7e6a42b1d484eb69629dc47ad219eda : SUCCESS in 12m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/54ab7cbffed741fbb9090e74ccff1409 : SUCCESS in 15m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/12c8f1a967d14ad6a7d743dd58ec5529 : SUCCESS in 14m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37af741b819342f187227f4609dd1471 : SUCCESS in 19m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c657a423e944822b6a2c463837ade37 : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b94c6d6a4d4424f9599dc3190486f40 : SUCCESS in 8m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7603e49257b94c2480a3e22a8e4178d4 : SUCCESS in 10m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d495d666975a47ac9fac2976d1e203f1 : SUCCESS in 1h 09m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/171a7c41bf4246d5bc1b6913e4177cef : SUCCESS in 1h 12m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a0ed4120599848699a0f05b5399c0007 : SUCCESS in 1h 34m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/61bac5f3120e42528b6780adbd36796d : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":4},{"id":"15b788799f82448fb4e3b85f44d606bb6bed740a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 03:01:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1a196150a66156714334e4a473ebf1eda83c96e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 03:01:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"53528cf82e42e6e8906ad132545a1156747cd421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:37:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"bdfc62882935cd07f449123d2f6b1bb1c27c5a11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:37:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7481cb9fd334e90a1c939ed162e7a04 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebca4bdd7e654677a91690ad6f175e55 : SUCCESS in 10m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1df9a2555c334c10a2587a5d8b33fbd5 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d972af2e465644c8937a29e8eb243ad7 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db3ee55699364b5e98dc88b8a602360d : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a545334d73b549c8ad28d316ba9d81ca : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0eb05f1384524c8caaedacc132ed1dc4 : SUCCESS in 6m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3701eacf630d4168bf7052e5b5397f20 : SUCCESS in 14m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4f1e3c48f5e495eb5284310dc1cc6c7 : SUCCESS in 15m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1f83251029e447d28d43065bf2def5f4 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8618b005db304c74816942c38ab799c1 : SUCCESS in 8m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f3d12bd1c15642ef9dda9e589886fd78 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0901456b1809423dbfa89f33aa1e1965 : SUCCESS in 4m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/33c6fba515dd4ab4ae8947334639f8e1 : SUCCESS in 16m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fb40c64483ac42c7a02eb5a63b0ecd2b : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0debd96ea78483eb5c049a3c29f7d5c : SUCCESS in 8m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a3baa9a81c5f48cb95624fdeabd4aa7c : SUCCESS in 28m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fdbe4f33a9354bcfbea05bb2d97198fb : SUCCESS in 32m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f681a63fbce24312aed81943fa767a8e : SUCCESS in 26m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f63d805c61234bb8ab21307f11d6a530 : SUCCESS in 22m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c844f3eda90e471f8f59664fa49fc770 : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/840d089617f7418483ed0864aea921c6 : SUCCESS in 13m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/494c54ff79d945058255a3d9c4cbebb4 : SUCCESS in 13m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c401cd74364347dd80110b8110ab9ed1 : SUCCESS in 20m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba0dea0ed7534363a4ade728ce431fdc : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/790b5b5416e54a3a80d71f072589ae11 : SUCCESS in 14m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c5a56569994426fb6c86dd329db3282 : SUCCESS in 10m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f892b4f57e734df099aa6974be861e5c : SUCCESS in 1h 08m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d8716448acd4b95a9126eece08d1de4 : SUCCESS in 1h 12m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d8681d2d9e9343e0a948f11931942d3b : SUCCESS in 1h 36m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54e46d691ae64c3db9c1d4fe2aa158e4 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":4},{"id":"3f5fa99e402c23887ea9f97ee19fbab235a23ff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 05:40:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6dcd8ff392e4535b4c6a9a9051f5de9 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"61aa8c683868817dde0c17616fa0c978fd1e4fa8","revisions":{"7885d708d5468ca930f3c7a95b8510b1551e4686":{"kind":"REWORK","_number":1,"created":"2019-08-28 06:39:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/679000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/679000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/679000/1"}}},"commit":{"parents":[{"commit":"6575d739aec00990a53abcd7d38655774dab7e27","subject":"Merge \"Blacklist eventlet 0.25.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6575d739aec00990a53abcd7d38655774dab7e27"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-28 06:39:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-28 06:39:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I3b423f1c64fc652c27b5e63038f373ef015d1e7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7885d708d5468ca930f3c7a95b8510b1551e4686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7885d708d5468ca930f3c7a95b8510b1551e4686"}]},"branch":"refs/heads/master"},"df3e281068c844f2bb8091dbb737357a06cf0ff5":{"kind":"REWORK","_number":2,"created":"2019-08-29 19:41:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/00/679000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/679000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/679000/2"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-28 06:39:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-29 19:41:29.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I3b423f1c64fc652c27b5e63038f373ef015d1e7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df3e281068c844f2bb8091dbb737357a06cf0ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df3e281068c844f2bb8091dbb737357a06cf0ff5"}]},"branch":"refs/heads/master"},"66036dd53350c023566514527562f613af7fe4ea":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-29 19:45:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/00/679000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/679000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/679000/3"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-28 06:39:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-29 19:43:07.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back the following\n\nlxml       held back for nova (active wip)\ntornado    held back for salt (active wip)\nstestr     held back for general issues\noauthlib   held back for keystone (active wip)\njsonschema waiting for some openstack releases\n\nChange-Id: I3b423f1c64fc652c27b5e63038f373ef015d1e7f\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66036dd53350c023566514527562f613af7fe4ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66036dd53350c023566514527562f613af7fe4ea"}]},"branch":"refs/heads/master"},"61aa8c683868817dde0c17616fa0c978fd1e4fa8":{"kind":"REWORK","_number":4,"created":"2019-08-30 00:46:24.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/00/679000/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/679000/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/679000/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/679000/4"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-28 06:39:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-30 00:46:08.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back the following\n\nlxml       held back for nova (active wip)\ntornado    held back for salt (active wip)\nstestr     held back for general issues\noauthlib   held back for keystone (active wip)\njsonschema waiting for some openstack releases\n\nChange-Id: I3b423f1c64fc652c27b5e63038f373ef015d1e7f\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61aa8c683868817dde0c17616fa0c978fd1e4fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61aa8c683868817dde0c17616fa0c978fd1e4fa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
