)]}'
{"id":"openstack%2Frequirements~679216","triplet_id":"openstack%2Frequirements~master~I55d8f0c1fa2db519a19b72c752e339fa27e233b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I55d8f0c1fa2db519a19b72c752e339fa27e233b9","subject":"update constraint for oslo.config to new release 6.11.1","status":"MERGED","created":"2019-08-29 09:55:14.000000000","updated":"2019-08-30 07:00:03.000000000","submitted":"2019-08-30 06:58:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679216-1567148288610-91f773fe","meta_rev_id":"9330820746c97c25009d52d417f5a483067acc0f","_number":679216,"virtual_id_number":679216,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-30 06:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-29 19:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-29 19:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 19:28:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-30 06:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aade251f4cd166e0a65610f926e3252f203f6da8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-29 09:55:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30fb7b91791c1e0b385be42e7e3e43f21ccff6a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:47:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5712b26d740e4212b399820e13419a70 : SUCCESS in 6m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0a42bc70ba1443828319cf97fe30b0ec : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1574a6aca4b4003b504c6fb0e802b70 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7ae263e5c284fa3a771b2791ccee4b0 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c44f7952a67443699c70165fc1353739 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c687c3e62f94615941e4e4249302cd9 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edfb34f32b214e0da8764c7f98443c14 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c82e728aea4640b48a64f8f6354fa094 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9cc09176d0364ce8906d33b49f9751b0 : SUCCESS in 24m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4611999cdb63410293b93c7e8ae061c7 : SUCCESS in 24m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cb5e98fbba534be197d30a7dfc126f67 : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ebca59cbb9146abb8e307544a261580 : SUCCESS in 8m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9d4ccf21f87d49f6853dd4cd8365c6d6 : SUCCESS in 5m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/992dfb675df54ffebc15fc801ce157bd : SUCCESS in 5m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/84bdbbfa77ea4758973c1ef4a172e585 : SUCCESS in 12m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb3e2c0e9a6d454e810d1d768188f7b8 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b903bcce9ba4747ba959b1e787393c5 : SUCCESS in 6m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bbd402cb866e468cab942a091582b57d : SUCCESS in 29m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83c38e483b664e26a5de17c0d544666f : SUCCESS in 31m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/73ee2d8db2cf4e8890b819165b92874f : SUCCESS in 29m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/29298a01c1b84058992e36231e935997 : SUCCESS in 26m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a51a65a3113e49f8bd1e423afdb654f5 : SUCCESS in 13m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fa682d6f97e644999d38db61f70bec45 : SUCCESS in 14m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/13768ad8ac44478fbe608f0ec887b1de : SUCCESS in 15m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24b400fafe2a4a7488e211c33898fe5b : SUCCESS in 19m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05fd82c1d21f4a9ea239a160564b42c5 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc2a825cab0847739fe887c6572f1bc7 : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6703eef4cbf44c2283e926e85a741c0c : SUCCESS in 9m 08s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/16300edde8f4497492df81aff91678e1 : SUCCESS in 1h 17m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/338d6bf8eea048b6b12456b315b983b9 : SUCCESS in 1h 18m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d13bd2044a00401a9d33c34b90919788 : SUCCESS in 1h 45m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d5c107621b7843ad97588ac863eb74ce : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":1},{"id":"18829f816d718feb2abfc9ca867b3dd6d4d19eca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:28:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"687f842d16c9a302ec80148822ee8a5892432733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 19:28:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b990c287b81a20946c287662dd275fbb17bd6a51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 23:41:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895a61c7515543eea77116285713afaa : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b6b6e07b614c80bc70dba677f3fad8 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66504a91f6ba4c9394ad95b764f2f4b8 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3751794e54b84a299e615206e0672d24 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b58b03e30c641b1ae2a71a43f145354 : SUCCESS in 7m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5a79290cdcc4d53b93e2f82935f57cf : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2137bad01434ebbb430b30c7c0bf4ea : SUCCESS in 7m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fc1049f8dc1d4136ac6120f8bc75a78b : SUCCESS in 14m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5ce131ad1c94b32b7acbf2529d92a3f : SUCCESS in 16m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/30d28b041a5c4e379baefaea1efecd41 : SUCCESS in 10m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1f5855feb7504599bd0b9ff495b91da8 : SUCCESS in 10m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/415826da43004ad49ca378cf064a4a13 : SUCCESS in 7m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea444d70f1dd4712ae8922e9708ca604 : SUCCESS in 4m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d728d371401d430cb83ebce1091d9813 : SUCCESS in 11m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5852635741b944558e8a3b51326f72d1 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63979b38302a4c2d8ee4586644b1fccf : SUCCESS in 10m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cefffc30a1474d4b9ae5c5cb71da28d0 : SUCCESS in 39m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c08d7b0e4c464d89afde579e04a8da11 : TIMED_OUT in 42m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/af85b144b0d641a6bc5ee936d9ca8e4c : SUCCESS in 28m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c4a14dbfb47c4e9586b3671d86ae30c8 : SUCCESS in 28m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3bc898d4720143f1b270993c53d341dc : SUCCESS in 14m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/55fee0c9752d4f6e9297ee8dcf2c3518 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/920c8ae3e94f4ad29cfd7f4fe33ea32e : SUCCESS in 15m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/951e28e08c054e4ebe0f33470c936ee6 : SUCCESS in 24m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/887023806cf64a43b6020a7d4c0169d5 : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f706fa297ccb413086e0db49b385170c : SUCCESS in 9m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee6bd98450bf4b01b496c77d181edc8d : SUCCESS in 10m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1bd59a368ceb48a29e00252e61eabbc0 : SUCCESS in 1h 09m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbd9959f3c21425aa51c4c7597a1cb18 : SUCCESS in 1h 09m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6ed40d9efbfa458dbf4cf5751ceceb59 : SUCCESS in 1h 44m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d7530d66281a45e68d84aaf6f28ede4b : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"60ab512b1c33ca90ca6acef844e93455d343ecd4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 00:14:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a05c231d9e1d73ef060a153ef1b7ec21c8566a30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 01:56:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f24b426e6274a5d8cf73313968504b8 : SUCCESS in 6m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b7bb5c3767794969845f910ca152a1f2 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34a2971bdf14e04b5e2884774a1aef7 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f93466fb598b4cf5befcf35e3c231ed9 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ab3d3b440e34fa695b56b5ffee81383 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/093764115dd6482289cac53ccf1a959f : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2652551b88d485c892e285fd95b1fc1 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80aed6c665024679a99bbedd0053368c : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c1d53c977e194849a69c9ef5da5641f6 : SUCCESS in 12m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4467ef46b73d46768d3c78eb3a55026b : SUCCESS in 15m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/68d3f839a006407fa1fa9b390ace364b : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/610de445af3c4726bf6c03cd5379c3f7 : SUCCESS in 11m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ae8cebe7c113458c8b1b2fb17a04cd74 : SUCCESS in 5m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/39258594cc9c46f786211b2759f13758 : SUCCESS in 6m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ec793f93ccdb4171a0395f6853c2ae49 : SUCCESS in 11m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b5115e3bcd01415cbfe647d210f092b1 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4304994147324ca1962ba581ab7b56c0 : SUCCESS in 8m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/49466b2b7edc4cf5870750c508358cf3 : SUCCESS in 39m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/460ab8c073b2463eaca46cedd9442553 : SUCCESS in 30m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/753a1aab9267498cbcbd67b621371756 : SUCCESS in 25m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/888694c831f84938addc35c0d2cbd09f : SUCCESS in 23m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4dd1f8a6e09a4767bd866961a3cd3bea : SUCCESS in 14m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c18af780b0e4472b7e7dcf265c6a9e9 : SUCCESS in 15m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c626c04c03314b3f8d1d145bc25d7545 : SUCCESS in 15m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/770185ece62a41e9a59147ea29b70ad5 : SUCCESS in 20m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/281c86e55f574d4998ed8318c1b441ac : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c6e055419611427198f3e4d44c84e7e5 : SUCCESS in 9m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/663a61631b9544a6b6a652e02e9adf8f : SUCCESS in 9m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6766ff6fa9c74d7b9e5c4909aca8e79f : SUCCESS in 1h 10m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00966f586c394886aee4e43535562daf : SUCCESS in 1h 06m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/df6459d4a65e485f8f3bb60544a13c67 : FAILURE in 1h 39m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/796ce89861444bf4bb9993f714cf70ba : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"56a80389cfc2e48fab76f5d2d9e8c09e84f05649","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 03:00:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"884f45ab1ae5c8af719033bdfd15de6a1d4074e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 04:56:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5325eadcc728443caa6e94487aacbf2b : SUCCESS in 3m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7820fa0d691644298c664bc8f957c06e : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95cd3cb289ec4637a7c36a5470139f5c : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a20816e1780247608f4c6903dd0d71c8 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a739579ea2cb4de7b465f923613d745f : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84624ce5fcdf48cd9cafe90281284fad : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b3034294af8451d9faea66003f11d19 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5885293641e84dd1b0915bc7cd04cf79 : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7ad44fa0bf8b40deb9ff60dc996b1064 : SUCCESS in 19m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/40d91a3de6a84baf9ff77a4164908de5 : SUCCESS in 11m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a6fe5eb0f1447faa893e9c98b9454d6 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/378ccdf0e8604b0e9c49343c8d715266 : SUCCESS in 8m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/43c0836ded3e4d8394ebf7176d77bd06 : SUCCESS in 5m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7af890c28fc14ac8a57e09d93109fa64 : SUCCESS in 4m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0b96b38128cf4425951c6df4df94ad0a : SUCCESS in 12m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e65d0b764e0a40ef9ef0a59e9412818b : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/37c1a0c61b4948799bce639b62156cac : SUCCESS in 7m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f72c06a992ff465c8bf5157945632719 : SUCCESS in 40m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/82aecd05f39542ff9e678ab1a5a47154 : SUCCESS in 29m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/249b17da0ba24461891cc893936dd3be : SUCCESS in 28m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f1edf2a5597490c93fcb97c17b03aa2 : SUCCESS in 31m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/efba24a6a1e0479c9895db5b4955bd75 : SUCCESS in 13m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/279f3a974c274bdb97d0628982e6f80d : SUCCESS in 13m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/03f8a5e760f6455a9411aa90da6ab892 : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1434ce6fafd4e13bab40f9d4dce2a9b : SUCCESS in 19m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cdd91969a274e528f19169515bfc598 : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a18cc3eea1a94e91bca8fe4334540b7c : SUCCESS in 9m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e00c6b6103494bea806b4096eae62f18 : SUCCESS in 9m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d2eb8b8ebf334c469a719894295f3ef2 : SUCCESS in 1h 16m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab55319ca92449bf93da51456a6f9571 : SUCCESS in 1h 07m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f2fb026f93a343d09e8a6c5c2be63cf8 : SUCCESS in 1h 49m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3adf506c872646768646b6b90e9b2e2d : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7926014fd76612563fc56e0ebf13765ec6d6e52d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 04:57:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dad0d005820aa7a6c1d775e9825ad126c8e4592a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 06:58:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c9e73413c24ea993a6afe355828e28 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0680ae329344bfaf7cd5caee1475f3 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44adfa5e7974402a82dbfe038ab0be74 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9961f7fd7b414beaa70d4ee092ecbb95 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae33cf08cd5048a9ad2f664158019cb2 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9db353222db04d1bb7e6943759c500fe : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/923d324f87fe484ca2a833232693199f : SUCCESS in 3m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2d7dc648c1224be8b458f11f2aa94114 : SUCCESS in 12m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/890b471cdde54552b2dac5f58c9ee04c : SUCCESS in 17m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9bb4b4c564c34a0788e0c2bf5acb205c : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b3af7cbfaa94386ada082cf7f7bf9c8 : SUCCESS in 8m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/52811054979d490c81c58815e6540bb4 : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d050f31de40d4893b8bb8d57d9a1e2e6 : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/672004ded5564b6a996ba6372d0863f3 : SUCCESS in 11m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/273e96e5f8514bcfbbb43db3b761c43b : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29c61b74e30a4f8386350d4f200a5089 : SUCCESS in 7m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ec95d9b808714214a538175853b29d73 : SUCCESS in 33m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/67b773face844fb3b95c66427a8560bd : SUCCESS in 28m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9a0f6c52c8484afeaed19bdd686cc0cb : SUCCESS in 24m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/292e7d818a3a40d1aa85fa4381965429 : SUCCESS in 24m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0bfd33d7bea24dfcaeec8545602a8c75 : SUCCESS in 13m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b372c48b223f44f3a1c380cae2916ad2 : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1f7653eb4762465c929e71bcce36fe1d : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f2f3a9cb38c4067b860d3e72e45908f : SUCCESS in 18m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5ede9c0ff864b5c94cc897ff435b6ef : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7de0333bbc0d414f9ded8aedd28a9338 : SUCCESS in 9m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5cbdffde82f041cebc6397b0c801304d : SUCCESS in 9m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5291e96ed2b444d7bfaada9a45541949 : SUCCESS in 1h 23m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/20dc9def4eb747c8a11e8a9e3d8df4a8 : SUCCESS in 1h 09m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e83cb40b1bf4587bacb35387e2b3402 : SUCCESS in 1h 54m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f07bd460cf014e439c365c01f2247c30 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4dad1b5ece20f5fb05e8e4ed6e43eac98e7947e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 06:58:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8975ef2f626149fe01689a2f4b28e74399d8ee04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 07:00:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6edc18ed79ec444dab63bf052a64aaef : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a926c458d99e131552300e53700d2adebe97d187","revisions":{"a926c458d99e131552300e53700d2adebe97d187":{"kind":"REWORK","_number":1,"created":"2019-08-29 09:55:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/679216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/679216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/679216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/679216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/679216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/679216/1"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-29 09:55:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-29 09:55:12.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.11.1","message":"update constraint for oslo.config to new release 6.11.1\n\nChange-Id: I55d8f0c1fa2db519a19b72c752e339fa27e233b9\nmeta:version: 6.11.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I797f3f75af1b5243fac8a8bdc7e1aada8e9f4082\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a926c458d99e131552300e53700d2adebe97d187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a926c458d99e131552300e53700d2adebe97d187"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
