)]}'
{"id":"openstack%2Frequirements~679219","triplet_id":"openstack%2Frequirements~master~Id42c6b1faf882e56f0856582bae724aa4dfe1376","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id42c6b1faf882e56f0856582bae724aa4dfe1376","subject":"update constraint for oslo.messaging to new release 10.2.0","status":"MERGED","created":"2019-08-29 09:57:48.000000000","updated":"2019-08-30 03:53:46.000000000","submitted":"2019-08-30 03:50:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679219-1567137053227-727564a4","meta_rev_id":"c1584b275030e020f78a8320275b1bae47c757f1","_number":679219,"virtual_id_number":679219,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-30 03:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-29 19:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-30 00:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-30 00:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 19:47:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-30 00:14:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-30 03:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a59f40cfd2832ee9220b0867c40171604741dee1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-29 09:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8da8570e090c45c8bfc03c6b9e2170aa61d7847c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 13:21:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653651503379429cae644f33aa58c6b1 : SUCCESS in 5m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5324de18ed0a479482a8d619ef48bafc : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f4f6e799ff949a18d86a5910e1cb7b5 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cd0f395a2ab4293957fcf9a3d626ca8 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd8314c926d48f38bd684abae28e6c5 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39f63a361e224adebf1a2b9d745a7114 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4db2cf44099c4a109a299220940cbc22 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6547b878be7c47e29f546feb268832a5 : SUCCESS in 4m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e3c6f7f21f4448978ccacb1bc30d434e : SUCCESS in 11m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a06e923d74d146e6bc4c02281f914fe2 : SUCCESS in 11m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fd0b6a3b3c5c4657b1a23f7628b829b6 : SUCCESS in 10m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/436471dbfa4242c59ef5180bcbd16936 : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b465272bf6294121a8cff031b564e53c : SUCCESS in 4m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6c97a644b1f04b07a07d95926abdcce1 : SUCCESS in 6m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0f905458c4344c15b1155bf38fa0a815 : SUCCESS in 14m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66f2f01eac804e78928051e82ab602d9 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73631ef1320649e0b47af226985e229f : SUCCESS in 11m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/67eb3d27ecc14f7ebdd0c6f4ac8e25a7 : SUCCESS in 35m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/56ac2b92286746a8a85460707fa97d2b : SUCCESS in 32m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/32178ea1daeb4bb6b3f648f7de7e7153 : SUCCESS in 27m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cfef943f068348308390ff66769b2dff : SUCCESS in 23m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ae67350866b84f5f953292d7fcfa0146 : SUCCESS in 14m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f2f02dfc4a6449792d2cd50e572bd75 : SUCCESS in 15m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/292f58a630c04718ad54723ff5be87e0 : SUCCESS in 14m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/544840d35902499e97bd7849b3299cee : SUCCESS in 20m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cebffe06b00461394da46e85c035943 : SUCCESS in 6m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5adff3f5e2b9411a85b3cff6adab150a : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b93315ea74e74c8981497d4174b183e3 : SUCCESS in 10m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c8d5db8e039b4254a80db07357857431 : SUCCESS in 1h 18m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f8919475ea84e99bc051b2315dbb04f : SUCCESS in 1h 16m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9bd94c9d3da94a5a803c99beafc96358 : FAILURE in 1h 38m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/84db2d73c32a4566bd19030b67cfbd71 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ea1760916493c51c14a32ae087bc6846a5547bc7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:28:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b325489e87d9d905de4725678b981d564ad011c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-29 19:47:09.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d5396fe786bc415f66e15d698118b5760c30c05a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 23:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f487856ee44881a3709e9759c34377 : SUCCESS in 4m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c39879376ea84b0ba0c0a22fe06a1f63 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eeb091abe764d088de70055eae130a6 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5bbc7519192b49c1b0e8ce3e0a3b7c1e : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3835c31ba334d438a75da8ab01e8028 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ed75b3171d34e6ebeeb8e389d0bf344 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1d996faaaf24f9dafef25024bbbf029 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2292156f2b040278a4e3d59cdf23abc : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/74a360b2cbb144b09ce1b7554275f255 : SUCCESS in 12m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c35137fb0a1342f5a086a68f961b8bbc : SUCCESS in 13m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ab944173a4974e34863a11858ed96f0e : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d420285ba1d46aba44eed8411ada518 : SUCCESS in 10m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f3001de74bab40e9b825fdfeea772ffc : SUCCESS in 6m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b28f3be953aa4aee8f0e0c11995db393 : SUCCESS in 6m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/560d3a26bba445a799f92d130fc5f2aa : SUCCESS in 11m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f8eb6059eae8468c9261b02f8facb031 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/943b81273f324603985f4a09e0228403 : SUCCESS in 8m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bed0ba7de51a442eaf9fef5521f19f02 : SUCCESS in 26m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/241d5d9a57c148959e5d9601e734ac99 : SUCCESS in 30m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6baf938af0a54b02855c1cf0d9f5c57e : SUCCESS in 26m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7519e2b3e0e4d3bb7339862bc36c5f3 : SUCCESS in 23m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a893593310f149119d06be740f6443e4 : SUCCESS in 14m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4625bfe3e9ed4c0ba84fe587786b90e1 : SUCCESS in 18m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/806660ab67b2407eb4046dfcb374949f : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6073627b755e4c8da022ee91da8437ff : SUCCESS in 21m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d8330573364442489c29e53ac73b0d0 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8554f9d347d94d859e20bed672cc886b : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a74bc6642fe1464f84085385b5b9eef5 : SUCCESS in 12m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6c0dbeb11d5449eb8eb90357a59a0e13 : SUCCESS in 1h 08m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/87e63925ebe54410bd15c8d8ec72acbc : SUCCESS in 1h 09m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1743b02b2f4e4d49a0131ca3297c482b : SUCCESS in 1h 49m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7f0b17ba55494e699eb3dd3625b1a5fc : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0e99f0097a298201352b2cc2f36867bb246a5874","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 00:14:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93d40bb417ed37cdab4b0fbf6c35a6221cef785a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 00:15:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e83091edf1e826d7c7e699732da7f7c93f3519f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 03:50:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f709ec34dd45079ae3951ace3406cf : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db4142a5a554b9c9b552387e5dea268 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da744b8188f24758995f98f5b9a4bf56 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e64c6b70e940424cb02cc65b67bab902 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6127aa6ae3f6443a973c3c5b10a726a5 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5370b1e8be1b4ae0b2cd5728d9b09f0b : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/554bc9bb4eeb488c8f6650912664a28e : SUCCESS in 6m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b707dd32a7674eef856fb01807a884f4 : SUCCESS in 22m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9130d1720e0a4dfba2469792ce35281e : SUCCESS in 22m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0ab55ca22de64c759f19812c1389e3a6 : SUCCESS in 12m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f2a2432270cb40d1984f2c8b29c42989 : SUCCESS in 12m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d5b85625516d4aa68f660edff04b2fb0 : SUCCESS in 6m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5dfdf2fd34a4e538294e0ef8c279924 : SUCCESS in 5m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/efb4803c1e0b42b5938ebb86f550aadc : SUCCESS in 13m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/75a09c95e1954a6f929fe6d366b36a9e : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5863c41bbdb54ccd968f177a6bd7e8b0 : SUCCESS in 9m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/75a0cae5b6844b99961f3b6bc5aeab0e : SUCCESS in 32m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3afb44fb3f274db7a348aea1d4583587 : SUCCESS in 32m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4065639efda743b2a9ddac758dc1f9e8 : SUCCESS in 21m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f39bf37018364de094908090af7b9775 : SUCCESS in 25m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d21199494025451b8ccd4618b6e3ff2f : SUCCESS in 15m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5c3d87411cce4f4eb2feeab7288a121d : SUCCESS in 15m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2c5f600d4a494b88ada124c903bb9f16 : SUCCESS in 15m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9772051e1e274452b5a1cb5ee3b29620 : SUCCESS in 27m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1a30d8739e54e8181a0c7369c58a963 : SUCCESS in 6m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/30a3e6a25b4d4da9be6fb6cba4b3d82f : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bb442d8102e8485aa2ca7e09b017136a : SUCCESS in 9m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c7748e4b80ad4fc1aa723eee7e7fcb09 : SUCCESS in 1h 26m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cc8fc20fc364e669bd314d0f237d04b : SUCCESS in 1h 22m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/69886ce4d6aa4bcc966c972a5cbbb523 : SUCCESS in 1h 48m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e6e6a3842459487eb1ba11969d2db3e7 : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":1},{"id":"736406860e99b656ab78d26b42ad8e287f88cb64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 03:50:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"517050e4c2ff0302d5e60b8822cd4c2b6aa7178a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 03:53:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1056efb965a48e5af942002e2273cc2 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"721f4983457d6689aded8843d5902541068e9349","revisions":{"721f4983457d6689aded8843d5902541068e9349":{"kind":"REWORK","_number":1,"created":"2019-08-29 09:57:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/679219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/679219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/679219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/679219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/679219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/679219/1"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-29 09:57:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-29 09:57:43.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 10.2.0","message":"update constraint for oslo.messaging to new release 10.2.0\n\nChange-Id: Id42c6b1faf882e56f0856582bae724aa4dfe1376\nmeta:version: 10.2.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I797f3f75af1b5243fac8a8bdc7e1aada8e9f4082\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/721f4983457d6689aded8843d5902541068e9349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/721f4983457d6689aded8843d5902541068e9349"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
