)]}'
{"id":"openstack%2Frequirements~679220","triplet_id":"openstack%2Frequirements~master~Ifd34cc6a2ed0a4a01172aaccc7a27a8ef5991cf4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifd34cc6a2ed0a4a01172aaccc7a27a8ef5991cf4","subject":"update constraint for taskflow to new release 3.7.1","status":"MERGED","created":"2019-08-29 09:58:12.000000000","updated":"2019-08-31 04:36:19.000000000","submitted":"2019-08-31 04:34:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679220-1567226084743-3374a85a","meta_rev_id":"b198838428db833fd1551af5c04745dd2cf78817","_number":679220,"virtual_id_number":679220,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-31 04:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-29 19:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-29 19:28:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 19:28:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-31 04:34:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fe8b4600d8070c3977afbf5d890fa48b1c5726c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-29 09:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6b56c266301daab20eb8805904b55ab911820c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 13:26:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f129608618949bf9adf511b01861051 : SUCCESS in 5m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a055c567c01b47cdb1fa2ec36cd878ba : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde8ef4b4624404d9ba5a432ca429b16 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c53aa3e508e43aabbe3d190cc9cd23d : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/578d7043616a4906ac31024fecbe3382 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/239120651e39458f868a4dc1580decab : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9ad5cd66d8b477db0ee808e848fca8f : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f58c88208064cb4be3a287f166effde : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cb5a8e89a56f420198b745f2cb5c3d2d : SUCCESS in 14m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29b9c2c40feb4152ac260e4ffcf6e88b : SUCCESS in 13m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ff417ad98614420f9b215ffab7c06daf : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6df0b5859cff43adad54bffc47b53bed : SUCCESS in 9m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/83146b1f588845ee804b2d0af1c04a0a : SUCCESS in 7m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5bc9c98533ea48f9af3da33f0bae417e : SUCCESS in 6m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/270f050e42d84f62a689c0f484defe09 : SUCCESS in 10m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6739bd15763b448490edf92c853d2a8f : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5eed3e8f9c904921acbd4e6bb583bd64 : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b0049b7828784a8e9e9c831d6135ad5c : SUCCESS in 28m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8bddc7a5493d4c5d99d603acfa2519e0 : SUCCESS in 27m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0aed8de8248c428db3170e22676c4b9e : SUCCESS in 24m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5db3ff9e70364fd588ec799ba3a73504 : SUCCESS in 30m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f2612b74e89a4806a5cfc9690c2dcb77 : SUCCESS in 17m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3b0c015195a34c7e8025618fe5d6e593 : SUCCESS in 13m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/16c920d241fc4465b08fa1a06f7b5c47 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04a1cb60943e4f04b22501edc5f46577 : SUCCESS in 19m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79bf009317734a01bfd6dd7f74efdcbc : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2ce6b47fadd24803ac1c35ca99eabae2 : SUCCESS in 9m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2329b91714ab4c1fb03d30efeca7c903 : SUCCESS in 9m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/52d0d96c2ea84bebaddd790edd3e18a0 : SUCCESS in 1h 11m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a480140eea2f495aa1a7087acfe0c6aa : SUCCESS in 1h 10m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7d271e64f30e4d5cad8057041e56f78f : SUCCESS in 1h 36m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bdd3df03f12c4eb0a3fdc35a5a3b46a8 : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d7e4ea0500889643d9aadb5b5fc1e8b8cfbe01a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:28:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a450e6103860c62d4d1238b648008a94b8a1ce57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 19:29:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4240240dda5d559e7bbff57202351253eae88663","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 01:17:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35da8b7343f4424a3bc466d856fbf8a : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abed9b8f5f6c4205890700cfa11a1353 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18ce528ffe86406592111d1046ff9738 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d089e53c7c28491c93846e2d34f205ac : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88c14e3a02894a06aa1e6ff7cf661977 : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a23b18fdb3d4e259fe6ec34a553ed5b : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9e5ba965c234f56b070381e16f1eea0 : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8e054a125b3a4120bdedcb00475a675c : SUCCESS in 14m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39ff521c7d3c435f947ae6f002750c7a : SUCCESS in 14m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ef8e7f84d55249baa49876b338b14c30 : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/708d71a7a4364489bc9fa094d148354d : SUCCESS in 8m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/806d5d1b038d492e85356cab0dbffae8 : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/166004f3f13645efb3dd6502515ae06f : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fd223568f6b847569e7507505a9bbc62 : SUCCESS in 10m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e045884310d745818ab417db1b72e9ec : SUCCESS in 21m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db2a07f6ff224779827a98fb8ae36b54 : SUCCESS in 26m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/48c530a14406420e97895c417f133ef2 : SUCCESS in 31m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b5bfeb544d634b77bf5cc06e977e117c : TIMED_OUT in 42m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/85fd39a55b1b4be4b5b6736c8e197e87 : SUCCESS in 25m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/75d5a2bbae1544098944059c4b79ecf7 : TIMED_OUT in 41m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bfff43ec3c2249ba9b068faca3945fc9 : SUCCESS in 34m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/191de66abfb64733b4baf2dd995739c9 : SUCCESS in 26m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2d936f26247147dc91ba6397437c7ef8 : SUCCESS in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32d914864db34409b3b0a7e6f01bdcda : SUCCESS in 35m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84eadd13ceed46bfac0dfcebf13b78e2 : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5363f4805f334b469266757d23240200 : SUCCESS in 9m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/20761279eacc4587a1c6d382591097d2 : SUCCESS in 16m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cea84dc800d04618be9d833ddc27c4ea : SUCCESS in 1h 09m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b285800fd5c4cada8193df066358d8f : SUCCESS in 1h 22m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/73bbc9fdc9f347068f506fb7a5020860 : SUCCESS in 1h 41m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b80c6f0009c34e3cbb2a4ac0370de033 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4c484280208dfc3515d04f79e68163143fbc8273","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 03:00:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a36c1d409d128c1a970879e865dcdcc129d380e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 04:43:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783eb033840b47259f3e78c7a8ec0235 : SUCCESS in 3m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ad690c8388844c42843abc84616e252a : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf426040836c4c64a0b12c8adafa551a : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e0b8066d8be4730918dd7bb0a744a16 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4550397697424664a1eb235429be7a32 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dab2c202d08e439089ed8d21ba7175b9 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b7c7e8c82954fbbbecfaa949e9de980 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b75b9d8b271047c1bc5739132189b402 : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e1e08bda3a454b9aad6f00294557cdb6 : SUCCESS in 11m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c9e85c108214bbb9dbb043a758090f3 : SUCCESS in 19m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7be6ff64943c48d688146232520010a7 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a60c44011a1446ffb574feb63634678d : SUCCESS in 7m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/63c48f4c050e4010b4aa2fa487c8b10f : SUCCESS in 4m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ac3fe171ddcd49f98093f83ad9baa6b3 : SUCCESS in 4m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4463743f1cb7484a8c39d3939a6f4678 : SUCCESS in 10m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aee26a7ea65c4b0aa73b2a51f50408a0 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/931f4cbfdd8b44c58a9c6d01310d1e35 : SUCCESS in 6m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9c52ec29d5db41bba8f33e81cbf33c3e : SUCCESS in 29m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f12ba7431b84e9482fd99c347ba58e8 : SUCCESS in 24m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/acb894718e1b41df9b884b3ba6429c3a : SUCCESS in 23m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae20f9aab24244d88f5d09a06189ec50 : SUCCESS in 25m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a01f960c1c714d85a1322c172bc3ca8e : SUCCESS in 13m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/39e64b60e88b4209828bc6e1153bc76e : SUCCESS in 11m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4c52722583284a9a875a7300ac81b501 : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e83bc32d3712439faf503422c4900b15 : SUCCESS in 17m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80aaab6393db4ec796938a35fd4fe2a7 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/06dd650e030d4d2d989d7afe8115784d : SUCCESS in 8m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f167c2932284f799d504f6497a37239 : SUCCESS in 9m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bbbe9755a34b475bb2d36652794322ec : SUCCESS in 1h 10m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b52be644b2154652957696c30e9ee4f1 : SUCCESS in 1h 12m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cf8e62261c7246ad9556d546377cc7f0 : FAILURE in 1h 36m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c352f56ff2954bf1931812d043c16b6b : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e58a29a2b66cd3a11150db772827854cd8c4e476","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 05:21:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c53b13db8c71443ecbe9401cb082eae248e910f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 07:09:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d1ee8d0cb749e6965831ce49ec1ccb : SUCCESS in 4m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7095d19e0a0e4d4a864a650a872f0cf8 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f485ccaba6c44999a57fdf02d0e9915 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f2d4e7b6a8140fda42b76129016c8c0 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/685171e5a3034c19a0adcb24996aa527 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95588c9a0691439fa87b70fabd49a2c1 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3fcfb96e1134236aa4eef15f2f2e501 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/932c55078f9e46b2a9b6265e0692638a : SUCCESS in 4m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/03aa5fd547eb42bd81978b2bb2953acc : SUCCESS in 12m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b78c84bdaea43609cd13f886cf7b62e : SUCCESS in 12m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/898f196eac384219b5f47170816332d8 : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/79726c5f759b47b9883535d393315d07 : SUCCESS in 11m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/30b0a3572c414bf092f3dce19e997be7 : SUCCESS in 5m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3e2a45d32a9a428089d78d72ead8aead : FAILURE in 6m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fb4dfa9d506143d397b159bba9bd6607 : SUCCESS in 12m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/efb5a58612104f3db8c599193e6777c6 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a1e305cfcd249ec92c45ab2f4245273 : SUCCESS in 7m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6b0b8ba0d8cf4631a0d0195558859c12 : SUCCESS in 32m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/361df86871b34954bec3efe3fcb7045d : SUCCESS in 29m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/90f83e5942144acaab8c6405d79e16a4 : SUCCESS in 26m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ac3a757daf54278b93ed6019f12c78f : SUCCESS in 23m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3e640ea0473a468caaf1dc01b26ab40e : SUCCESS in 14m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7774dd550fd141eebd4e2c66fb5226e3 : SUCCESS in 16m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/98fa3efe80a348349ffac1630762f028 : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9beeb4e40977454fbf0bcba5dec2f94c : SUCCESS in 21m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7f6c42821f742908f7c2a619f2dd193 : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/be0795fddc0c4ecb8993535130abe8a1 : SUCCESS in 12m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bc46f55a5844187858702ce91af0c17 : SUCCESS in 9m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0e8ba3dd7f984cf6ae53fa8b8673bb75 : FAILURE in 1h 04m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b03bd1dc4a342b481a9163f2187788b : SUCCESS in 1h 16m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c859bbcd5ae74b08b12a434e1c96593d : SUCCESS in 1h 44m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7c662daa0479469dace2fd469963ef34 : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"080f77c67fad451f1d16816c2dc00f8b2fa818c5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 14:35:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f7a7d351b8a79ffcb1e22651f0a23e43f6e6a61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f86645e4ac42ecbb87000aeced98b6 : SUCCESS in 4m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e0d2e8c62af640f18080f3e91b604aa4 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f22a7bba74d4f49966525fce49f4fbf : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/096165e20713418eaae20734563fa2a6 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c47577a125a14fef83ff6e3926cf9e77 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab811663c43b479a915a354541052ee2 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9acd8ce34a5143c8854c80010da21cae : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0cb7706f307d4d2f8b47e21ad4f9b6c8 : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d6d664b3d6a84b39993587cdeb32d7af : SUCCESS in 12m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e85b937f141440ca9f85504ebef7786a : SUCCESS in 6m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bd5d8f97631b406b8acb2788e2040686 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/54edb0e7818c4a45b86c6d4c92cd249a : SUCCESS in 9m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1f1f0be901484ec4b2af43c84ad4e477 : SUCCESS in 5m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b825bc2b6014b83b1e987c8d5f4e7c9 : SUCCESS in 5m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8c15c95abd5d45d3bc4dd75323e04e72 : SUCCESS in 11m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/06aa526a3d3f45da97209f33e209b550 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b987c4b75c744ef9b2051c65bf7cce35 : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/66d73931be024863ab89c51e46dba610 : SUCCESS in 32m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a214476e07204223a312d30429abd9ef : SUCCESS in 31m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/358a8a442f1d4a968d909a63bc8929c6 : SUCCESS in 24m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bfef83b2a81d417fb3514032a4da378d : SUCCESS in 22m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/790671ae066640f99840e939462ed26a : SUCCESS in 13m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/897b1cb4bd8f493c814ef139e1709ff3 : SUCCESS in 12m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c7df83a68ea246eeb6636e81880ee537 : SUCCESS in 16m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec305d94409347b18dec6dbf2d2e7178 : SUCCESS in 20m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2fc2c5fcbe149d690dd8c394f4dc016 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/70d6f369d16c4770801af1b86feb32ab : SUCCESS in 9m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/91768a5a2e3d43ed97f06209ff092dcc : SUCCESS in 9m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/70507ee5a1be42e6b60b9e7d1748bf30 : SUCCESS in 1h 02m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbd8d0aeffae40db97234fb2a7bef870 : SUCCESS in 1h 10m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/33117fb17b5a4727b3c7a7334e314503 : SUCCESS in 1h 43m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e813bcfe91cc4ff4b26339992abb6190 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"93cbab797670657a1cf61326c8414c599e1c52b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 19:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89c1e8446e176582ecbb9fc8034e86cef2f7e143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 21:39:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fecc8a9d9144e7b942cf019f772d819 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df8e7167ebc410cad4b696b5c6d9e67 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/342bf4b837c346b1b9a48d04a688d495 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00772232c9d84fb4925c4edd866233ed : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1f5806787394321b88b0d02b5fe0dc4 : SUCCESS in 7m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c277bde7e5b4c0998598cea4670b5fd : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0933a903b85c4cf4bce76845ca54f098 : SUCCESS in 4m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c9428c000fcf4d7b8370617e21de20e2 : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/495a61e101d843ceb8a1ddbb3964f74d : SUCCESS in 7m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b6e82d0bdf634137a9a48638142cc01d : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1699f21c3c9b4187bf56512bec01d799 : SUCCESS in 12m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1f40a7904c7747098753823780ba3b00 : SUCCESS in 6m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5eb46352d3564325b18199bae81914fa : SUCCESS in 6m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/77e98510a35f451a9f63b64cbcdb826f : SUCCESS in 10m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a2b225266deb41cf854ad9f17881084d : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/034f241e30eb4a4f9a595a5a755cae34 : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1c7c5ef7a9714d7e9ef9f49f5eaddb8c : SUCCESS in 35m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/35bb911b646348f08a854f5c0809fa33 : SUCCESS in 32m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b23e55212cda4749a5b9bb9577ae1fd6 : SUCCESS in 22m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/594d0d834243459c98c866defc4ecd66 : SUCCESS in 27m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8f88824240a4c1681992984bd67d0e4 : SUCCESS in 13m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2838eec039dc4174a1faf311d528dab7 : SUCCESS in 15m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fbf4eecba7c74e678fde52d13cec8e37 : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5130cc55d5a54fe89589963571a24c46 : SUCCESS in 20m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67fa67c44397414c93d9165f71577526 : SUCCESS in 7m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fb46e0703df94380acefc9308f040198 : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71c11b14020347c5a3e3eb613300a02f : SUCCESS in 11m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0a5ab7c976f3423eb010e32ab10fb35f : SUCCESS in 59m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1b7d207414246829959c878fbdc2b2c : SUCCESS in 1h 09m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b6d708151e634440a05626a72e0d4779 : FAILURE in 1h 58m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2be95210f08040928d98bc6d0180d757 : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":1},{"id":"44a7bea0ce0a4ac0f33af7d8e71f367395294c4a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 22:02:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d52ee3b102819e2ae311a0edc28b46d996e0159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 00:37:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946d7a57fbdd48078e48f02fa2c2048d : SUCCESS in 3m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3bc7dafbdc64466c88c52a3f64799869 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0b9821ed874bf09c08490d6d79494b : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c639f78c5f314ca498a445f62645eb41 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c5d07fc5e074f85acc5e63b87e109f7 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5fdffd5a2994a49bdf9df955640b028 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15bd21f2197745bb850e50775f41061f : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f95f45760584b9b81fb311cdd077596 : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2e42b294996541969ec7be0fe1ae4168 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6190ac3dc67140119ce31a3f255688cf : SUCCESS in 8m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c725ce66cf144b0d9615c04b1e26ee45 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b38b82847b24a1397ef98747264f5fd : SUCCESS in 9m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/948a6735f913428eb501e0f733a2a4fd : SUCCESS in 6m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/384f112eba8148a6bda6d08d0213e8c5 : SUCCESS in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6d21c19e05124dc8be1624e80a8d6eda : SUCCESS in 10m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9e8c6df2b714a8a8a0265cd3c6cc615 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f967716d516c4147a4892462e3e87cf3 : SUCCESS in 8m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0a7c272674b041629bbcc5c998655469 : SUCCESS in 34m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5a8a052fa4ab472dbf207d315f250b67 : FAILURE in 32m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2d86917696f8490f991f65e401f9b350 : SUCCESS in 24m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7d21b46224c4d64a30c36d74a3386f5 : SUCCESS in 26m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/06d1a01e5bd2499cab71c8809446085d : SUCCESS in 16m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/50fcec14b5a447a8a8c3915249d6d29e : SUCCESS in 14m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c25eaf8239ae45c09a5ba71a107a3827 : SUCCESS in 17m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6be7e548be1480f9c9e0acc8066400c : SUCCESS in 24m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35b8278d96a942a782b850f66597a586 : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d9dd07ed8e7245899b8caf685bdb8848 : SUCCESS in 12m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/97a776537cc44b3c9273ad0d51da1709 : SUCCESS in 9m 41s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0e75d36bb0214ff4885c8262f474ac73 : SUCCESS in 1h 05m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a7789a828a54f8ea620c4369ad74449 : SUCCESS in 1h 17m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b8e8f69376d34984957b3fe4c3e1badd : SUCCESS in 1h 52m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9fd11b1bba5c4e9d899e2600a036c586 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c700c1b928de5474133eca90edbf51c5db896433","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-31 00:58:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d2da9cb9a2debfcb9cbb80371557543ae13684a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 02:40:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd889318f50b4c28a84ab97cfad8668c : SUCCESS in 4m 19s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ef6a727179d46c4b50d340a27227e34 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a721c8f351f4412e9266f7348f7e2135 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0faf1cfc047e427e906c512e0899baca : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15254570e8514c82a08f61b084b7c4e1 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90f5e4b14ba54a0aabf9964a3cb1315e : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc2cec2f9a7644f9a5662bc22b024c90 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8f3ecd9ede04217b1c430d078af3ed5 : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/60dcbb15877c4bf8971a576a059364ae : SUCCESS in 8m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f57f93c290a4f5e94c16f32376cab2a : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/862e8588ab6c4c9198e1111f93b8a456 : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8bf5233582e74a02aaf7be819a31356a : SUCCESS in 10m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bf23a53ebb5f4b7c97b18c04e592c27a : SUCCESS in 5m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c344b543638d412199aa5fbf6d8012cb : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/add1c7570a06452f8b66ade05e4f7c46 : SUCCESS in 11m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e6a778ee60234268bf37a1d8611fae93 : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce92ee6eec744aed83f276ada548e595 : SUCCESS in 6m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/51e4b3da734c45c7862c289c04f3965f : SUCCESS in 31m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a229799ef3b41699e9d19c228a0468e : SUCCESS in 24m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9cfb1cacae7f435db809bd82e819af98 : SUCCESS in 27m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae98735361a74de5bc2cc8dbd6ca877b : SUCCESS in 20m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c5f225f5e3464d74ae9fb28f28d6cfe0 : SUCCESS in 17m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd12d1208aef4ef99cc9d3c06188fff6 : SUCCESS in 13m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4d4555e04b114a01b3cb09242cde8f82 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/963762737fb846778677ef5adb3e3ad4 : SUCCESS in 17m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ebab470f55a41439c1fb83ca79058ad : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/48eeb139ba82476093056af5b4d2eb99 : SUCCESS in 9m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d131d8d3bda1445ba394042be93430f4 : SUCCESS in 8m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6dde374078204862b03abd0ccf08ddb5 : SUCCESS in 51m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b2dbbf2dbeb4aa89b51ae667ac604c8 : SUCCESS in 1h 04m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0c184b30c1f64cf4800f9841cbd2373c : SUCCESS in 1h 39m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/42317f58a5fe4f9281d738e8b40a53a9 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9a91ba940f1659c55856c5261578e964e0a9fcf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 02:40:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c72da2b1b3bb8321cfa4fca023d49a1f2dc1f173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 04:34:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dc9f5e2e7240a864075ae012c9244531bfdeba38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 04:34:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/932f09863df6456a89d9b607113c66a0 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde5d87287ed48f789271e11359d31a3 : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d144703a45c3459590a8d5ababfdf226 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc85649d118d486692e14b81fdec9db5 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cae0e0a10a9b42fdb6d3034e1837cb72 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33b2b372f06b44b88d788f01daed73f0 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e7f8f6ebe0041eaa47c9992bf2d139d : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c296f874314444d5958a8e4a87959ebd : SUCCESS in 11m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6317d4eec18448d5b33b90bb6ad256ad : SUCCESS in 7m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1a403f1fb1ab48d89becfef611523452 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8672538f01f04015bfb14f98533e3605 : SUCCESS in 9m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b2afb2d022b648f7bf9aced1fcf37d10 : SUCCESS in 5m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59182a4313c64ba5b081fe88a8f5cb43 : SUCCESS in 5m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5beb7f2265f640c9b9f57d84eb5e8e17 : SUCCESS in 9m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/33047f467d2e42af96703c53bc864139 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c73967c1016a47349878ee3439a453e9 : SUCCESS in 7m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/49f444b3a2ce495aadcd48bfb27907c1 : SUCCESS in 26m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d5d9b176dc24d26b2de40c7f0e7d9b2 : SUCCESS in 26m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7d30e13350274fdf8532487a4b9d5740 : SUCCESS in 25m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4621583d889842b08f7f2300fdadd51e : SUCCESS in 19m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d24975bde9b94dd8be1c644456eca834 : SUCCESS in 15m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b08c289aeba84455b6640f992a0e95d6 : SUCCESS in 15m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc2d58bd879347e19d0e84240f672615 : SUCCESS in 13m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9731aa11ffd64b02ba8eabebe1b6a73a : SUCCESS in 21m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/428b962630fe4171a35b037bc16c7393 : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf716d9bf38040afbafdecada913013a : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b9a1c1dbbb414e44ab75eb50b834d5c6 : SUCCESS in 8m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/bab3eb7621f34fc488e72a902f605409 : SUCCESS in 55m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c0ff54e85ae460f9b8d3120bdd0339d : SUCCESS in 1h 08m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f4d3b9526bc468c8fe2f33d4a8bdcbb : SUCCESS in 1h 47m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b03054d238e94077adfb5c229d28f304 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9a0323b6f757dd4fa2627eb29b76aef2366aa078","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 04:36:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2678ac51c7848aab2968acf37767494 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0878792d008ea01a415da67bc2d4b4b83604030","revisions":{"b0878792d008ea01a415da67bc2d4b4b83604030":{"kind":"REWORK","_number":1,"created":"2019-08-29 09:58:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/679220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/679220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/679220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/679220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/679220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/679220/1"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-29 09:58:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-29 09:58:10.000000000","tz":0},"subject":"update constraint for taskflow to new release 3.7.1","message":"update constraint for taskflow to new release 3.7.1\n\nChange-Id: Ifd34cc6a2ed0a4a01172aaccc7a27a8ef5991cf4\nmeta:version: 3.7.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I797f3f75af1b5243fac8a8bdc7e1aada8e9f4082\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0878792d008ea01a415da67bc2d4b4b83604030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0878792d008ea01a415da67bc2d4b4b83604030"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
