)]}'
{"id":"openstack%2Frequirements~679238","triplet_id":"openstack%2Frequirements~master~I36602e2f4ca1542b6c60df1ae9fcdc551451a3e0","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I36602e2f4ca1542b6c60df1ae9fcdc551451a3e0","subject":"Adds ansible-lint to the blacklist","status":"MERGED","created":"2019-08-29 11:36:14.000000000","updated":"2019-09-03 20:49:42.000000000","submitted":"2019-09-03 20:47:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679238-1567543663415-c377062c","meta_rev_id":"20be78809abec833a4a96e8795995bb8e87a5702","_number":679238,"virtual_id_number":679238,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-29 16:54:14.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-03 20:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-29 22:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-03 18:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-03 18:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 22:07:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-03 18:43:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-03 20:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9e17e112c19b2f614f65d37a71c5bcc55fa7d32","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-29 11:36:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"213d5fafd8c4eaebfe3f415a684aad3eab372483","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-29 12:39:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"50b17bdb3326d2a9ef328296705a33375e564573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:26:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1ff53cc92845ffbba544dd2d789340 : SUCCESS in 5m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/de7fa882011948be93968e07c52da860 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd7b10a3d09404aa5315a035530201d : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a07f387827941fc87777d519ebb81da : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d633d4dd08b4bf7a8ced5caf1abf159 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28121d4ea6c64ed7ba65a96d20c14718 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e19e14a930c14f21939226e7e1dc2aa9 : SUCCESS in 3m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c51e8c39784e46d98a23b07ff63761bc : SUCCESS in 1h 18m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8fe894df5c643b0bbc4b40905b10ecb : SUCCESS in 1h 15m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7df19599807540b7acf91821ecab8b59 : TIMED_OUT in 2h 10m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/48f8f3bfb8bb49438557ffb735916051 : SUCCESS in 1h 21m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3e274f0e18c9aae749be961b71fca6057e706d75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-29 16:54:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c0d7018461c741f89ab986d7125d0e6f749caf49","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:28:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dabae08ebe6fb88d4dc2873b12968e3e1dfb0afc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 22:05:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f412f906afa45a2a656b295d875121e : SUCCESS in 4m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1d0a4b9e678d46ee9583a70cd1dddf63 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7bd3adf9afd46e4ad098bc31157876d : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/adb6daa7367449d6a85370c0539f87d2 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b08845efb99b428b875e62ffdcc3f22c : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae71d0ca1fbc4eca949de4930592e22e : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65b70380cc8b4f42b8d724a0761e55a9 : SUCCESS in 4m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6981759793494a36a38e5ed078f47aed : SUCCESS in 1h 15m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10841d7676ab4f6a917d3e9cf97def9f : SUCCESS in 1h 10m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/14dd6f6c18d74cf58116657bd0299097 : SUCCESS in 1h 43m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1fb4b74eb5804603981f6db871eb44fb : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1f8b34a1cdf9da3c33a4004d153f63da9aee67db","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 22:07:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4ea8a75d58712132aeb23f678cfb143dad460d82","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-03 18:43:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d3d15972437f4ae030e9a5b60986b3d2da63ef9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:43:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"042bddc83bf5c654818acbc590f0deb3c5ac7aea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:47:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e56249cab694c39a91f06cde2459560 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9499aeda2f2a48ddb38e359855ce55d8 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36d64da0230340f19f50f25ad356f8f0 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26b2ef3b0bb2463c9f38e7881199f829 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/763e2de44ef04b84954ea0afe3b3fc2a : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/531eb939a9174d7b9492310427302a90 : SUCCESS in 4m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/936cc4986feb4ef593352ea52f3a7ea7 : SUCCESS in 1h 14m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cc655eabd9e4e35ae7530047d39a1f2 : SUCCESS in 1h 23m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/227b3c47d2e04ea785976fabdb32e1e1 : SUCCESS in 1h 42m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8fd3a07be6504feb81647406061686a5 : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":2},{"id":"47475fd2e05149658b7c4fb0735a44133d266637","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:47:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"20be78809abec833a4a96e8795995bb8e87a5702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:49:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf82fe98dfa4ba19c98f762821b6448 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24db3f89537551613af0d742ec62669c9f0beda0","revisions":{"e69331eca49d2165a2996944b55301db40286dc6":{"kind":"REWORK","_number":1,"created":"2019-08-29 11:36:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/679238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/679238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/679238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/679238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/679238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/679238/1"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-29 11:35:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-29 11:36:11.000000000","tz":180},"subject":"Adds ansible-lint to the blacklist","message":"Adds ansible-lint to the blacklist\n\nAvoids check-requirements failure regarding usage of ansible-lint.\n\nChange-Id: I36602e2f4ca1542b6c60df1ae9fcdc551451a3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e69331eca49d2165a2996944b55301db40286dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e69331eca49d2165a2996944b55301db40286dc6"}]},"branch":"refs/heads/master"},"24db3f89537551613af0d742ec62669c9f0beda0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-29 12:39:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/38/679238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/679238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/679238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/679238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/679238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/679238/2"}}},"commit":{"parents":[{"commit":"281a03d71bd80ab341e6659fb5095bcf18722710","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/281a03d71bd80ab341e6659fb5095bcf18722710"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-29 11:35:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-29 12:38:18.000000000","tz":180},"subject":"Adds ansible-lint to the blacklist","message":"Adds ansible-lint to the blacklist\n\nAvoids check-requirements failure regarding usage of ansible-lint.\n\nNeeded-By: https://review.opendev.org/679101\nNeeded-By: https://review.opendev.org/679105\nChange-Id: I36602e2f4ca1542b6c60df1ae9fcdc551451a3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24db3f89537551613af0d742ec62669c9f0beda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24db3f89537551613af0d742ec62669c9f0beda0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
