)]}'
{"id":"openstack%2Frequirements~679381","triplet_id":"openstack%2Frequirements~master~I179da689d619af2ab9555e74e0284ea6dbfff99c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I179da689d619af2ab9555e74e0284ea6dbfff99c","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-08-30 06:26:08.000000000","updated":"2019-08-31 03:08:53.000000000","submitted":"2019-08-31 03:07:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679381-1567220830831-91f773fe","meta_rev_id":"adbb79b5967c31a1a92575199d8b104de24e2f4d","_number":679381,"virtual_id_number":679381,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-31 03:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-31 01:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-31 01:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-31 01:12:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-31 03:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1309b89fb090c26574230575df51359837c924fd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-30 06:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5f1ee6b7a4574b32b17b60912daf6685365e920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:45:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d079a455c97547ff9835d6aa3dccfbd9 : SUCCESS in 4m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a6a3c08dd81c4208aedd63c867eb1213 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc37fc0a96b244279db5f69aff05667a : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d43e7b25390b44a1823741bf0734785c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b76530897ca245c59286c409094c4937 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46638c3a22704d59916d1c4a5fc2119a : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1621abf98974c07b6ae01c55a4bacb3 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/668f94943fd943a4bd2ddabec5517898 : SUCCESS in 7m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ff0964bcf7a2437db0d5cae0555bc589 : FAILURE in 13m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/319eb02db251424aad7db4288c719833 : FAILURE in 15m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a4af3ec9dff14539b30dcfe9b452753d : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7811bd65f1a04fbf857481e2bcc74568 : SUCCESS in 8m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/11c34b37418a48b5a75ca04606c60553 : SUCCESS in 5m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/16c9214392fd4ec0af1740422fade7dd : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0b918a613f7b4d098cbb857d96f906cd : FAILURE in 13m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de0d9205caa34239b456257737436b2d : FAILURE in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f040e2a69724c95b1d66d20479a4733 : SUCCESS in 10m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/14812b92e9cc40e7961966601bd72811 : FAILURE in 34m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/45da340886a34862a6c949b20449e26d : FAILURE in 34m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/301e31aede8a4a4e9b9c15a3e6ac22c3 : SUCCESS in 31m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/16756dd81cdc4ec79a71909baad1945d : SUCCESS in 23m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/18bac079b4004880889325572746ae80 : SUCCESS in 15m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d35b25b274d04e36ba01ffc41415336a : FAILURE in 14m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f608290da7544b38bb519a37476050f2 : FAILURE in 13m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbf788fb540f47858483bda41def4816 : SUCCESS in 19m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d60606ecde34ddd80fc4b269dd4fa56 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2bdedc178e184fbf87cd28189af44216 : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/94caa5e0c7c148f8a4eda5f08fff5151 : SUCCESS in 10m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7c12a64824c241efaf19c2f42908ece3 : SUCCESS in 1h 04m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a456de97f2fd4fe39f0f46301e2e7646 : FAILURE in 1h 12m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a86518dc11a940fdbc19234bd27456fd : SUCCESS in 1h 37m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1af78a9182d2459cae8126a5fa161086 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"18ad0832aef83c42965b0b1ea25bf260062f070e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 17:31:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43e67c0a3ba995f4536b5b61db733240af1a30d0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 17:31:53.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"8543f66ab9b2cdd2a54a574f0849ff9f0d552759","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:48:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8454dab7a70d48428c4d19b22fad918a : SUCCESS in 6m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ea1985b6efa84824b1b047e71a6b615d : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48812a8186da4efd837b61a0a1e4634b : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40dc55c94ef64f2c8749e9245709024d : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e787b1405cd4f05be5fb366a2556ae4 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc583d8d5df9467dafba5e6114f6e442 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eca5ca24a6a94a03acabb80ed5330572 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f12c07a622c049229fd93b55ba764971 : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5ba2295e175e4ab2a3e383cc5718af0a : SUCCESS in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37b686ce3004440e8629e89d0b8dbb28 : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/44fcbf49bac84a9eb90aeae8251be8ab : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bcd3b6a357ca471d89e6384de38455cf : SUCCESS in 11m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/79cf9aceb3d54904886c95dc2f9ce013 : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cf709759f2f94266979a12fd6053814f : SUCCESS in 6m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9f69d3fee9ce49a991c23091e03bdf4d : FAILURE in 11m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d09ea4d1611e4b63a8e97add76ec601e : FAILURE in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4bcd1b9e2a8b461786c8f386d3a4c32c : SUCCESS in 9m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/19bda0da5fa849b1bd760f55433b1053 : SUCCESS in 32m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dcea2ab1e4474a1b92f4ce7ab8c2de48 : SUCCESS in 37m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b5030d45452a4197b77cdeb1e636abdd : SUCCESS in 25m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2cc55e1b7644400986bc1090a5e6929e : SUCCESS in 27m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d6131014d97c41c4bd6cec1c9c8223d1 : SUCCESS in 17m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/539dfd3ad7424b5580d9f107f5fe41b0 : SUCCESS in 14m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6ab93fc1e64441218cfe240511e60557 : SUCCESS in 14m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2048bcea23a49728f6914fd7ff747ec : SUCCESS in 26m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a37395d755d54e82aa116eed1d3d7d23 : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/99e4903cdae84388a2b355859e5f0fe0 : SUCCESS in 12m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/156590a399e24e7093f075ca8a3f1913 : SUCCESS in 10m 29s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6218a371e23a4a1ba5e5aed1622e8c29 : SUCCESS in 1h 17m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9626ca8e59fd40b98ab6c1485ef8d099 : SUCCESS in 1h 14m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0cb84afff0fd44f39fc4ca850d58982c : SUCCESS in 1h 41m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dfc28f7e13de4f9d8b671925b00ef192 : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":2},{"id":"623f1f31a7d3392edf2cbdf0b9256892e31d3712","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 21:02:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9d350cd605a4628b50dc9b0d5f79d3bad3995100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 22:52:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a183106472841a7ba6edefb2d893ca3 : SUCCESS in 7m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4588669f600243c29edc8b279049d8d0 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dace92b1839847dba55d90785b4e84bd : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55c2e935889f4001b355ec5d4a449a85 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f576c5667ef447aaf5a381917e33138 : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95f545887d7f4f13b026bfe68d73fbfe : SUCCESS in 9m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c10f89b3bc146a383a32c2446f29648 : SUCCESS in 8m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0edca11fcf124fe8be8998854ab6bc3e : SUCCESS in 3m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3ed74d0b98d44579a487ed7235ae394f : SUCCESS in 21m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/991a324afa1e4e9d97b8649c97d0c345 : SUCCESS in 9m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ee6a15dff5c74f2480891de139e717c1 : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eeef087704804a0ca3fbc825991c14d9 : SUCCESS in 10m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d7159f0a81a442ca46a1377122f3e12 : SUCCESS in 6m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0bab9961a3a84b0d88713cf6f091c8ad : SUCCESS in 4m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/05d8cea686a6476e9a1bc8387e9a6cf1 : FAILURE in 12m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2a0a234d5abb4812953d2474b5ecd1dd : FAILURE in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6eb460179f50473ebfa54107287ac40b : SUCCESS in 7m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a93eabc069f14857bfc322f87f845fc9 : SUCCESS in 31m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/580f554071f14dc98828d0cbb8452210 : SUCCESS in 35m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4263ff3ea645484a9e236343e530c45d : SUCCESS in 23m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c70f4fd013fe4e66a1bb72b0cba6ac36 : SUCCESS in 23m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/710787ae33c14d9eb699aa41c75b0c84 : SUCCESS in 17m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eae9929f7a974dacaa0a3a49e309d795 : SUCCESS in 13m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/11a8a600baa345fa8f4d1352b093e4a2 : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3be85269423647b683687c3cb8f13802 : SUCCESS in 19m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6ab3987f6844c99a20bf3592d62654c : SUCCESS in 7m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/994f1ddd3b754f40a12a7bd1d185ff92 : SUCCESS in 9m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d9b5ccbf74784bdeb6b078b729c64d0b : SUCCESS in 11m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5df323b2bf06413ba3a912033ab50bbe : SUCCESS in 55m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f265efaf8abd450eacfc9c783543dd9c : FAILURE in 1h 02m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a6cbb3dfbf1647dd891a71f4092cb566 : SUCCESS in 1h 44m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ce2d60980b1045d3b1a3f169c2a1f04f : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e2a10aa23a253e97e6875c8db84bc85fa67cb985","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 22:58:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b8bebfd164266d85ad6290b3c611b0f7c8f9202","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 01:03:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2a31b99f9a4810af7a159df64bb11b : SUCCESS in 5m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1bf399fa4b7e41f38056b29efb0cff51 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcecda7fb8484b18983955853e161729 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b957d993619340af86adf9c968035ce6 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b04bf6fdabf446bb103bbf87492cc46 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1ad3d2658ab419c97648c4f4b387dd8 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c99e78ac71ce4f5e9ff2354986131557 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/429f41645b664ddb8ec1ae317808416c : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ce9884993428474e9b31c771b0488d9f : SUCCESS in 9m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7547dce7be3e4952befe5e8c7a272b03 : SUCCESS in 8m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d15b33ebfb664cce8bad5f4fcd0b0364 : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3fc9ffdc2c40485587f086b6ebc413c9 : SUCCESS in 11m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b57959f453c54189865dc63ab0b25c63 : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9bc0ad125fb0461293b921c8b6e2c3cf : SUCCESS in 5m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/831d163fb23a4151b6db4e05e353cae4 : SUCCESS in 10m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6da070851e8740db8087b76049b9fde7 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3383f4415b704e90b0b15c64c298c981 : SUCCESS in 7m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e08f9b97e3ec49b0b674b0f2ad88655b : SUCCESS in 30m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a3b11c3813148ed90785dadc03f5bbc : SUCCESS in 28m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/28143f3a260b42e29e9c41f1d57afaae : SUCCESS in 23m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f0d4475e293c4dd4bdca0d5d7ca9ff37 : SUCCESS in 21m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a65c4bed16b54bf184d522b72dd83129 : SUCCESS in 12m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/41c9ab60d1844ea18b32a0bed2879d05 : SUCCESS in 12m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/33fa3285d8204046b234364e4284be55 : SUCCESS in 12m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecab40348bca4f05a4677d5f9a20a7a9 : SUCCESS in 19m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09eebabd705d489bab728565e4a2dfdd : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eed6a26f30ec48adaf92ee0305361110 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d81108c7b0b342878e08ed0f2fc23884 : SUCCESS in 9m 54s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/129ed213585a430098a521182a996d6a : SUCCESS in 55m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/386965dc8bc94573872f9726b67dd2c4 : SUCCESS in 1h 08m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1eedeef0dea04879aac6aa99b7796aed : SUCCESS in 1h 34m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8dbe4cf196014a5896a463ef90e7c6d2 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a37e0d91bcdd81e536aaac8ce3c85598a4f925b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-31 01:12:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a88c6d0914670a46e03d0cf2569f77c5bf3e05d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 01:12:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"475f6f2a2896437285e81c57bd9aad667bef742c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 03:07:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d611822222b45f189650a5c086293bc : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38f4052c9694d7fb7b4cd0027356773 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f776c8cdaa5247fc9d4347bc836266b7 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4097036635944b899a203c879440c247 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c694b7afd23849188eee1dafe399914d : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfe7fc26a2314f40be3c95726a5b38cd : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ae0c3eed58a474ca31748891428945a : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3e091756e2b0413086f44e5236073b7b : SUCCESS in 7m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a7a5e9e1ef54c6c805f4f470e2a0b21 : SUCCESS in 6m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ecd76a5f7b2f4c1c8e0b812b2a508e06 : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c294f16c2dd4a188c8c166b5f5db7fd : SUCCESS in 12m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f6440c98dc044d92946cabeebce29ef3 : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c061d58d88b4adabb9f3265e004663a : SUCCESS in 5m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4021ed444c034083899abfeecb935ad5 : SUCCESS in 10m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53fb3d5778634743ba0fa660b29dcd7b : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/871251b92e3f4533a4e9de32297a0c36 : SUCCESS in 8m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3b7aa9cc3f9245ef97bbaa2e35db7ee8 : SUCCESS in 39m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7edef55f04b946db8cb1436e73eb828a : SUCCESS in 29m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d9675603071e4538bba22087dad976ed : SUCCESS in 27m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c157391f158d473296f25ee83cf0c5a6 : SUCCESS in 29m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ab0644fbb1814bbbb27e4c33a532d766 : SUCCESS in 13m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1e7f65254f7f4003839b5150d3d144d6 : SUCCESS in 17m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c3479c8aac0046a9907f0bbe4f206ada : SUCCESS in 13m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9955728f0bb4c5ba7fabea63fb33cfc : SUCCESS in 17m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/095318ed6b6541cc8c29115b2e974cca : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a3bd404301794a33894e4b724229b891 : SUCCESS in 9m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f0cd7e8019bf48528612d04af2cd4f84 : SUCCESS in 9m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3b1cc76040e24b2cbba2c62d924fb102 : SUCCESS in 58m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/934b82557c6a425f86bf00f35944b920 : SUCCESS in 1h 05m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ac0bd07127e4658bebf1e79b60465ff : SUCCESS in 1h 27m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/50af8a6e94354e7cbf58915aadc41a65 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":3},{"id":"ea8c9b8b05218850a77ff5b885cced4b1e013e22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 03:07:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"adbb79b5967c31a1a92575199d8b104de24e2f4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 03:08:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389e7b150a34454398020de2c806b8ab : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"54802078e0cbd15bdd97f24a1d084da83cf0ac97","revisions":{"f2956aedaa96f4af3e9a1005bcfac1731b8bfc53":{"kind":"REWORK","_number":1,"created":"2019-08-30 06:26:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/679381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/679381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/679381/1"}}},"commit":{"parents":[{"commit":"49f889fc02dab51cc0c9b87697442039994b0a5f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49f889fc02dab51cc0c9b87697442039994b0a5f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-30 06:26:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-30 06:26:06.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I179da689d619af2ab9555e74e0284ea6dbfff99c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2956aedaa96f4af3e9a1005bcfac1731b8bfc53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2956aedaa96f4af3e9a1005bcfac1731b8bfc53"}]},"branch":"refs/heads/master"},"94e142777954e9f2e4f4443ac7929bc465a7435b":{"kind":"REWORK","_number":2,"created":"2019-08-30 17:31:50.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/81/679381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/679381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/679381/2"}}},"commit":{"parents":[{"commit":"2908e0238ed281ff6d82d8d7142a06bbb9ec14cd","subject":"Merge \"Update constraint for ovsdbapp to new release 0.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2908e0238ed281ff6d82d8d7142a06bbb9ec14cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-30 06:26:06.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-30 17:31:37.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I179da689d619af2ab9555e74e0284ea6dbfff99c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e142777954e9f2e4f4443ac7929bc465a7435b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94e142777954e9f2e4f4443ac7929bc465a7435b"}]},"branch":"refs/heads/master"},"54802078e0cbd15bdd97f24a1d084da83cf0ac97":{"kind":"REWORK","_number":3,"created":"2019-08-30 22:58:27.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/81/679381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/679381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/679381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/679381/3"}}},"commit":{"parents":[{"commit":"2908e0238ed281ff6d82d8d7142a06bbb9ec14cd","subject":"Merge \"Update constraint for ovsdbapp to new release 0.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2908e0238ed281ff6d82d8d7142a06bbb9ec14cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-30 06:26:06.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-30 22:58:18.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I179da689d619af2ab9555e74e0284ea6dbfff99c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54802078e0cbd15bdd97f24a1d084da83cf0ac97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54802078e0cbd15bdd97f24a1d084da83cf0ac97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
