)]}'
{"id":"openstack%2Frequirements~679534","triplet_id":"openstack%2Frequirements~master~I7fe0ca4689b0fa99fd8ec5d3a1c704ed77062157","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I7fe0ca4689b0fa99fd8ec5d3a1c704ed77062157","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-08-31 06:14:40.000000000","updated":"2019-08-31 22:03:29.000000000","submitted":"2019-08-31 22:02:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679534-1567288930441-c9d86742","meta_rev_id":"e2127036e01ed6bf143bb7a9cea2b985b7d1e1ae","_number":679534,"virtual_id_number":679534,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-31 22:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-31 19:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-31 19:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-31 19:01:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-31 22:02:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e138ae3a2b0adc4f508fc5c90f193baefa65564","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-31 06:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfe3eb098d242a053d03b4f1264540ccb974a099","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 08:07:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d291a1035d43b9901df1f41ab652ba : SUCCESS in 4m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c440350860a24ed492b36442ffea6c87 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcdee9ca68f349f0b589477beedb1196 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc33ad3d3d0e48458d539c7bbf89b060 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df5bfbd582ac46cf8891400ab8b5646c : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b18fc02017d4bf1a5ce77175d7d4f3a : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2c00dd968124b918acef0c66098fe16 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a958ded38ea421786b412a687056311 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5014c20f96f7425c93654bcabb93ba10 : FAILURE in 8m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66b72ef486ce4cc2b43304c81597e6f0 : FAILURE in 8m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5d88ff0acab04577bae3631fedcd4fa7 : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aac7188fe0f14000a49befe970c06251 : SUCCESS in 9m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2cfbfa661c4b4557a5a28165e5256927 : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0f8f6424647741c38ee654774c03aa25 : SUCCESS in 5m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5ba41dad0e0c47e886821cff2951e4a1 : FAILURE in 11m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ebf8b1ae5d7c4b20a5c7eb0bd9a8f32f : FAILURE in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34ce8849eff54c198f703423dd70e368 : SUCCESS in 7m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/04f7ed3e9cb9460f9dae3abd6ad08def : FAILURE in 30m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f64425468a9341d28cb5a0de5e44475a : FAILURE in 31m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/22521e64b05640b1955b13a55d0b0a22 : SUCCESS in 24m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8f6431d0bdac4be0b9f60e2076ee0384 : SUCCESS in 24m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b3735cdf4c8c486f8e9495fc3f31d8e8 : SUCCESS in 18m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/deee1d933a97423da62d5f8bfbe84a23 : SUCCESS in 15m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/31cb32694a33484795f6db1c96d8fd91 : SUCCESS in 15m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51059665a7c24381837bb931683d98dd : SUCCESS in 19m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71616a42c22b41c687028dc770e4fd9a : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7022cf5fe754b7d8f653721f1aa4828 : SUCCESS in 9m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75106e35d58e4e38a237c07346132ee7 : SUCCESS in 9m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/effcc6197294427a9cdb8b22eb67d4fe : SUCCESS in 1h 30m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e9453a0c20544929fbe3c63787cd5e4 : FAILURE in 1h 11m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d275dbb9e563457183907f921402c4ea : SUCCESS in 1h 44m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/292ccdd63ed44da2b968c07540d50849 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"99edc8f1ed18b88873c916a1e073e00605d9ce7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-31 18:38:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"273e5af5e708ef9fedde1d0f2994a341d08a364a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-31 19:01:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nsemantic-version is horizon\ntornado is https://github.com/jaegertracing/jaeger-client-python/issues/242\nansible-runner removed a release...\nstestr fails\njsonschema needs python-zaqarclient\noauthlib is being worked on by keystone","accounts_in_message":[],"_revision_number":2},{"id":"d42f240e2f8d352f8f4cc52d65ad7206379bf07d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-31 19:34:56.000000000","message":"Patch Set 2:\n\ntornado should be fixed now too :D","accounts_in_message":[],"_revision_number":2},{"id":"7bbc21c4582ac9e77d8ad32d963a629dca54557c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 20:24:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c849a83226492a90d7b511bc58666d : SUCCESS in 4m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/049043c8a4c94dd8a2120e6a848bedc9 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492e6c5f82c945ada22961c17b0bbc38 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/563a3839c9e84af79d57f296dd51a76c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afbe5b702bf54825b24ba45e07d1601c : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59596bbfaab84d4191871165e20ac9d7 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a47738fe751462ba003fbf6da66e34d : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17abc26898554163a121173c71953534 : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f021ac497ed64eb79a019bd8fef2ddec : SUCCESS in 8m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/63409720e8c848708334a75292c9d2c9 : SUCCESS in 7m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3784fac5e93b469e87325513236851dd : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1fc35c4c340b473fa60094c49e05a700 : SUCCESS in 8m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ca981eb2267c46369e94e91bb55952c3 : SUCCESS in 5m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49c95c3d10fa4fafa9b7f46556ddb7f7 : SUCCESS in 5m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/989c003e98134e03b5a38c3b8cb3eb3a : SUCCESS in 10m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/394b52c5d91a4432bc0f9d6dfa964b3b : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4423624b9dbe4d96b264488628c866bd : SUCCESS in 7m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5b6c9c5ecbf840e9a84d71623a17bb6e : SUCCESS in 27m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/12851ec76d974f5d802434c685238169 : SUCCESS in 26m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/64416c3620a9405395249ae29958859c : SUCCESS in 22m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b2c170474fe4c8f8b10322a84363997 : SUCCESS in 23m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c1521c9653f04a0ebb0826ac399130d2 : SUCCESS in 13m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d3a8872564da4960b687cc1bd6af0a59 : SUCCESS in 13m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3be650796b934a3cadcbda47ac775c35 : SUCCESS in 11m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcaa41b313e54a2bb4a9900096cbf113 : SUCCESS in 22m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78fcd613e6aa4165ae6ae4fbcaaea57c : SUCCESS in 5m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/288e366ca2354cccbd86f1f8ffd21dd4 : SUCCESS in 8m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/82bb3d2d5a70418cb407e25d58171c67 : SUCCESS in 8m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1bb7921665eb48b1bfac052def208242 : SUCCESS in 56m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/148ebf8c084d4b6188d7cf09c2bac68d : SUCCESS in 1h 06m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/46e0dad8f1474e60b454413064a7215d : SUCCESS in 1h 40m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/59bf43d65d684cb68c37ebf7c00b47bc : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c5a8963413192997d3a3a822248741aeac3849a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 20:24:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"364d8d7e73c7fd977fc3fdec4e02f62d010827d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 22:02:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aab2a5df8e04654a9bac6adf35ec3c9 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe41b4a61a1c4236b66f332493d86141 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e691f74044b94764bd7c05b0d4382f49 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a975bc4d42d4c77af9c6540e8762c02 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d20aebeff5854f42b6d7bd67962d1b37 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/781eb709155d4923b4ec67f857a8dc5a : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/860013471ca547dd875a45bf4f881a3b : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7985775a534c4aa6ad21b50b180db708 : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/854139d697704d3f9d9f52319ef868ab : SUCCESS in 8m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e76938fd739942cb90a54b051af077d9 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/87fdf9793a8947e489a69d2da5d34085 : SUCCESS in 10m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/04ee30aa677a4051bc9852e3685fc9b8 : SUCCESS in 6m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a1c3f4f6137042288f23e4c596af8493 : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/80408fda0c6e4c208f58faec0e1d135f : SUCCESS in 11m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7e51efc30d6b4d2a961d52a267b50ba5 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3dca6de6b47c401b979138a381e149fa : SUCCESS in 9m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/260cd906963043759358c78c644b720b : SUCCESS in 32m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/38a89b359b5e47f29ec1b20d5d0a701f : SUCCESS in 28m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f91759e0b16f4ce68c0030e83998e4e0 : SUCCESS in 20m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1b649a315c5e4fbe8b25fa0f4912d38b : SUCCESS in 21m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6855ae7e2a974f169513e2433c6cdbe4 : SUCCESS in 14m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/983a607f0efe41e4b1a7b51709e42722 : SUCCESS in 10m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/982d603e63524382b941fddf34f7e51c : SUCCESS in 12m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26d3f22dd04242c79078e0be249fa0dc : SUCCESS in 21m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18f384aa94fe4231804b6245d6c8791d : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6dac76109507496a9f51beb7e13884f0 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/779bfafd7abe4534af1caeabba96c31a : SUCCESS in 9m 07s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/66d8e0289b9f42da8c78f07437f64190 : SUCCESS in 1h 09m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22d24de50aa64f3eaf67d7399d102560 : SUCCESS in 1h 08m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1626570ea2154959bb87962385128eda : SUCCESS in 1h 35m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5af9c4ae6c28475c9c9d86ee073cc2f7 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9d33f4ee05091b6b8a6970bcc79cc0ecf4ada78c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 22:02:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ba4eca9ae6a701ed898f163b015d2750a648cf8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 22:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4d4cc027904b5a95984980d281a4ce : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52a7ea4000042a441efdf7acfd417af069f8e6f0","revisions":{"430aa248794c2b698df3b78ed274c8bafd3884b5":{"kind":"REWORK","_number":1,"created":"2019-08-31 06:14:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/679534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/679534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/679534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/679534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/679534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/679534/1"}}},"commit":{"parents":[{"commit":"6fda2f6ee7cdfca0c4c4770761fab5998a66944f","subject":"Merge \"update constraint for taskflow to new release 3.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fda2f6ee7cdfca0c4c4770761fab5998a66944f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-31 06:14:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-31 06:14:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7fe0ca4689b0fa99fd8ec5d3a1c704ed77062157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/430aa248794c2b698df3b78ed274c8bafd3884b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/430aa248794c2b698df3b78ed274c8bafd3884b5"}]},"branch":"refs/heads/master"},"52a7ea4000042a441efdf7acfd417af069f8e6f0":{"kind":"REWORK","_number":2,"created":"2019-08-31 18:38:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/34/679534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/679534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/679534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/679534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/679534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/679534/2"}}},"commit":{"parents":[{"commit":"6fda2f6ee7cdfca0c4c4770761fab5998a66944f","subject":"Merge \"update constraint for taskflow to new release 3.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fda2f6ee7cdfca0c4c4770761fab5998a66944f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-31 06:14:37.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-08-31 18:34:38.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\n-semantic-version\u003d\u003d\u003d2.8.1\n+semantic-version\u003d\u003d\u003d2.6.0\n\n-tornado\u003d\u003d\u003d5.1.1;python_version\u003d\u003d\u00272.7\u0027\n+tornado\u003d\u003d\u003d4.5.3;python_version\u003d\u003d\u00272.7\u0027\n\n-ansible-runner\u003d\u003d\u003d1.3.5\n+ansible-runner\u003d\u003d\u003d1.3.4\n\n-stestr\u003d\u003d\u003d2.5.0\n+stestr\u003d\u003d\u003d2.4.0\n\n-jsonschema\u003d\u003d\u003d3.0.2\n+jsonschema\u003d\u003d\u003d2.6.0\n\n-oauthlib\u003d\u003d\u003d3.1.0\n+oauthlib\u003d\u003d\u003d3.0.2\n\nChange-Id: I7fe0ca4689b0fa99fd8ec5d3a1c704ed77062157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52a7ea4000042a441efdf7acfd417af069f8e6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52a7ea4000042a441efdf7acfd417af069f8e6f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
