)]}'
{"id":"openstack%2Frequirements~679614","triplet_id":"openstack%2Frequirements~master~Icb768f17baf5677173908c4258d8327e27681ffa","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icb768f17baf5677173908c4258d8327e27681ffa","subject":"update constraint for python-troveclient to new release 2.19.0","status":"MERGED","created":"2019-09-02 09:37:07.000000000","updated":"2019-09-03 06:22:46.000000000","submitted":"2019-09-03 06:21:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679614-1567491666052-26e757ed","meta_rev_id":"8c84a5978bb8209d24597f8391d4b54c2105a5d7","_number":679614,"virtual_id_number":679614,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-03 06:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-03 02:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-03 00:04:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 00:04:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-03 06:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1488c378f4b0f5a2293610426b973bd0e3741da","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-02 09:37:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcbb90ca2745dd0efce9625ab44e29f60444b4f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 11:28:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0988fb9c5d9b46838f0e53661dd7801a : SUCCESS in 4m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7d9f97cedddf41818e1773a08cffebac : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04fb0f39cfad44e5b3497ac673876bd8 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56d5355923984b9c95c18c124866adec : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d441e7ade33d45aa8446a8b73f922934 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4437b170f55c4092b3fb85158b5f1a39 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8da9d270b7364a2b9bc6a353e91b63bf : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae4a1a345e9b449e89d4cf450d4233d3 : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3a24e500a85449faa22aed42a08c5178 : SUCCESS in 10m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac702c792ae442ebb0495a0dcc80ec9c : SUCCESS in 9m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/085e93e23cf24383816f095f62cef950 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3649a2d156fd4da8a782509c5a1bbbab : SUCCESS in 11m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cfb00377454e4550b5d8b1566f06bf4c : SUCCESS in 5m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/961f615b1ee04e1f8e092b36f1c83f0e : SUCCESS in 6m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/de14edc681e34c8abe0c21007356a9e0 : SUCCESS in 11m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/09383ab47a634fa7b7f7ba2c69feaee6 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2105e84c7fb4496ebe0e5abd9f55cbbb : SUCCESS in 9m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e2516bbbacc245b8b1cc56aa1d341ed5 : TIMED_OUT in 42m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6640db6afdf24b9dab0d3c25b44b5bb5 : SUCCESS in 40m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/96d5ffd42a994cf2bf22e9da70b373ec : SUCCESS in 29m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c02179be68684869ad7901a796b0d01d : SUCCESS in 20m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4e93db63dca144de8ca068dfc34b27d8 : SUCCESS in 18m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3424264355844045bf08e7f37067f1d2 : SUCCESS in 18m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e3698204015e4b45bf4a23b5a052ef9e : SUCCESS in 16m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45d9fa2ca2f54613bbe6a3491927516d : SUCCESS in 21m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff86a56b2211424e9189b025faa62915 : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bb0d075983e546648c8891d0335241a4 : SUCCESS in 12m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/651ab0632d07485da71012e3d5c6f0eb : SUCCESS in 11m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5c122d798e6741a299c8701c1fce293f : SUCCESS in 57m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a07aa893894244ae923f414088682876 : SUCCESS in 1h 12m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b834c8d581904c709aad46aa8ee36f88 : SUCCESS in 1h 47m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/49bfa24c57714ee2b1082d66e61c3710 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":1},{"id":"71e4f0b90010c7cdade6c3dbec1ea869d7e6df96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-02 18:14:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7dc28598f48842df4f381afd2f8e3696bdc234e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 20:02:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3988cd7cbae142f396cf9bf3c8cd7b68 : SUCCESS in 3m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/88670b6a099d48fdb208de486274a7c9 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e222b9ad29d24b23afad5b36c6d44610 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d8488d99d014db5afcf8c255155061a : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ee8cb96346f424898943200c394a9ec : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b83bfd281f7a46a58fa85274dc69b4ba : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b40d10872d44fa2aec77fdf357b6c61 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdd377043fd9437f85b637ee4a0c1603 : SUCCESS in 5m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f6a5260f78854bf59a7cac14f3de51b8 : SUCCESS in 9m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4739012763be4a03a9208e465011dd1b : SUCCESS in 9m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9427f9ab2c2f41ac9df1a3aab5dcf596 : SUCCESS in 12m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1a905fa5d82c45eba7d7625aa9fac606 : SUCCESS in 13m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0ccea45269434b08819effc0345fcd48 : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/443d570b0b9f4ad09c7c9e555c9fe63e : SUCCESS in 5m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9301c05b837b46aebbad9320f9429c61 : SUCCESS in 12m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/296679d8671d4482837de20e41915dca : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/12152fe1eae04f4880f34944c1e97215 : SUCCESS in 7m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c898a8741a50451498c9a4dee24be33c : TIMED_OUT in 42m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de32664ad4bb4663936d9d331e2a6722 : SUCCESS in 37m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2404e3ef971c48258a72a46560b909fd : SUCCESS in 26m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7af02b88aadf4316aaaaac6bd242e9df : SUCCESS in 22m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9cce66969b684c99a486d6411405f805 : SUCCESS in 22m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9523930a4dc4f01847de30884c71e46 : SUCCESS in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eee466d0ff074de19228cb62d117edc0 : SUCCESS in 23m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c6eb4af47724b6f8940e761de93c7c2 : SUCCESS in 24m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/434a2d2dc05344c0b3f6315f4c5b3524 : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fee2cd614c224ec8816a36898542e63e : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a9f8f5f7c3ca4b6984ca0e83abb70adb : SUCCESS in 17m 20s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/baf26a893ebb47769d97f7ad78bffd56 : SUCCESS in 55m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/826ac54c48c9464fa8c2de353700822f : SUCCESS in 1h 26m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a19eb3ab90c4568a130e467db33c2e5 : SUCCESS in 1h 42m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0aebf947b95f4041ba2b482bc9c15dd0 : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"51a357c9260ed0805f26421fc79558801959e392","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-02 21:45:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72f71be5f859b0b0ca9df6a2e14ba46234f1fd58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 23:29:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111c96c6a1084666ab2aef908f844023 : SUCCESS in 3m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5e2a1c8634914aa386d1570cf64a442a : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c9002d023c4f599287238269783b91 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0fcdd33d90e44daa89fcffea46e9d3a : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c3a78e784c34e71ba1dde2566436f3d : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a0e189223c04194a9d156e0f166f671 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/746bd8729dc246afb3b544074f8d3c3d : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4a3f9b0b9c541f48c56f75600ba7aff : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e1b68ffba3bf48f5846dfad148565aed : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/523a81a4bcfe4faab35acd20bd9782ed : SUCCESS in 7m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/066dab3d60c048d88e49edf8d9e8c0da : SUCCESS in 12m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4550f490116c4c6984586aabf59502d1 : SUCCESS in 7m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e5c6459350464f988ce08481868ebab8 : SUCCESS in 4m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4aed059e28634f40824631c2d18afa25 : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c46add0e21004e57b234f41f73de72f9 : SUCCESS in 11m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92c776ced0b5431abd8071c7d26a0b3b : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad8eff8039644214b93e7dcaa9fbacc5 : SUCCESS in 8m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1ea880d4b441481daea8992f164b1bd5 : SUCCESS in 29m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c9518a1aa764196940c79eb0a44cbf4 : SUCCESS in 29m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5944e112c7274c0f8468a970a51a3542 : SUCCESS in 23m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ca5959dba60a481c942bc1fa02085b66 : SUCCESS in 23m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7af1dd6e533a46bd9f23e308560be876 : SUCCESS in 15m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a76ee4598adf4bb0bb0883f2fed54d6f : SUCCESS in 11m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f0b35cbfbc1b46098ccca5487dd4050a : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83e9cbcd38044adb8a44af23297c3ace : SUCCESS in 18m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/702b2f1eb61a463cae99c338d8de2be2 : SUCCESS in 6m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4945e72eaa4d4645b7f59db2ab9d3036 : SUCCESS in 8m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f34dfd0e78a543618b8f7d22724dfda9 : SUCCESS in 9m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/287e2d2e1200433881f787e160b376ff : SUCCESS in 55m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d9cf1af72624a6397c0aafedae6ef84 : SUCCESS in 1h 12m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b0a19c7e8f6b4e25ac7fff291227bce4 : SUCCESS in 1h 42m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/98944ba97727407bbeebb2abe9787877 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d67c2988e8acdbc6d414be0493054b22c2a63a84","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 00:04:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a05850098dfe578cfd1beaaa0e3781ab15e97c76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 00:04:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e71aeeea321033f92ab1497030616e2069bbcff2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 02:15:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6092250a7241acb5f24f015a94a2d7 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad4a29b95994e6f84a1411de72dac0e : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05e0121b4fcb48c6842e6f2c550af922 : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd91e51d3fd445f8e8c36428c0e9f0c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03dd02b768874d73861e78c3914ef8fe : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/472f317a60044b53a5f316481d1d7c71 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/585494d56b4b43b98f0b74cd6fde5c81 : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/081d08c96545408abb03f68cd1c8b1b7 : SUCCESS in 7m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/363bd69f15034266aeae91a41d1d8f58 : SUCCESS in 6m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1b49ae6f7a1540e78961746c8891fe57 : SUCCESS in 11m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67f5b2113e4548f5bfd5aa35e11d471e : SUCCESS in 9m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/574a75a185db4a72a9759c79db4b6082 : SUCCESS in 4m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab9e6bd55e394db88954cbd0c5cb1860 : SUCCESS in 4m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b3839552b0944434a293c4b05c3053a3 : SUCCESS in 9m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/93141c1ea7d94da0904038d40bade19d : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0945c0543e0478597970a4993b0b952 : SUCCESS in 10m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/32ce52a7ecb747b8ace77f65ec379218 : SUCCESS in 30m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f9789dce65a44338b61edb4c68639215 : SUCCESS in 26m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b9b4ea26bd7343ec8987bfe9c27a0432 : SUCCESS in 19m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/91e6770dfbac4ec9be7a16e99e77684b : SUCCESS in 21m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4dc28c1158bb4d1982b5f940e16ee8ca : SUCCESS in 13m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/887e8cf617aa4a37a3b4c2146ec7d1dd : SUCCESS in 12m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/54be04a97b544b3586286050e5e3102e : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c60a609392ad4d54ae575c8e680282e5 : SUCCESS in 18m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03f8415335004202bd193a5ac561a4d9 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7af53c03681a4c70a2fdcb9f1745bbf3 : SUCCESS in 11m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/92d2ed9ccdd44ba4a853a1ced9484d99 : SUCCESS in 8m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/07437403216840389202443494407aa3 : SUCCESS in 54m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/75088649637f4f50bd0785221317b0e3 : SUCCESS in 1h 07m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/be12272b0b7d4d3dbcf8452a8e6a5ad2 : TIMED_OUT in 2h 05m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd11ed64e34b4236971a790fa884211e : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1a7a652e34afa099edef0471843070e8c00a357a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 02:56:25.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3cf765be5d9756b10ee3c81a349336d2533ed3a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 02:56:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba6ee53005588b3de97b6956facd271d7811df28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 04:34:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af4b4db38d74f9092456cb5ceb78aac : SUCCESS in 4m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cf38605771984f64b1b07c61c9f36392 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5527170ab3d4943aa19ad1d8d89a48b : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2172888dd614258b8922d8d1346f845 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d375b9f712bf4a548aba852989887377 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53b67dbb55f34dc9947fe21f6b9a98ba : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/169f8e9f24fa4b3ab313e32cbe1d103c : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bc60f054e6546d4b71e62d080fcc4e7 : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c7fcc7aedf804ef09ad3357bccda2879 : SUCCESS in 9m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4a06a0587c848548567ae47e1276940 : SUCCESS in 7m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/da52efc3865c47b4966a9ef386d1bed4 : SUCCESS in 11m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/472a2cf3dd81443c97550c5f40b74b13 : SUCCESS in 8m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/56491608617e49ef8b110daaf25402c0 : SUCCESS in 6m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d53556b86064f2bab7a53a79cd40735 : SUCCESS in 5m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d4043018da3547c7bb654cb6fa1149c5 : SUCCESS in 11m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/286c2890dc744c5c851d5689e45611c2 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0c60bba56a846ff960d682e06099408 : SUCCESS in 7m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/23e8593131a14e30b9a66c4699d065a5 : SUCCESS in 33m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b39667d0d8204f46a0a4eebc7e96ad87 : SUCCESS in 27m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/be2be7fd473648d193c1b97609c93e00 : SUCCESS in 29m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cee6792366f5464d9da909621a06ddf0 : SUCCESS in 20m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/05367430fcc949dcae039e7cc9422059 : SUCCESS in 19m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/47c8b34ff4a344de85e80ecb791bf213 : SUCCESS in 14m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/96647d74e67246be84c8c197eddf7fef : SUCCESS in 15m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4c8e62fa1fa4fe182daba67674d5df5 : SUCCESS in 22m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8d4515bf0aba48f7af1fca70fa11068d : SUCCESS in 6m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d0d958be3a5e4d2991c20b219ad2c860 : SUCCESS in 9m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a7c26a1c560a43cba3b88125466e152d : SUCCESS in 8m 24s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c0ce0d13b9f741e3aa0655a39ae8d118 : SUCCESS in 53m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfeb6b5a8ac04e818f2e1ed9a322a7f1 : SUCCESS in 1h 12m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c470c3fc553b4c56992281691cf407c9 : SUCCESS in 1h 32m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b27298b3c04b45e1b5deca80360508f2 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a481a0104f919e1327f225e51595e76b23372d0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 04:34:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7edbd14f53c72d4c891930012ccc9542f3d0acb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 06:21:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f877ca4d5e64cdd96e0da45a9596c0a : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854087b7cc8c4fa59a79c1660a8cb0b0 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e67138903723414bbd4bd320d02dd9cc : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9737d4bd483f49ada0bd5bf4ae7bc217 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc63419cd4274c11a98c4eff55721ac8 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/774cbd885cee46cca8ab774868a17e73 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c9605509d514f13820ca922dc215ba6 : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e9a3ab445695401b8d097529af907600 : SUCCESS in 14m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5fa1b18f56145809d199dd7f402390f : SUCCESS in 7m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/985cdfb3d882498bbebcb1514b2f0a66 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/076fb5c5c9ac480fa6532ab6ee655c06 : SUCCESS in 8m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fd09a0e8f70f4e9a97d912ae09a9198a : SUCCESS in 4m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e765d72ab62041c7b530207fc0277938 : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1b88135dc0ed43068853f1a45b3bc5b6 : SUCCESS in 9m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/23408eff57ed4294ba27447a0ff2d0a8 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/771b150bcca841539b517105880e6c39 : SUCCESS in 8m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/21ae268e9ad24f8c996fad3d1dd1edf4 : SUCCESS in 30m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e4395faa156467cbd0175977ed7506f : SUCCESS in 27m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/959b9be23f9e4312975f57d7bc3cdd2b : SUCCESS in 19m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/63aad7bc89cb4fe7a5ac8d8a8fceb366 : SUCCESS in 23m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d0522616f91f4fb7b29d505d8bb29dac : SUCCESS in 14m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d35fc0ba425143a0a6f137fba9a70b8c : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6e38071a434d4ffdb3363445f0e66d77 : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76913bc38123426e97dc3ece2faa7220 : SUCCESS in 16m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63a3d4001c33404db3f3c4b0f061f953 : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f3993bb8e6e43d1bbdf20f9d38c95ce : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fba3f4268c7a4ea3801bb216c623fa64 : SUCCESS in 9m 24s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ba46833e438545bcba4b1de50e9717c0 : SUCCESS in 1h 05m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1344d369777e470b9975b46856e12840 : SUCCESS in 1h 15m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d75cf54365dd490098374c98be23aed9 : SUCCESS in 1h 40m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c20d0fe7c2de47d29a13672a7d9b2966 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d93163b604b7c1840d82e1e17b4ba9ef65e95c88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 06:21:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f8e9e7deb1499c4c4d4d61bbc428dd65c0972abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 06:22:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5dbe099eb040fd8cea45460aacca37 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"926309010f7affc959e6367d4b2be2c2951e7e71","revisions":{"926309010f7affc959e6367d4b2be2c2951e7e71":{"kind":"REWORK","_number":1,"created":"2019-09-02 09:37:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/679614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/679614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/679614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/679614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/679614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/679614/1"}}},"commit":{"parents":[{"commit":"08b58dff06e34fd1f70ae5500f50f7461ba372ee","subject":"update tornado now that jaeger-client is fixed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08b58dff06e34fd1f70ae5500f50f7461ba372ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-02 09:37:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-02 09:37:05.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 2.19.0","message":"update constraint for python-troveclient to new release 2.19.0\n\nChange-Id: Icb768f17baf5677173908c4258d8327e27681ffa\nmeta:version: 2.19.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: I495965c12df4023f5036a25da51bc8380e53e4b4\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926309010f7affc959e6367d4b2be2c2951e7e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926309010f7affc959e6367d4b2be2c2951e7e71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
