)]}'
{"id":"openstack%2Frequirements~679615","triplet_id":"openstack%2Frequirements~master~I7ce72878eb782ac3f8a5e7f3a3344c54699e0117","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7ce72878eb782ac3f8a5e7f3a3344c54699e0117","subject":"update constraint for python-novaclient to new release 15.0.0","status":"ABANDONED","created":"2019-09-02 09:39:47.000000000","updated":"2019-09-03 00:05:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6890a91166d6bd005f86f0ae35316f46ac825b3a","_number":679615,"virtual_id_number":679615,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-02 21:45:31.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2019-09-02 11:21:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-02 11:21:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-02 21:45:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4b3da11201bcadd2f445d472697094dbb967426a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-02 09:39:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcc0894e6fd11cf0f69ca11d9f9475f4c652cb9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 11:21:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058086b944eb4c1da0a9ff521cfdc193 : SUCCESS in 6m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/61ae5578d649482088d6503565fd4712 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b937c6becb2d4689b0cb633e7e8d33f0 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0d17f84630c453eaced8b1e08fb0005 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5645acbda4f416a8776465bd63842c6 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/446be0d996ea4ef09d4a2736836aef1b : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d597c75f00545d4b55b699284d76230 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/066aefbef74647c5966043afb1a83c51 : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0576b679f32d498a8e235ae2e072b6a0 : SUCCESS in 14m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/78eeec3cce4843d48b6b36a343b66eef : SUCCESS in 7m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b90f3599afaa43d6be094974b8971415 : FAILURE in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2097f93f0dae45a8b0c799cda2e144e2 : FAILURE in 8m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e66e727fb8f84b118e5ca769483611e8 : SUCCESS in 4m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/801f7400ebeb47498922c82aa5d79240 : SUCCESS in 5m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2ca643e4563d420b8ce929c8d72bf483 : SUCCESS in 11m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db8d1c775be74617b7d3c4b79fcefebc : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0f58bf037254240b67a018467c51d86 : SUCCESS in 7m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/94bd6ef1500d4e34adb4fbb9c1ac13a0 : SUCCESS in 33m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d6a792fbcb843879817e489efac8a21 : SUCCESS in 28m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/664a674baa9343a7a8b0b46b65299e42 : SUCCESS in 23m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1a5627dc67c14e4e9baaeca387ef890a : SUCCESS in 26m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6a36f7f31b3c420b95f64ec0911213ca : SUCCESS in 12m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/653671377f804d0aac2af49b4fb1fabd : SUCCESS in 12m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e23b8f4f55a4f65874cde19f6d75f67 : SUCCESS in 12m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4275a1c091b472ba4050cad1ab1539d : SUCCESS in 18m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f955cab27d940b6a0a826da84fa40ea : SUCCESS in 6m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/962f899df56f40dea4f00e94b7a9cfcb : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e36a1d0a138c4510b7025611a6edb4fa : SUCCESS in 9m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8f7fbfc8e0d34cc48f29121ebc14dafe : SUCCESS in 54m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b63859a88da44e9ca492fe131963d02a : SUCCESS in 1h 12m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4d213c2425646efad703e5281dd5bb8 : SUCCESS in 1h 36m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4dedca96b83640dca8bc90d53f715399 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"faf460d9c035ca01be4b8598c0ef020c556d670a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-02 18:14:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"225aff202e0331e35087964fbc7b9e28087358fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 20:02:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7046f579db0410a8739ef3dfd514a70 : SUCCESS in 4m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9ffae7e29c6f456a9b73567b945333bd : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd4c77bbd2a4822bb2952510a1d0361 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3518a4e375734b188d7107e04ee06275 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e034d6a3004707b6a893c782c558c7 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9da7ec0148d64932963f9707884054a1 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0631f3bdcc9341d59d01b2d6a8288f44 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/973a8d9547294d6ab3fa25ad78e912ab : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dfd151994a724909b074d7007e747526 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/14604e1d3c314ae6a582d0efc6216de0 : SUCCESS in 6m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c09f76405e444e568b8cf9ab9c46b0ee : FAILURE in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8593411f2f9a432890513f396bf5e8ab : FAILURE in 9m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/15352a116096429594a89577d285d6a7 : SUCCESS in 4m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1fe2833aebea43d6af170f3f6baae4ae : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8d25dbee870d46528ad3ead7431ef2e4 : SUCCESS in 10m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/86ffe24fc83c410a9fd78ef48b637787 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77b2de3043bc464e9b009f16f3d3048f : SUCCESS in 9m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/29cf7246548d4919a042775a547e89b3 : SUCCESS in 32m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68b27e0d2b2846e0acf4a560e6354584 : SUCCESS in 33m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2bc7df1022734edb8ae15b13be7bed63 : SUCCESS in 22m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ad2197f9c84346cd976be9477c7f0f19 : SUCCESS in 24m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0b2a573f64ec4e02b6c03c8966a8d7ad : SUCCESS in 13m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/787ad99631b7476a86a54a30fc64acf4 : SUCCESS in 11m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/536ccb4b444140dfa1709bbf1d9af9d3 : SUCCESS in 12m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40253968c4644806bc9123f98cf2833f : SUCCESS in 20m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05150ff06b12426f95874d87788baa1f : SUCCESS in 6m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cbc0578a186b4607abbef72a995abe4f : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a98606fd30884d61bbf585adbefc8487 : SUCCESS in 11m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5ea4044f014849c2b2505a69d43836ff : SUCCESS in 1h 08m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/83c53df91463448db4a8670eeec75c8f : SUCCESS in 1h 11m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a07ea8358eac47baa451f497a64b1b53 : SUCCESS in 1h 42m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e32f73e156094e889d1da71a0e029213 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bb71f7bb2548d35a14bc542632a8cc797f988904","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-02 21:45:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c3931b537d1f08435726cb446b2957b4179966a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 23:28:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f1d0f0c6494aefa630057c5f5e3b19 : SUCCESS in 3m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5779653b625e4346be1d59cbb3276b17 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa9fc66709a4f609a9e9bbb612e07ca : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2fab737bc7f6494f91c7ec909e5eb097 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25c1d293c71b46ac8855a6068bb065e9 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e49a823e9b64236966786aacab6ca7a : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f1e18d5c2824cd4b67abdc64de868ab : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c11c3cfd9e2442ae835d98ae4af79fc8 : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/73fac4f83a5a4319aa57dfa608dac2ea : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d00c58b719da4d7aa585c376e6861c57 : SUCCESS in 7m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e8473c4499a84cda872f97840b55f4c5 : FAILURE in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1bbb7ae3123241c183ed4a8b356e10ec : FAILURE in 8m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2894649619b44ffe8a2076d9dfd20435 : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/64d1905867ff4f29b020417598bd2eb7 : SUCCESS in 4m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c077ec0c0bc545a68da934bfa25c4886 : SUCCESS in 12m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d407a16fb6dd43a4a752c7cde54b8b32 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13d720bf86b249658e222f6051dc77dc : SUCCESS in 8m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ea16eb65b30045239bfbd5326f6a7c7e : SUCCESS in 32m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64d21dc112074ba090d1760593958963 : SUCCESS in 30m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fabbce8555a64af9b727de47975e67d0 : SUCCESS in 24m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0acc6511c4cb448387c636babc7afcb2 : SUCCESS in 22m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0d5943852f7b4b6c9c88740de8a8eb92 : SUCCESS in 13m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9059014acd4a44f0a343c97cc675b9e0 : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4873e96282354008b8a2a45e05201820 : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ea7593622324baf9955f8c760a09693 : SUCCESS in 21m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fd70958b5ba455ab7ac3bb69ee55921 : SUCCESS in 6m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/445a231449e849f7ad930544c60ac09b : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/076fe6b7130444c69c17fb0209c7734d : SUCCESS in 9m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/eba119ff75844592bf699a4f0d6edb3a : SUCCESS in 52m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72327bfc20f447c283e30047051755f3 : SUCCESS in 1h 09m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e63f3448e8004e439878ce0e1db18f43 : SUCCESS in 1h 34m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2ea9f8626c354e65bf03ece5495f19f5 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b77ad529d44720a5bebea790085ec9d5dc6eb8fe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 00:05:51.000000000","message":"Abandoned\n\nuse the manually proposed one","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e9d77995c57960c6e41d304fb2a06544063ac99","revisions":{"7e9d77995c57960c6e41d304fb2a06544063ac99":{"kind":"REWORK","_number":1,"created":"2019-09-02 09:39:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/679615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/679615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/679615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/679615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/679615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/679615/1"}}},"commit":{"parents":[{"commit":"08b58dff06e34fd1f70ae5500f50f7461ba372ee","subject":"update tornado now that jaeger-client is fixed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08b58dff06e34fd1f70ae5500f50f7461ba372ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-02 09:39:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-02 09:39:46.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 15.0.0","message":"update constraint for python-novaclient to new release 15.0.0\n\nChange-Id: I7ce72878eb782ac3f8a5e7f3a3344c54699e0117\nmeta:version: 15.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I067e21cd3b33eb77850afc30ab9557d435c5d800\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e9d77995c57960c6e41d304fb2a06544063ac99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e9d77995c57960c6e41d304fb2a06544063ac99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
