)]}'
{"id":"openstack%2Frequirements~679704","triplet_id":"openstack%2Frequirements~master~I55f20ff8cbcfd74e4d6ddab11201cd4a759ec1e8","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I55f20ff8cbcfd74e4d6ddab11201cd4a759ec1e8","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-09-03 06:16:35.000000000","updated":"2019-09-03 21:27:16.000000000","submitted":"2019-09-03 21:25:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679704-1567545923212-4fefa147","meta_rev_id":"679217bc4efb81a421d66dc4f75ff1bd0b2dae77","_number":679704,"virtual_id_number":679704,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-03 21:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-03 14:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-03 18:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-03 14:47:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 14:47:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-03 18:44:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-03 21:25:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15d9b1397237960bfc487d097bf1c22df80fbd11","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-03 06:16:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ca8604daf109ecdfecca8f7952093a3f6b967c7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 06:28:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"636bc0bc76bf7272ad80679b68629d2c39bbcb0c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 06:28:17.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"5ea49057a35eb6d234ebaf43fd016e170bd172b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:32:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f03dcda3e0842b48d3d437fcb048459 : SUCCESS in 3m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5dda16618e4f4c8785ccd83be82c6e77 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5d76c2860c4985a4fb7f9f96075230 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a4bd6d6207949438a5e468b5841c2cd : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3ee57a8ca354ab99f17b6cce7e4f29e : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7973780df9c84dde99d7d95613623ae0 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/975c5d5e3d74461a97c6e429e2b2a55b : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5db4a6038f0942edabf46189ba6f5d2e : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/838e45461f2a43828a89cf0ec18fb620 : SUCCESS in 9m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4056abeeb04b4db9b6454906931c9972 : SUCCESS in 19m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/92ab0238caf84bb7b1caee94b76ec6ac : SUCCESS in 11m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5e138b172c264a44be8de7b780f7bec9 : SUCCESS in 10m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bd3482250ca0415e874ea430cddc0731 : SUCCESS in 5m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea1cada4d1c641e78627b30eb1d95cc4 : SUCCESS in 5m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a628f5b0fbae48ceb68f03a6be0f25ea : SUCCESS in 11m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2c653ce80b4e4e1a9879c64a5f078760 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b24505c4e484ff89f49e531b861d14d : SUCCESS in 9m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1dd72e0e0d90419da1cad9156277e010 : SUCCESS in 33m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fe67786abab44612bb4c04e11f2843a3 : FAILURE in 32m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/857088aea2fb43b79ad8645d7af54411 : SUCCESS in 23m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/628da0a6f5b449fa81d6d6fe2636807d : SUCCESS in 23m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8186b0fd1b94b64925d534ba7dda946 : SUCCESS in 13m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8b6139dbefe040c49c2ee063b56fbbbf : SUCCESS in 13m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/56d4bef622eb457d9bcd8f91e8e2535d : SUCCESS in 13m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49656634f96a4cd49dd19fd8cb9a13ca : SUCCESS in 18m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47dcfe804c96400aa2f9ca5b1eecbf6e : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4a2f549e1128491db6dd2a2a3083654c : SUCCESS in 12m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c9a8fec976b4495d95694deae715a197 : SUCCESS in 10m 08s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3b559efcdc25463c8ac45e0996ba7c8d : SUCCESS in 56m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78dde391bcf04683ac3b351e21d0cac7 : SUCCESS in 1h 12m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0e1608b7ec2b46d28bb031f7a1e64437 : SUCCESS in 1h 48m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c04f3571b50a4cbe9cac86da7e7fe13b : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f3428f584693df088a4b6614cddce764d3ed3496","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-03 14:47:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6c8ae72de3a44b437a782e9579f72af29882dcc2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-03 18:44:59.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0a919d3079b4c5e951e196447a00d1116088ba8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:11:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11f45f59a614efc9e320547dbcbe076 : SUCCESS in 4m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0dc06a859df9404c86de8acc5fe89306 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5cdf87fc5144e05b45c69b7b8cd011f : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98f8408fab0649888701e4db4d34354a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5f8c67ff954f39a42c83da21663074 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdd4cd5c71574dcbb7a1f925e856990f : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/052b338a03444f649756d4add0515b12 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98c7da2bf0534c2fa897c379e657437d : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/21465596ad5245ddaa12068f5dcbff93 : SUCCESS in 8m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2adf8488e70f42c489311204a8fca546 : SUCCESS in 23m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/deb2e8023ce949a6ab631afa890af0eb : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eed247ffc4df4598aed0cac8fdb10db0 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9152c2918cf6479db9e824449748b397 : SUCCESS in 7m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b0e7a47fbf09449091fc5978a0857775 : SUCCESS in 6m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a63f844f84f142c19eb17eeb212dfdc2 : SUCCESS in 12m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bb9561db1e9046d7a8221650bf50c525 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44798df4d560484ea6c64fafe6216eaf : SUCCESS in 10m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4e94d605fabc42c19d58fcf84c68db80 : SUCCESS in 32m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/14c7793101eb4b7796da0061dc0abd4c : SUCCESS in 26m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f72dae5cc3cd416691a327ba1a705d03 : SUCCESS in 30m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6adbebbc3fc24620b3be487160f1d98f : SUCCESS in 26m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e2f18a417ce94e99be81e5f7ab508217 : SUCCESS in 13m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a0407c2b305e41fc87e44bec83201b56 : SUCCESS in 13m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9381fd467bc74b65ac72b98a3370cd3f : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de50e655228741c4873f92aaa64adb51 : SUCCESS in 22m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32816915b8824b2caaedaf98b04733ce : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/71ad22e7476b4ace90362a4d2e500b0c : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef9a8332a7fb4cf9a34c444e20a7e20a : SUCCESS in 10m 15s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/cb5a3773394e49da8dcc74f6cf415c55 : SUCCESS in 1h 24m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bd8c8935c684e7b93dc699ac3021cd7 : SUCCESS in 1h 22m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cce384cce4fa48bf8bc381894f27b86e : SUCCESS in 1h 32m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dd2180d0e3094c7b8b3e19fcf7c2503d : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1fa13e4f25ff54026eb1e766ff1ffd2ab75b5748","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 19:11:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"afe3d7ee0e59b13ac70cd1fdcc0a66d2cb3f973e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:25:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf4a3f2ffc2485596f726613698d7c3 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cbdfde986e14ed2b20c315a80c39fb0 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dbbfde0a9c44082aea2f42a2965f212 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e79d41a93e34d3b97527240af57b061 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/950e12dcbe384daa83d421e39f506d91 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/203061321b074417ad28b6faafc697bc : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3c21dbaea8e4ee3a8ef37d663cba060 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b693ddc9fb174448a79ac53012cd25ca : SUCCESS in 17m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6c3b15fcb1e45ae8300c7e0bd5397e0 : SUCCESS in 8m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3de5a67ea52a48bc9627d321567f5a16 : SUCCESS in 10m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/905c6eca6f594fc0b097de647b094a55 : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c4494192ce654d8e855e6be76b025b25 : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/851b9c7d94ae4ca6a0c55230879f02b1 : SUCCESS in 5m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/06aa6d7f5bd3421a9158e0d791622307 : SUCCESS in 13m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4fb913197fa4f5ebd2ddbc852a02713 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92fed4f3848942f6829e9b478be7d8d8 : SUCCESS in 7m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/65bf0b80ee484b41aa8942b1b1e92651 : SUCCESS in 36m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1e016c750b3a4c68a6d2ab00368f4cdb : SUCCESS in 31m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c53e8fc806174d0fba9514c97c7348cc : SUCCESS in 23m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/022aed9a83644bb2b08d279956bcc59a : SUCCESS in 20m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0c3beaf159514e29beddc4223897fdf5 : SUCCESS in 16m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b226983395148e486f99c811e5d2013 : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9f19ba2f140140aab3b0df93bc75e6d7 : SUCCESS in 11m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c32106eb8d0e4c8babd07788d23313cc : SUCCESS in 19m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/742c6979ead34debb0aff8ecba7ea0e3 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c18f9ba407f44db4b8746a2288cc39fc : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d0ef44c461e145278a074bed8271bff8 : SUCCESS in 10m 00s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c00616cc3c744ddabc570980c8314bcc : SUCCESS in 1h 07m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66cea93fcdb3411eb7bbac6105b62d6d : SUCCESS in 1h 29m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0055d0c784754f52a8695ccd7d5f3cff : SUCCESS in 1h 35m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/81e0ca8882d6413e94ed5fc1c5ee8a83 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ea2f1bb97fa41c2f9bd077709fd0aa5cba9197d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:25:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"679217bc4efb81a421d66dc4f75ff1bd0b2dae77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:27:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f06262dd6b54d298c7b9f3c63a52ae3 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc7667eb1cb8f139d9fb549cdf62859d3a619d94","revisions":{"e0505c6d19d1d4b90f87c567eb226ff8573edd59":{"kind":"REWORK","_number":1,"created":"2019-09-03 06:16:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/679704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/679704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/679704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/679704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/679704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/679704/1"}}},"commit":{"parents":[{"commit":"08b58dff06e34fd1f70ae5500f50f7461ba372ee","subject":"update tornado now that jaeger-client is fixed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08b58dff06e34fd1f70ae5500f50f7461ba372ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-03 06:16:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-03 06:16:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I55f20ff8cbcfd74e4d6ddab11201cd4a759ec1e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0505c6d19d1d4b90f87c567eb226ff8573edd59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0505c6d19d1d4b90f87c567eb226ff8573edd59"}]},"branch":"refs/heads/master"},"fc7667eb1cb8f139d9fb549cdf62859d3a619d94":{"kind":"REWORK","_number":2,"created":"2019-09-03 06:28:14.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/679704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/679704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/679704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/679704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/679704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/679704/2"}}},"commit":{"parents":[{"commit":"08b58dff06e34fd1f70ae5500f50f7461ba372ee","subject":"update tornado now that jaeger-client is fixed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08b58dff06e34fd1f70ae5500f50f7461ba372ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-03 06:16:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-03 06:25:30.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nHeld back for horizon\n-semantic-version\u003d\u003d\u003d2.8.1\n+semantic-version\u003d\u003d\u003d2.6.0\n\nheld back because waiting for zaqarclient\n-jsonschema\u003d\u003d\u003d3.0.2\n+jsonschema\u003d\u003d\u003d2.6.0\n\nheld back for keystone\n-oauthlib\u003d\u003d\u003d3.1.0\n+oauthlib\u003d\u003d\u003d3.0.2\n\nheld back for known breakage\n-stestr\u003d\u003d\u003d2.5.0\n+stestr\u003d\u003d\u003d2.4.0\n\nStuff below here we don\u0027t need to worry too much about\n\nheld back as an openstack release (bot proposes it instead)\n-python-novaclient\u003d\u003d\u003d15.0.0\n+python-novaclient\u003d\u003d\u003d14.2.0\n-python-troveclient\u003d\u003d\u003d2.19.0\n+python-troveclient\u003d\u003d\u003d2.18.0\n\nheld back because release was pulled from pypi\n-ansible-runner\u003d\u003d\u003d1.3.5\n+ansible-runner\u003d\u003d\u003d1.3.4\n\nChange-Id: I55f20ff8cbcfd74e4d6ddab11201cd4a759ec1e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc7667eb1cb8f139d9fb549cdf62859d3a619d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc7667eb1cb8f139d9fb549cdf62859d3a619d94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
