)]}'
{"id":"openstack%2Frequirements~679847","triplet_id":"openstack%2Frequirements~master~Ia6f480c10b0b70833929fc524670e6892295b126","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia6f480c10b0b70833929fc524670e6892295b126","subject":"update constraint for oslo.service to new release 1.40.1","status":"MERGED","created":"2019-09-03 19:18:20.000000000","updated":"2019-09-04 02:57:43.000000000","submitted":"2019-09-04 02:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679847-1567565766823-fd577826","meta_rev_id":"381933283d4bbfbfc212d15df0e0de14d11dfc9a","_number":679847,"virtual_id_number":679847,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-09-04 02:56:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-04 01:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-03 21:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-04 01:08:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 21:39:51.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-04 01:08:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-04 02:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71167055535dfcb24626c611c93fba310f821941","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-03 19:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3017a6d0b8780b0e3678b8d7942c2d2936f52338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 21:34:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593c7967d6d84352a109591bf592460b : SUCCESS in 5m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/96232b9aae4f4fa2a23680ce4b8caadb : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2d04868d24484c8631db0835d1c400 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c443d2e1659042ff96926a094fe5322c : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12ad38cb8b0e4495a583a6c2eb5350d6 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a8b1fe8eee14b7e9e236fc33de6724c : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc3c2b1a30d340da8a4f230b737bf2bf : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d624ced3a36f4508936b23897edff653 : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6680cfa786da4be3ac7eed6c24389eba : SUCCESS in 7m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c93a4f0062524c7e87d2004976807ed6 : SUCCESS in 9m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/afa2744b913c4a5cbdfbacfdad445d1e : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e5a98583376d4efd8a151b84715ca9fa : SUCCESS in 8m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c990922ead7145ee9165fd3453672c4d : SUCCESS in 5m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32ebf515e2bb4426ae7b5a7f2778fc04 : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/527986c78f394149a96b3ee42c120890 : SUCCESS in 11m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c1ea63fbecb548e5bd49a70038075f6f : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/040c468b6c6f4c45a3906cd9b1ddbbf8 : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d366728487ee4e429ec070c55c9de5e3 : SUCCESS in 37m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a444a4bb7fa44f3b534884b51eefdc5 : SUCCESS in 36m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9462803b84ff4610b552b2c18a39d637 : SUCCESS in 26m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2721435736674ac0a8ec7556fdab2c81 : SUCCESS in 22m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/21435fd778fa4039b215e58b147d9fb4 : SUCCESS in 15m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/93325fbf1fcb4cd08a22de510b962009 : SUCCESS in 12m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3542dc778244443eb4cfdf7f1cad1760 : SUCCESS in 16m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf598386b6b34cb7a2b302aa5b19a6c2 : SUCCESS in 23m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/759df4ffe36c431293fffddf4f3a9dee : SUCCESS in 6m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6493a8187abf4648908ba909c4ab3f2a : SUCCESS in 12m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03dfcbf6796d4cae90149fc47a565352 : SUCCESS in 10m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0b3a614be0524ad397c2af7f33b6c3e7 : SUCCESS in 58m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d4aa5cc2ec34255afc3f526a8c841bd : SUCCESS in 1h 11m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/45f586cfcf744d2da9125c80d0e2bdd0 : SUCCESS in 1h 52m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30baa192a5ce461ebb11b747c08b94f4 : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dd55616823d5794c70ffc3632d72965bc52a7dcd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-03 21:39:51.000000000","message":"Patch Set 1: Code-Review+1\n\nmake it so","accounts_in_message":[],"_revision_number":1},{"id":"5ed40c993364416d4373c8710d7b069fddf3a192","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-04 01:08:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"371355192d99484c1d3cd35f9c5630c029f1e11c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 01:08:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06d92b3ffdb369afb7bf557b1dc2e80dc04ef725","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 02:56:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d689aed69ec3e274b1c6c7f48f241e787d4b8bac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 02:56:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8debf224f5fe4028811266aba1811e74 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74476f3bd3c4373b58d4670a1b949b0 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cb870c9b8db4828a9c79c1de3329b00 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ee99c97f8bf41a6b5fc5c39d08356af : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29721fc31a6a4ffcab0f03fb3281633e : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b702ddc1c29246fc973bb29f16d48054 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0347678f304b418ba6a62d3738af9670 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/83813e7f703942908d1dc68458a15e4c : SUCCESS in 9m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/40ae9be3e6944e77ae0ec77e916b960b : SUCCESS in 21m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/91e989f3f54943309c78cdd3e5ea53d8 : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae5a0cfc1ec442fd86a7d964b7331eb3 : SUCCESS in 10m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6a1641b7cc4b439aa8e92b4299ea8b60 : SUCCESS in 4m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21c6c3b1a63343ebb59309e0425cf671 : SUCCESS in 6m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4764356551b84989b18b2c9419c169ab : SUCCESS in 10m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/14a25724ba40405eb58e4d35ffb9998c : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82a3011c9c8240d3a758fff3600247f1 : SUCCESS in 7m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ca848645206a4707b3fd49ef718d520a : SUCCESS in 31m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1022c24612dd495bbfac5f18bdb2a22c : SUCCESS in 31m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a7fad5596e0b4688a9164cfad9c4800c : SUCCESS in 24m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ba1b86b50fb41528af9f2fd8245195a : SUCCESS in 23m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c3b136421afa4f18a83c65037e1d2064 : SUCCESS in 14m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c4f211675fd1485f828a4354e79b3a70 : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cbb171fb4e544a3088b6daf164bf92e3 : SUCCESS in 12m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/778b096eee1c4f40a3206693569a5d6a : SUCCESS in 21m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8677ccac703e44989ec7e8db8ee840fa : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/33423bb60ead47f58c22c78088a7b0a7 : SUCCESS in 8m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bba14221a1549ab8066c39d77a8b288 : SUCCESS in 10m 01s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2e0e6ec516854086be05d7b15257c7e5 : SUCCESS in 1h 00m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a24bc0f8e92404d9db24d6a5e321e32 : SUCCESS in 1h 08m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/747dd43de44243f4a574ff816d0efe37 : SUCCESS in 1h 41m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3632ee805e734b8d91351c38237b7407 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":1},{"id":"db2453b614a52894b4e07d313d6bf9b050cc4a51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 02:57:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dbc0da8c1e4d6e8110033dc3c9d7fe : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06888ded9a5f55e9084eafaee72785b6ece47077","revisions":{"06888ded9a5f55e9084eafaee72785b6ece47077":{"kind":"REWORK","_number":1,"created":"2019-09-03 19:18:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/679847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/679847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/679847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/679847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/679847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/679847/1"}}},"commit":{"parents":[{"commit":"926309010f7affc959e6367d4b2be2c2951e7e71","subject":"update constraint for python-troveclient to new release 2.19.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/926309010f7affc959e6367d4b2be2c2951e7e71"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-03 19:18:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-03 19:18:18.000000000","tz":0},"subject":"update constraint for oslo.service to new release 1.40.1","message":"update constraint for oslo.service to new release 1.40.1\n\nChange-Id: Ia6f480c10b0b70833929fc524670e6892295b126\nmeta:version: 1.40.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: Iba3992f4f88b2007d135e6b05875b2ed17fcd93f\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06888ded9a5f55e9084eafaee72785b6ece47077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06888ded9a5f55e9084eafaee72785b6ece47077"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
