)]}'
{"id":"openstack%2Frequirements~679990","triplet_id":"openstack%2Frequirements~master~Ib22ec108288111aeb99e5b5ce7ab31f44a61c612","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib22ec108288111aeb99e5b5ce7ab31f44a61c612","subject":"update constraint for python-zunclient to new release 3.5.0","status":"MERGED","created":"2019-09-04 13:25:47.000000000","updated":"2019-09-05 22:29:12.000000000","submitted":"2019-09-05 22:27:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679990-1567722473187-7e60562c","meta_rev_id":"116f78d6574d92c56e7f8e093a1d247fd99db238","_number":679990,"virtual_id_number":679990,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-05 22:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 22:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-05 20:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-05 20:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 22:09:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-05 20:17:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-05 22:27:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2393196da9205b3308f95aa6e930ef547499856","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 13:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc015864ddbc64c1ab5fc230ba7d381abc1c1dc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:05:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad2a4eaf3d74ef286a99ab97d1696e2 : SUCCESS in 5m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/de2950e4ba24427faeb731dcb0bb12b3 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e351040497430e891d192cbbbbd8f7 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ddd0e3e20cd4b34b9505bd9759ff3ef : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/969ae553b7364a87ba8bba7a45df013b : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8c3436c4d44474cb6e8173588544de5 : SUCCESS in 9m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a751a614e262486aa00116f8d82f8632 : SUCCESS in 7m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/482c1c39dc19432c80cb02eb9829d508 : SUCCESS in 8m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/61940fa36ded446ea4bd22d3f45b98b3 : SUCCESS in 9m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdc13c342eb244ca942c884717b3879b : SUCCESS in 8m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/af9ed88b0b3c43a18adb71de045b0b3f : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/24091a5b3eee41a8888169726bf2c742 : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1624f2afa88a43a1a6e21ca44939e26a : SUCCESS in 5m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da8d8fb22a564efea0527dc332d7ec7f : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ca8130326e8e430d9fb1c525489b4b4f : SUCCESS in 12m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/42882313fea0449482ad05f6771e0411 : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7cc4fc80f414d2ba4fdb16970a96e6f : SUCCESS in 8m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/11dc25adb0214c9c897adbcb20803647 : SUCCESS in 27m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/801caf7a3b70444fa251aacca5b2f259 : SUCCESS in 30m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/75e1b2dcce984b7a8f04105bcd2929b8 : SUCCESS in 25m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8a0f808cc4e4ef3a4e2c25225c68a42 : SUCCESS in 24m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/abcbec0413864b56a6fbb6f22a3a77d8 : SUCCESS in 14m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5cdccd0b84b94ebf86bcde3129185e04 : SUCCESS in 12m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e1c02b9690ba4d2cbe18284f13532551 : SUCCESS in 15m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f60779f0c2704ad3a1b137b66ba3760d : SUCCESS in 22m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb8d862069854c5f8461545690453f22 : SUCCESS in 9m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5f233c6deb814f4a8c5d7655a58ec309 : SUCCESS in 12m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1818d2b2aa5d4ed6a7a262df2d05b652 : SUCCESS in 11m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5449e07ae9624f519b72e2bf79be9068 : POST_FAILURE in 1h 05m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70bffaa02bb04d2780613ee6507ea0fa : SUCCESS in 1h 21m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/acdd3a34d78543788fae2248fd1a896d : POST_FAILURE in 1h 49m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/515606913204490fb5b01eb1f28b6c87 : POST_FAILURE in 1h 04m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e8b9eb5861abb29d9358266e759ee7cdfcca141a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:09:10.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd6c61791d274288997521461a2581f0044ad1ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 07:07:55.000000000","message":"Patch Set 1:\n\nrecheck\n\ngate stuck?","accounts_in_message":[],"_revision_number":1},{"id":"2fab5b366246cdabb2f8dbdb79390077ce7d2ec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:12:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d45cc0b36bb4ca5953c67fe8b365c5e : SUCCESS in 5m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/86f8762b30c447e4a89af7ee38d8c1e3 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2bafe9f43e42419a4b5dd8575b48ec : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5a05e0e067947be95844b12ab9fa83d : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82e5ba30e0014968b2645b064592963a : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed350233399a4c1fb19a8947d64a6d40 : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22a223a1fa1d438aba4b4963cd760909 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cca46586afa243a9aad3c106c7007f89 : SUCCESS in 5m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0b1cfb63ebca40808265e31ba7ef2f44 : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c0e93dd428f48b5a55fdc83a9a279c2 : SUCCESS in 9m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3c308d2cc20641b09c1fea3ac942b782 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e5b9e55ecc534b58a8b21f6963f1a6a9 : SUCCESS in 10m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c980ccf4e4c040b18f04d6059039f0da : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/930cdc6d958d469cb7d902d8bd9cb38c : SUCCESS in 8m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bf61bd546973453b9a0241875bfe24d8 : SUCCESS in 14m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/27dee36137b94569a7228c44b911adbf : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f5264cff85a4c58813e5bdbf30c10a5 : SUCCESS in 9m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9dd3eb59a2bd4bb58dc1c3b603f45356 : SUCCESS in 34m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/640befa085f04bb7968ee354b0fe96e1 : SUCCESS in 31m 49s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/183505c922bb45a495b91b6d79fdc25e : SUCCESS in 25m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/666cc4588d8c42799d2dc4a6b2aa39be : SUCCESS in 26m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/933f982d39764e93bc3f314332ef8b8e : SUCCESS in 13m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0d49645b2f364f70a5764ea00339e6d7 : SUCCESS in 15m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0007a955a7bd4cf980240fa7cab9b248 : SUCCESS in 13m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f00166e8c50342a293ebe0ec38dda6f4 : SUCCESS in 18m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/973b527337004bd381786f6472fe4f3d : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bec3cb9d7bdc4319b720008de3a5e300 : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/31d268776dbb47bab35b90c5a646b187 : SUCCESS in 10m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a5a2b2b8c7084de5b2faa4592dc2011e : SUCCESS in 1h 11m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf1a604b0d60405f8249a7af4af71f3f : SUCCESS in 1h 13m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0715f0eb37bc467489b4d729cf23a67d : SUCCESS in 1h 52m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/02aff02669c3475bb3b3f02da77a335c : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1903bf27fa7c17018665b26b1c3322b4f96c83ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 20:17:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0388a18b8cbaebaa22b17bf45d3bba12a2f1c4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:20:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36db8adbbe9a5e54658ab5c5fff09fe4e4d5301f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 22:27:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2f1949e1e6462d9fabdb8f0f308723 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d131b9d3b7904417aa8812f02ba13f30 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e6ff48cede6417ba7ad6cac4cbdd038 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490288e648b1460e83e777f4cf810501 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6812a550cddd44a1a55a0361063fcbe9 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52ef544b424448bcb183a6709eafc120 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b93fb03442594f1bb4a45fc2ba55d6b7 : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3cf967be620a40d9ab37578775a5ae14 : SUCCESS in 10m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6ee87b418ae43e2a52e57ed2bb9bc1c : SUCCESS in 9m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0631a93718db45ef83910fc75ce4df65 : SUCCESS in 11m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d189a8763ae4b5993fb8af8ca90f432 : SUCCESS in 11m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b754938b8a1445c28204123dda9bf547 : SUCCESS in 4m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d48aa6f00ac64a5f8ae747aa90308381 : SUCCESS in 6m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9e97332e8209491aa169aca13153d0f9 : SUCCESS in 11m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd5967d5325947b5acf919988bcdd1dd : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0dc8421e31da411bb14b2df0cbdc65b2 : SUCCESS in 13m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9488c9b5aac743bc8a781b3106bf76bb : SUCCESS in 39m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a8355442f9b401c83a2588385a03d82 : SUCCESS in 33m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7296da6779324520976a0a5640f1c425 : SUCCESS in 24m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7b62c66ed964df0bf189ea5baec9b79 : SUCCESS in 23m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e76137a0fc2844a6ab7f0883d5cb4860 : SUCCESS in 28m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fb9c54f1fb6640019243a22b64fc6abc : SUCCESS in 27m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4840b4cf27c24a52bed7eda8a395835b : SUCCESS in 17m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e28a48e12c8249479a48202f6978c4d0 : SUCCESS in 33m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7daaf25f9fc9454a951ac98808431a7e : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1dc1b1167cd423f965afae0a547dbba : SUCCESS in 18m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7d28f3c4c66e494ba1aa6e9ab284850d : SUCCESS in 17m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/678be3add2be44a08bc1306c8b3fa9d7 : SUCCESS in 1h 19m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1963c51a40649039d7ffa525f5cfd3f : SUCCESS in 1h 45m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7fa35ae73a4442638800a2713556ee63 : SUCCESS in 2h 04m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0861182cce034001be7b580502f93d98 : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fe9a58bf6e5d908276bb3d62abf6a7952a10599f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 22:27:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c4f00004c59c571a5f9c7cff4b68503bc9c189ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 22:29:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f0ec48733e4283945e99b7617c484e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c7aaa9187a8c8e2893f08478ebf913fe64fa71f","revisions":{"3c7aaa9187a8c8e2893f08478ebf913fe64fa71f":{"kind":"REWORK","_number":1,"created":"2019-09-04 13:25:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/679990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/679990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/679990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/679990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/679990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/679990/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 13:25:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 13:25:44.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 3.5.0","message":"update constraint for python-zunclient to new release 3.5.0\n\nChange-Id: Ib22ec108288111aeb99e5b5ce7ab31f44a61c612\nmeta:version: 3.5.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Commit: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Change-Id: Iff7ab68635c0851a01f450ca709235c7e130d0d8\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c7aaa9187a8c8e2893f08478ebf913fe64fa71f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c7aaa9187a8c8e2893f08478ebf913fe64fa71f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
