)]}'
{"id":"openstack%2Frequirements~680009","triplet_id":"openstack%2Frequirements~master~I4543b210ad1a8a52ff8ecf8b1bc011050c62f388","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4543b210ad1a8a52ff8ecf8b1bc011050c62f388","subject":"update constraint for mox3 to new release 0.28.0","status":"MERGED","created":"2019-09-04 14:28:04.000000000","updated":"2019-09-05 02:51:59.000000000","submitted":"2019-09-05 02:48:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680009-1567651709039-36715ff1","meta_rev_id":"d3c76c313a9ae0181222e5639bd151369a218671","_number":680009,"virtual_id_number":680009,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-05 02:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 22:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-04 22:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-04 22:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 22:11:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-04 22:27:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-05 02:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7657ba27f57f4f3d80c4adfcd7e3ad0671c150e0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 14:28:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2efcf0f5bc9d761c62369dbc5c61fcd223ed3f6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 19:21:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83437e2971404e64b801a52ac6df6f16 : SUCCESS in 5m 15s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4a562a40d7444d56919f6d001e38e846 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907033f5a9bc42339399d59a91de4392 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/397605f77a394a029782cfdc663e887f : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08e6375976ca48c3a5cd074068bdac81 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6fcc71e89f70451d8ca9f8d180bb88c9 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf69c9a45a3d4c31b14cf5e9625e9046 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7dc93a89dbf448eb9a303a405656fa83 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2771da02b6014968ac62d8e7818840ad : SUCCESS in 9m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff265072686e4f6686ac6194f553f1e7 : SUCCESS in 7m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2d3170fb0d8c4aa59e969044725d80ed : SUCCESS in 11m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e3f468b90fdf4205b05f2366555dcc19 : SUCCESS in 8m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7223a0efa1a84d84bbba156abf40f657 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b792e88ffaec4433a269055359158f49 : SUCCESS in 7m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a7c0066af7e54408a0bc30963ef796c6 : SUCCESS in 14m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/547c4a4306ad4158b08b50483a0904ca : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5a61e25ea88427a9645554c34f301fe : SUCCESS in 14m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4170c4127c774e92bcd5cd375971da1d : SUCCESS in 29m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/48507a61822c4e1b8759d003a576cc48 : SUCCESS in 39m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/145343406ce14077bf51a9250766afe8 : SUCCESS in 32m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/40ca00ae2a324950bce6d9d7dda7bbee : SUCCESS in 26m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/acff371b1a16442788dbdd9cbfb92585 : SUCCESS in 15m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/edd6fdef3a5e406da2c66620bf45c61b : SUCCESS in 12m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eef0504210ce4ef49153119806caba63 : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c471ff41c5a4625ab4df2eea1857ee9 : SUCCESS in 21m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e03bc3ed5484a308f3d2417e6c7fefc : SUCCESS in 7m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d56cba6193694a68a040a0b055d7c857 : SUCCESS in 9m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/57463e5de5534b678e86aa33a0dd6d2f : SUCCESS in 10m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5dfb1880afb94110b9393be4282e24df : SUCCESS in 59m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86f29776814044678543f27c4e28bfd1 : SUCCESS in 1h 05m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba2a84121a39480b8f3ab3f3247b7ee6 : SUCCESS in 1h 52m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9d731671545548f3a15501fa7ca8f84f : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3fbced3ad6af11dbb1f8189092231944b195b2f2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:11:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5a983fa81f41f9752773b04fdc2a178e501b48e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-04 22:27:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31463a603d091fc29d2c46763931d250a0b51452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 22:29:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3868cb97e55296f3e61e98c641641001b2b43720","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:48:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffa229d59c5462e932c54c3cbddbcc3 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f72a501250447d88ade54bbc1eab59 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ace946768dd496889c73c46d6086f57 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3525774a84714f2d8009b686100c96d8 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9168f9076de1400289c1344097031cfd : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b1dda92a5684978b5b1c84b3a37c230 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fcb581480bf41a1bdd5bec1974df63f : SUCCESS in 3m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0800a8db98e148a08b03ab55d5e80203 : SUCCESS in 8m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/603c2a8ee23f421e9d8841bc8723f3e7 : SUCCESS in 6m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/79812f561cfd4dd7bb45f12d27fbb981 : SUCCESS in 8m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9eb2c76c3565437a85169a20e8d5557f : SUCCESS in 9m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b6446c0f85f74ba08e0fa81db39e4d69 : SUCCESS in 5m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/308013a571f24ebf9af9102a0a7b25e5 : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a46c6924bea9477d81272a72bd585672 : SUCCESS in 11m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b7557751b2c04917a08bb7c6554a285b : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fcbf0a113494a3188f16d6b3f3fbecb : SUCCESS in 10m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9ae06cf3d94b4ddc875a924bc2705c75 : SUCCESS in 40m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02380248d88b4d0f9d888487624a02c2 : SUCCESS in 26m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7689688fd4014113ac2dcfa9fffc2ca2 : SUCCESS in 23m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a6e0866dca574276b6c4f29f6546a726 : SUCCESS in 25m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/74a431418759485f87a5e29288689ce1 : SUCCESS in 16m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1492e09d3d341b9b9c9af68843be6f4 : SUCCESS in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a84db21302164319bbb8d799a5fe653b : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa7fd3d894fc4d7a866462570948af6b : SUCCESS in 23m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/083071502225418cbc03b5dbdad6543c : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef1403c400d2480d987d3b6523c20b74 : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/104fc166db7a4b61a899e876ed5cae19 : SUCCESS in 9m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5bafac4366e84e5f8d04e0896c02d7fa : SUCCESS in 1h 00m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8676cc9088344a4b63ec84092d7cf8e : SUCCESS in 1h 21m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7f5ed54b888f4d2daa3ef178864f3357 : SUCCESS in 1h 48m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/054cb8d31ab748fe85a81924cf77c418 : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0e2474c6b0aecb42cfa35d4d3ebee599582fa0e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:48:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"85cb318976a17e60cd7592dadd7e5554c81163e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b44dab2fc81496c9f8ab72372f481f0 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fa25c75279542ff8267e620174d83fa540f094b","revisions":{"1fa25c75279542ff8267e620174d83fa540f094b":{"kind":"REWORK","_number":1,"created":"2019-09-04 14:28:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/680009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/680009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/680009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/680009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/680009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/680009/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:28:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:28:01.000000000","tz":0},"subject":"update constraint for mox3 to new release 0.28.0","message":"update constraint for mox3 to new release 0.28.0\n\nChange-Id: I4543b210ad1a8a52ff8ecf8b1bc011050c62f388\nmeta:version: 0.28.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Iae495750fe9d18b3dc8bce3e34e8ccd57851fe83\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fa25c75279542ff8267e620174d83fa540f094b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fa25c75279542ff8267e620174d83fa540f094b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
