)]}'
{"id":"openstack%2Frequirements~680016","triplet_id":"openstack%2Frequirements~master~Ie3d122e97e9b5af1b4ebce9687436b15b69014ea","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie3d122e97e9b5af1b4ebce9687436b15b69014ea","subject":"update constraint for sphinx-feature-classification to new release 0.4.1","status":"MERGED","created":"2019-09-04 14:33:21.000000000","updated":"2019-09-06 18:53:39.000000000","submitted":"2019-09-06 18:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680016-1567795896266-82917982","meta_rev_id":"882d9e8d4a695bfb2dcc56ebce7c70321f68aeff","_number":680016,"virtual_id_number":680016,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-06 18:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 22:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-05 20:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-05 20:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 22:09:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-05 20:17:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-06 18:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c455d680e617334d8c8bc26c1f97ba1d2c83806","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 14:33:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19c9190be2a7e93a91c417d2186fbd3a3d2aa8c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 20:11:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598c08a736884b50a3bc6acb66d5a5b0 : SUCCESS in 4m 47s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cda4eae0516248dfb8923454d456418b : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94faa9f2aaf0464093868c1e5680a8bc : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a70341b9827f4cc497b2d3f5bc6d5bc2 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32475929a667423c8abfd9fa225270a6 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/049cdff829324b93bea0d8ca4ee49f2c : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1aa102e3a1e463b96d974ff91335b00 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc6bba994c06424f83e48def9b0bed30 : SUCCESS in 7m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2094284423a74109ba1dee13c5197a1f : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66e9272523d34081aa0512a57325f312 : SUCCESS in 6m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3ffd3875ca7c4c9da37b6e1e5b12aac8 : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0feee8e559542729a7f9084c6e824f4 : SUCCESS in 8m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/50f6ac588cc54228b83708f80028dc3a : SUCCESS in 4m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e876272702284a9da4d383a2ebc78309 : SUCCESS in 6m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/773581e4243c438faed01d7bc34cea6b : SUCCESS in 11m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4db812f89eb4833a523b34ac84997cc : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3ca3884ea2146efb2e47930bc31d453 : SUCCESS in 12m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/12f01e222ca843b594a7f535f880db2e : SUCCESS in 38m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7f111cc3db724c9eb48164da3ddab3d6 : FAILURE in 34m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9f03960688724e47b8d44f067bdb1eef : SUCCESS in 26m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b07dd644a6ca4601854816f2710e144b : SUCCESS in 25m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5b1cf2f5fc4343d2b9d500b7de95707c : SUCCESS in 16m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a0861396aba843a68e7a5087301466fc : SUCCESS in 14m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5e8f1c3a13a74b1aba0ed04027297ff1 : SUCCESS in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/894ecd60bd254713804d0ceb8eb52f0b : SUCCESS in 25m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb5f9002e14d44abab72fb9fcaac0257 : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/872edfc27e4e49038399c221f9357b08 : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fbc2905400c44455a9949a9f14d5e5b5 : SUCCESS in 9m 08s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/413eb3d7dd9c4b7f9d5b8ff43fb7ced3 : SUCCESS in 1h 07m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c13816bffadf4c1db87c5883a0f03c88 : SUCCESS in 1h 24m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/180f386120894d0cb94332d6f1b43b3b : SUCCESS in 1h 52m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d84bdb9a0784427b43ade82302b2004 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d86f8729303be3ffccf95e64a686b06713ae9204","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:09:39.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c6db1dcc0ef0c87b649d73ca65550841ca2e829","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 07:07:57.000000000","message":"Patch Set 1:\n\nrecheck\n\ngate stuck?","accounts_in_message":[],"_revision_number":1},{"id":"93574842a4662993efc644422ba1d2db19081256","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:57:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e4b9887ee242a4afb4ad4741b1cdab : SUCCESS in 6m 15s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/59b474ab397a4e598bc73b8fe940943a : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/827679e98f6b4888bbe9700b6598dec5 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f1dc3d1ad764414a30fcb8b9a41821b : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/481e6bfb216e47118c867b245bba3d41 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf029a7f7b1f422382e3c15913780288 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db9210b766ee44718c81fb427ea253e0 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0b9a4f05cac44fa8d9f5c3e2c21dd77 : SUCCESS in 5m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/548606aaa2684e3f87feb605576e0cf7 : SUCCESS in 8m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/97b2a755bf904b10bd0b3b5948c30587 : SUCCESS in 6m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/61a6e31539f64a229de784de97ed19e7 : SUCCESS in 11m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/398cee119d1e4936bafce99c78ca93c6 : SUCCESS in 11m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/198792e55999480bb98fbc8b3a7cd7f3 : SUCCESS in 4m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/040a9c4464ac4bb6b54a48056c805492 : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fdce5999b24a4ad8a0fef683a9424343 : SUCCESS in 13m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0587dbbf7fd14ea6a6f5a0b576654ea9 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6c7accb5a3a4db0ab92501f8fb4da14 : SUCCESS in 11m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/981943dd52ae488d95bbacd0c4c7fd35 : SUCCESS in 30m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c2ad44ff0ceb487aaf40fe60eaa8ca92 : SUCCESS in 33m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a681f2959114420b924e7bf45f3c9582 : SUCCESS in 26m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2692df26e2164b908c6545b288b46fa7 : SUCCESS in 22m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9cf037d228c8408083573de0f37de99c : SUCCESS in 14m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b04a0916b76948e7871b56567be0a7d7 : SUCCESS in 16m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f1e72e4ef0d64af282306d958453b098 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c50bcfcab8ed4f3ebba99e0dd58fc14a : SUCCESS in 21m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64d3045ac7f64495b117264328d71c5b : SUCCESS in 10m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/16116600fe664fc7b274f20ae3b3a00d : SUCCESS in 12m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc978d564d5f4181ade44f5b4b4acd76 : SUCCESS in 10m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0913ef1e9ce64982bc6a1586bdf79e22 : SUCCESS in 1h 14m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df2d428c50a24a1aad9cf1c43c8b7015 : SUCCESS in 1h 13m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a54c8af06688488781a11b871c8e2195 : SUCCESS in 1h 43m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ff4ff04fb30c4630a73da4ac2849b636 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c2c52ab2ecba28be5c994faa25d3fb79f15efe16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 20:17:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5eeee54f0a827f831c0b932f57b0a79dd8241e11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d798ac1c00e2e228d81639351fd05f70b7d403f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 22:44:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46242bcfb02d4a7c8ed22dc37cc7521b : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f5d0ac61546447d903a068482f4d793 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da473db2ce46465c8c1741fe9982ae7e : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca61f45d30da48f58a0198e1da0719aa : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da824b58f93946f0b66f0672ecb66162 : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23abf3800f1c48c780c3bd57ad8f8237 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe1acc300e3c4b2c833d974652f47f2f : SUCCESS in 4m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ba06d21f35cf477db53fabffaa1c3cd8 : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70f806e23b594fab838324932918d9f6 : SUCCESS in 15m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0aaa517d374a427fb83d537a63a0daea : SUCCESS in 11m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/49a86c03fb744723a213a9b5e7b201cc : SUCCESS in 12m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b5d73b7120eb44009a73b519df4d4e8b : SUCCESS in 5m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14ae1df3f6ed4862bd806ddc36be6bae : SUCCESS in 8m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e9069f81902c4399bbf7839c4578cd1d : SUCCESS in 14m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e84be82bb9164df388c4628ec4e0d0e3 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f20b376c09f7461f9116eb0952ca6e12 : SUCCESS in 10m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/64fa00514705435a894e7a0c2f29425d : SUCCESS in 35m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e4f088ac583d4b4e99f91093c8e9eb59 : SUCCESS in 38m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0834cf5ac3604739b582547eb80d884f : SUCCESS in 24m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0a184e4a82f840abb304021a55c98edb : SUCCESS in 30m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/42a91cecb37c4cea93fcca304c3cd2d7 : SUCCESS in 15m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ffc336ee93264682a93e514224748723 : SUCCESS in 13m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b739ce030f454b859735882b3bd0dff7 : SUCCESS in 14m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/815ddec666d7491f8c4854010d0c7010 : SUCCESS in 17m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f0e6ec494ec420d9f792dd931d60a13 : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/da5ac8df57934c4491da9e046da010af : SUCCESS in 9m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f6e34909be044b36bbb4970b37869ea9 : SUCCESS in 10m 20s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1283441572204080a6300ecbd5fabe71 : SUCCESS in 54m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ef2ce2551c34c13bf40edb0fa75cbb8 : FAILURE in 19m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/30e6b900687d4d25a7c38a68ec791e33 : SUCCESS in 2h 11m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1dcaf8d918cd4a2b95c0dccad03a9a78 : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"6540c79f51f5ed89c19843f7f619e9f039389e80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 22:48:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c6514e3efffd2b57dc8ea39ed8496ddb2e96e90c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:19:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea77625dda640c08696853608ad8731 : SUCCESS in 3m 51s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c8132749b3b046ffac4c8dd8a2b3a640 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b4d4a60eee488db4019ede1ded318a : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da42fd325cd54311a07964c5ba143f82 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8c9793c210a445f85d19a7ec6b99f06 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c4205a259054daaae2455dcf5a3ed88 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfc67390ea8d4ffea9e5ac4211ad6535 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47b4f9dc19f449d3b6ffd3c0e6a5a946 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cab5bc9c2852430db7e9291a72f67b4d : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cc64621727a149f18e0cc6ae0e34b338 : SUCCESS in 6m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f9e44156c06a4a489278af4aa757aaa7 : SUCCESS in 7m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/718c30772e6a4e3a8db396c477367c57 : SUCCESS in 8m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5fc49a5563f849c6a1c6798b0f8437c2 : SUCCESS in 4m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8c668703038c4504b9f212942bab44af : SUCCESS in 4m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ac1f890eff56487a8bf5d2fe15d20261 : SUCCESS in 11m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bf42f074c3b94f14bbf9fcfa0399b001 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea35537c25664d4db5299f0185dd45a3 : SUCCESS in 6m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/132a3cae05b94d7aa963f2149abbe701 : SUCCESS in 35m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7041a0c92fa14531a61d6b00854c47b1 : SUCCESS in 27m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6e9993f45feb43a5941461e41b960101 : SUCCESS in 26m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aa5561d71c7944d7997b25016f4e2581 : SUCCESS in 21m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/936af89761c34361b7d98d514143c85f : SUCCESS in 15m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/baed731a086b4d9aa42ea0ac86a66867 : SUCCESS in 15m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b27c14e6bcfc4ea3add7b32ec9cd6579 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2fc0c712c704ae59eab57548dc773ae : SUCCESS in 20m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17a6684ec64c4189a1edc807625370c8 : SUCCESS in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7013ff5b43f24d0a87c33689335ab9be : SUCCESS in 9m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/960946355fc04d4290802910c1c4b89a : SUCCESS in 9m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/54062b6a327c4622b993f1fa615639a7 : SUCCESS in 57m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9553a371fa04dd9baa54651a5faf899 : SUCCESS in 1h 12m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e301d743d2d342a6a0db02e7bc1aaa52 : SUCCESS in 1h 41m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/815eb786a0b2427188e5b25a51c080c6 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5116af5cced52466cf0c664e18e449411b4d1ad2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:19:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"281082cec3a46aeaa1932c463a9b945e39533ff2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:51:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669ea3b957ea4360a3572dac65450b07 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73e1f9a7a7b4d0abca182b8ab29fdea : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8a96faa2bbd456e8d59a46270b81f59 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3655c2208d924218b15f292d221ee27b : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac6c6497cbe84305ad7a40a8dff2217a : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91dc0ee21325462d8a9c03ca1ce4fc64 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20826216eb6a41b191532cbe2de36c9b : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/17bbbb1e4dcf4b3a9b350216defe7dc0 : SUCCESS in 24m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a8a35c4f5dbe4958b48503c78cb0ca1a : SUCCESS in 20m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d62d73aade694c0e8f4cb3524494f155 : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91c89fe174e04612ac172777a625fee6 : SUCCESS in 9m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f411e93a037c43a486e44b48e2e369ea : SUCCESS in 5m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/34a089e6e59b4e399c4ef97ff577c314 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/744afcbad5284b6599128a17a36ef1f5 : SUCCESS in 11m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4f3ed2bc286e4556adaa1325caf7cf1c : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d7673652931427fbbaa845c60cf5efb : SUCCESS in 7m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6a8c0dbe2f4b494ca334c8649765e55f : SUCCESS in 31m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8262d96169345b7819634535141cd88 : SUCCESS in 35m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c28008d7372142daac9a2443cb1ea406 : SUCCESS in 23m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d22105587aa4facb6d238da4b953fa0 : SUCCESS in 26m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/13b949ca6cf44bcdb81b58ba2d77d55b : SUCCESS in 20m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/96ad61b0fda84b4798b89b2d853d1fe6 : SUCCESS in 18m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bed6c520c5d1426f9d7edd67ca454395 : SUCCESS in 22m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24626dbd41214d17afaec9e77136b0a8 : SUCCESS in 18m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96bcace533d54027836f64c249d139fd : SUCCESS in 4m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8032698fe5645ee9c98bd029373cf59 : SUCCESS in 9m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ae7b7b33a1564cf08b34724a0c9a5b73 : SUCCESS in 8m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6b74e5429a19489a8c1aa388d29c5c28 : SUCCESS in 1h 12m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6746b89aa6c24ddd93352d3d444b2ffe : SUCCESS in 1h 15m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/595c532430b84b0a90783beb37b93c24 : SUCCESS in 1h 48m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c4ad145946724db8b4c866116695c787 : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"62974b35bac6710b128124d7b0279c2478acbc84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:51:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33d5ab3cd98f338b36d33ea8268626921c1d9798","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:53:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f14669cbf0e4c43a1e0c3dd59b6b0ab : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acc486d448ae339743c267bc09b961a7c51a9e9f","revisions":{"acc486d448ae339743c267bc09b961a7c51a9e9f":{"kind":"REWORK","_number":1,"created":"2019-09-04 14:33:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/680016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/680016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/680016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/680016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/680016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/680016/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:33:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:33:19.000000000","tz":0},"subject":"update constraint for sphinx-feature-classification to new release 0.4.1","message":"update constraint for sphinx-feature-classification to new release 0.4.1\n\nChange-Id: Ie3d122e97e9b5af1b4ebce9687436b15b69014ea\nmeta:version: 0.4.1\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Iae495750fe9d18b3dc8bce3e34e8ccd57851fe83\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acc486d448ae339743c267bc09b961a7c51a9e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acc486d448ae339743c267bc09b961a7c51a9e9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
