)]}'
{"id":"openstack%2Frequirements~680018","triplet_id":"openstack%2Frequirements~master~Icf9bc0d418520132f696de4d32d06cf5dc25dea8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icf9bc0d418520132f696de4d32d06cf5dc25dea8","subject":"update constraint for oslo.serialization to new release 2.29.2","status":"MERGED","created":"2019-09-04 14:34:36.000000000","updated":"2019-09-05 02:24:48.000000000","submitted":"2019-09-05 02:23:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680018-1567650183208-99383969","meta_rev_id":"46ae3bbe3a9322441b7a8f10ab3e9e7a3163d4b8","_number":680018,"virtual_id_number":680018,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-05 02:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 22:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-04 22:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-04 22:27:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 22:11:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-04 22:27:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-05 02:23:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49cedc6aafb3c829e8cf693821d727cb861e2f26","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 14:34:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec6aaedb0261a80c50421634612072273f2b1b17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 20:27:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b20d6390f77452c833c71e240cf4579 : SUCCESS in 4m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d475a9d553964fc596bd0371942ce78c : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2494f99217646d697d0af2d58f2686a : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38c9947982344f8a911fefa07ad317cc : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa57bc32435149c09e9b32ae61ca3088 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6be2db4fe14d4f2c9a16e52a257a13e3 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85b3e667b93d42638dc5c6e6cb124f49 : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4da6c1c1138e4023b5b6140daaa44935 : SUCCESS in 4m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/03d23e878d5c402c97241ca904993dfb : SUCCESS in 8m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/05af436385214b67b22e6b825c5b2fbe : SUCCESS in 7m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/60e7d72b64cb4892b3913186d4002992 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/75dff23c983144118b886986c36fba8d : SUCCESS in 8m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f6d71f546daa42ffa2f832f2f846d674 : SUCCESS in 7m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/43932eecaeab4866be5679cfc2db6c6f : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/46eb817e0efb45a8ad492d529e9e4671 : SUCCESS in 10m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7823489e45f14be29a0eab33a2054700 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa419693dc1147ccb1e988ed8192a122 : SUCCESS in 9m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a947f1080fc64d7a83b8b96ae0f32a37 : SUCCESS in 30m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c9cad5b178946bd9fa676589295e0e8 : SUCCESS in 35m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/67e1e6f2481841f395eee514746c441f : SUCCESS in 25m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0e74b04e7140434fa4053b581dd31fff : SUCCESS in 22m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7e0048e9a4df452bb92f18c983323d0e : SUCCESS in 13m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f914f7373f6643bd971745cb82a94d06 : SUCCESS in 13m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f6a5721ff8374d7bbbba98afa40c1240 : SUCCESS in 15m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d87e9c2cd0f642ffa867e8b472ffac09 : SUCCESS in 21m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0101382bcad41cab1090c7dddfa9914 : SUCCESS in 7m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6fac3d51c2e742cfbd4d19a2197e86ab : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f2d96fb4dc344aafb2704d9b8cfc0bbc : SUCCESS in 9m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a132076b70104a039faeb4c919d510d5 : SUCCESS in 55m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/268c646450594da9bb4bdd1bf5f92077 : SUCCESS in 1h 13m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fc8916da5e10418cbabe8b40cc8cfa55 : SUCCESS in 1h 42m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cfd90c96336c4b7d8245d15b355550ac : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f1f15f8d0c18a9c57e768c6d5505c6e274380a9a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:11:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e02d16c9121114d88ad72c0d87151467decef5ed","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-04 22:27:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e11c1fb4d5b3fa02cba7d366e51480edd71cf2b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 22:29:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47a79c866b1f68dc24fb2b4a7dc462cd79aae21a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:23:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0643fdaf9206492c9b4056b528654d9e : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ba63ae5ad6d4447bc7869847feb2255 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c4272ddcf5544ef8bd95777ebf09ec7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed03b479d53844d6917df73871966b22 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fbfa751f89049f5818a5f711cc6227c : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/509ef093c8a4447faf75efec0d33ddd6 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5456091478224cddb1331adf345284ff : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/808851edbc0c47f5b4c4f94ee3da1495 : SUCCESS in 8m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/580aff9b26a843d89642e03d977b6402 : SUCCESS in 7m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/55d1093e71b24cd0bc948f52f0417088 : SUCCESS in 10m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/328a6f49094e4f599bf78a5d138dc39a : SUCCESS in 9m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b25e2b2dcc9c457f9e7fc047e4217686 : SUCCESS in 6m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8b34ee569f534c358b855b22a4d32256 : SUCCESS in 4m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/68e141541edb4edf8146070a5f7bc7a9 : SUCCESS in 13m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/313073c442794ec0aa2ed981e1e73821 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9417e3d078f24ae4b2b593d827cddb36 : SUCCESS in 8m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/656cbb7f67204417a21cf2f003bc7049 : SUCCESS in 32m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a8b74a1680741f8a1c8893943ecfc7b : SUCCESS in 30m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/14a526f61d7f4fc29249fe97b5a9ac66 : SUCCESS in 26m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3746dc63b14f4c3dbab0fc915fde8b08 : SUCCESS in 24m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e7dc662e41094c2fb14f21a747fd78dc : SUCCESS in 15m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ad8cbb7b39c74952854063e950b1a6c0 : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d2a6bec598f84ea8932c8619ea66319a : SUCCESS in 11m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa1287426b8a424895fb8a1982b85b17 : SUCCESS in 20m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f470a62b7134c69832083c0fd1a2d18 : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce9cfaac190e429ca8a50f0bd61b786f : SUCCESS in 11m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/85b09822f0a841c7994032cfdc7ca784 : SUCCESS in 10m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/624d6896912548d6a8f84aa12187dba7 : SUCCESS in 56m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8341fb44f3c430aa121d2352844b562 : SUCCESS in 1h 02m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ffee1d8facc44091b8b50db4984a5126 : SUCCESS in 1h 38m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/484b27f400d2491eaaae7130477e7526 : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fb86b4ddba79164f018df0c2ec4d3dadd134828c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:23:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c88554a12e7b5d69355dc65751da4ff7144185f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:24:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8019252eed4401fa390f8bd7758f1f2 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbbc2cda36f258fe44c7ec67ed327aa8989879e3","revisions":{"fbbc2cda36f258fe44c7ec67ed327aa8989879e3":{"kind":"REWORK","_number":1,"created":"2019-09-04 14:34:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/680018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/680018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/680018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/680018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/680018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/680018/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:34:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:34:31.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 2.29.2","message":"update constraint for oslo.serialization to new release 2.29.2\n\nChange-Id: Icf9bc0d418520132f696de4d32d06cf5dc25dea8\nmeta:version: 2.29.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Iae495750fe9d18b3dc8bce3e34e8ccd57851fe83\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbbc2cda36f258fe44c7ec67ed327aa8989879e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbbc2cda36f258fe44c7ec67ed327aa8989879e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
