)]}'
{"id":"openstack%2Frequirements~680019","triplet_id":"openstack%2Frequirements~master~Ida2615cf33a326b24eb410596ea97dc59f28bf64","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ida2615cf33a326b24eb410596ea97dc59f28bf64","subject":"update constraint for oslo.rootwrap to new release 5.16.1","status":"MERGED","created":"2019-09-04 14:34:52.000000000","updated":"2019-09-06 21:04:46.000000000","submitted":"2019-09-06 21:02:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680019-1567803774262-82917982","meta_rev_id":"4cb16c6c6575f2c35d300cdce96de19f0e3995bb","_number":680019,"virtual_id_number":680019,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-09-06 21:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 22:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-09-05 23:36:48.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-06 17:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-06 17:59:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 22:11:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-05 23:36:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-06 17:59:41.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-09-06 21:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba2fc346ef40131f7227a141a6798872983a5126","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 14:34:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c318ba4d9faded057e09542b08ad53633e0f5260","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 20:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e045c221312480baf7ed2d2fe15f445 : SUCCESS in 6m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bc88eb1b06104dbab5bdafd95f374c80 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db72450218ad4a758d72c7f3fdee5610 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e28934857d2841b09462f40e4a1867b6 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/895af0c82d344b61919f2425800a40da : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cacf9ce33a8241d1879c567ebc95718e : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0870a77f9584778ab6d1c56ca3a07bc : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d3efef0665841949b4faf04ce1d96b1 : SUCCESS in 4m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ed50b35836724d479cce25a6537a36e3 : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/69ab93ed11dd40ceaa7ae2821c13a2ba : SUCCESS in 7m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8ac3c2da33764f11b4bd6aa04795bb21 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/076451a935024d41beebbcf91f934192 : SUCCESS in 19m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6bca9f1b68324630a97863e2c5558384 : SUCCESS in 6m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/734b40b54f9445a696f76c4ab6c66f29 : SUCCESS in 6m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8ea8b184dfaa4e2cafb8c669ca7c017a : SUCCESS in 12m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4372d45b54734903be55a80227df365a : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22f763659fbd466ab603c751a2b12420 : SUCCESS in 9m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6b394f97149a47ed9d1852fbec418f57 : SUCCESS in 33m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c85459ec83f14d6f87113f2b8bb2fa88 : SUCCESS in 34m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7d3c2f3b480046a5b6649371e25102c8 : SUCCESS in 28m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6322a4006144619bdaddb8183753327 : SUCCESS in 28m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9146a2c9f58d476a82a36b10f5b92603 : SUCCESS in 22m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fed06ed5fcd34667bfd6d93355eee5a5 : SUCCESS in 14m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/644799d756b143e89e2410bf4815ff41 : SUCCESS in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da42bf07c2d946b49b12375c68acbc4b : SUCCESS in 19m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c163ef10a624f3c9b67937a1c597b4a : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8b831006a21e48068935ab79e037d97b : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c927ec72e042415bb25a71c87a8ecfa9 : SUCCESS in 10m 56s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/725c39a691de4007b9b7f3be0c8e3325 : SUCCESS in 1h 50m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfc2d1528a3544d79f139d5fc1d04d41 : SUCCESS in 1h 34m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/563febcd030648f982493ee07e251277 : SUCCESS in 1h 55m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5f392d468f2e4c309c545e7224834734 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":1},{"id":"75e0720fbcfb4b7f8fa65468e880a38c388a24f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:11:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"024af2a8548292e59455ffc5d9b47ae1f361303a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-04 22:27:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a538b33b292f7ef77f3ce284c0e4003aff8cac8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 22:29:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36df23ae43b4fd140234c8e8fccf3022683887e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:50:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c806c3afcb43cc88ea39c579cb7960 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ecc663a80514f1eb23643f60bc649e7 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08f7c38a30c949ca867df04c8fe59630 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/631d4d52d5b041ca89309c2359b1f70f : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3be879b251544ce4b4cb0a53434099f1 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da0217b039484750b82a43aeda008b42 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c92b83b9140a42d2b551928826778cb2 : SUCCESS in 4m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1933557c44c3423589214e2bc0507c89 : SUCCESS in 8m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b21df10ed8db4112bdbd4d17c9542f19 : SUCCESS in 7m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5db7d964fa284fcd81268650539b4d13 : SUCCESS in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aac2bdb1be6a41fea1002cc1f2554b45 : SUCCESS in 8m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/df3700eeac2b4828be084982cea43691 : SUCCESS in 5m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1ba812c49464b15be8022ccffe58e67 : SUCCESS in 5m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f3cf6a74c66948b98c20a68303684150 : SUCCESS in 11m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/74ba12e90f9649d3a2f440624993c7ee : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b15574fc8be0486e9e57e2417825ae71 : SUCCESS in 7m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/540a803fd41f448f842d1fa622361f4b : FAILURE in 34m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/204cc871859c45c883c80ffeb4285304 : SUCCESS in 30m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8e2827f5ddfd47bba78035642af15986 : SUCCESS in 28m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dba90f6a79034418896e4dd092a124b6 : SUCCESS in 23m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c0b2de3c6b214dfd9416cf427772c3d0 : SUCCESS in 15m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e3770fe7867a450badea2be717cd8d8e : SUCCESS in 12m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3c92a5c875444a4d9245418211a5da38 : SUCCESS in 14m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f626060375942e0b82dddc9339890ab : SUCCESS in 25m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11fbba99e5c44f53990785d5f59ff711 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e880dbcabaf442c68ee47ded21771ac8 : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/379889a1881d43c18adb9c7274cd908f : SUCCESS in 9m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e471acb8af6847268572c57c4711dc93 : SUCCESS in 1h 06m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdfda6c15240407f9715fffe0b5cecb3 : SUCCESS in 1h 10m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/47ee5e2a8ef6443b92464fc57dd3f732 : SUCCESS in 1h 41m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9a0049887cf74d8897fe6e188bea6952 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"62f3f424378f7c1c18750d8df927fa95ed0a4a6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 01:55:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"453698801e926ab72f19cac7febd7f7f4bc1a8cc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 07:08:07.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck\n\ngate stuck?","accounts_in_message":[],"_revision_number":1},{"id":"55151724880c5c2fa6c33d4385073760aaf0a4a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:19:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b849dceba43a45e1a4609bd4d1ee8339 : SUCCESS in 4m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/53bc9f2ccf424f7883790b5455ea0046 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec7202df2dd4b688e427d97cf291ac0 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab079bb5f22447fcb2f359587eb24d64 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec8e6d3374234eee87ba8c22c7540a80 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75c786edecc549dda12d034922b45997 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac3d944bbb2046f4a747a389aa1c96fe : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fac22739ca52467bbefeb2dde48dbf6d : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b70c6911d51040dc8244dedd860037d5 : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/396b3bba042645d28a1f9651b9a96798 : SUCCESS in 6m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/98b9dfeeb871492bbec2f58417ccbf54 : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70938317bbce4003816b999fd1d64e8d : SUCCESS in 10m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/33fca33bc4894518904aa125ba561866 : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/97c9eafd17ba49fab0ac6e1172e5f836 : SUCCESS in 5m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/370e1820741d4f93b6855ead13241875 : SUCCESS in 11m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5bb59973f29c48bda3eb2ff1a69331c2 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8bccdf60886474baa1fbde9d317be6d : SUCCESS in 10m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f630973dc3543249c9cd7aa9fd6b9d7 : SUCCESS in 32m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/07b55502bdc04ef98be905bdca6be405 : SUCCESS in 31m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ab753ab5f5764f3cb37012a09d6b6a5d : SUCCESS in 28m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6387f788eb91411198e2f965e8cd8fb3 : SUCCESS in 23m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/65a2eb8cc0c34a168a4c739d8663867d : SUCCESS in 13m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6afcc6792dbf4562b468b672e3f26b23 : SUCCESS in 14m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/491127b1c69340209171ab3805ab91cd : SUCCESS in 18m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98cf8c12e196499d8d3ab82637773bfa : SUCCESS in 18m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5cbdef752069489596f76df631d14cd2 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/531f5f0a7bd14001b380e623127b1baa : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6c341727bbc243dba669b85d217a2de4 : SUCCESS in 9m 54s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c60090e93cac43e3899a72f88ab96a0b : SUCCESS in 1h 06m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d54d27f4ef294becb1ddb4e915f739a2 : FAILURE in 1h 43m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/57bcc2d0b38e48459c195de7974da80f : SUCCESS in 1h 40m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9d150c6cd2c54b1b9bf919589a975fea : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"aa38c790f3c6c1eea86aad7e3a3dcb5c17eee817","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 23:36:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd9db1d5d4fe5a185e3d46b4fad0fdff9bb0093b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:39:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f30b8787154c8586c780a320924909 : SUCCESS in 4m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2cdc21206ce44919a263aa7cb7cbcfa3 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919534cd265d4b29b08ac1bf0547b24a : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4a393335ebf48f89202df41c6a88016 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3749e85319554b729586decec1861b9a : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ec1ec0b2f14443084b0446b98e259ff : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0265a88a666d490bb7ae0491bbd4cd48 : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/049c59099fba4d1f86a84dbbb0ed7f3c : SUCCESS in 2m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f775ffe6c50b4f569a03e6357354d4ac : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/784cf7bf8425448e8a67cc4f4d00915f : SUCCESS in 6m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/048b2749ba8b4a798c57f0732ad3e731 : SUCCESS in 7m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db51e47afcd54794a8a31c9a1212231a : SUCCESS in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/53cff071ad5745e496b31655cfa31414 : SUCCESS in 5m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/adc63cf1be074be289fd0eb1efb2ea6a : SUCCESS in 4m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/429bc70d10f64b829ff10b242e7b696e : SUCCESS in 11m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ecf5c56e4f04618a38dc0475c367e6d : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5411814e960441babf48f5c835d5a7f7 : SUCCESS in 8m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/72f958c30d3a4548b6c63dbd890f35fd : SUCCESS in 38m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b69d60e59bb140088aabc9561386daba : SUCCESS in 28m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f32578b169d34f1d872d8824c724ca7f : SUCCESS in 26m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8fcbd828f0d34b30b8c06890f70f277d : SUCCESS in 28m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/73acd0f207154ac89ab726b5608b6b06 : SUCCESS in 19m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36ef1411659c408f90d42602fc6b272f : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8a6bc225364e475ab80b3d037e0f9116 : SUCCESS in 14m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3cf8ea0af27347e4a205fa29835dd925 : SUCCESS in 18m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf4fd02e44814af1b1572559ab86020f : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b7f821aa934e44268b1cbf71e539ded7 : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/924b9583a39a449f9cc75437fd377d6a : SUCCESS in 9m 49s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/031213ff74314f40a4e5cbfd52e10f68 : SUCCESS in 1h 00m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f42fcfadf324c02b32e53263486e00c : SUCCESS in 1h 22m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/14ed0ffc20ea435995225329aac9974e : SUCCESS in 1h 49m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b76ce08e2fcd46ecb0dee70b9756d645 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0f42e539f66d0d8202456a70d837f739053ec217","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-06 17:59:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b325898d167bb5d4007c0ecc5b8f59b93bd75245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:00:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aac042d3d629dc7d1600e7f728b90da8cb4e3734","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 21:02:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad9ebb98d8cd4d388918d0b9e5791a31 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca97751292f4b61bc7ef0015116ddd8 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edec0d70e95848ab91b0bdf0075e0634 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef43dafaf4044c1888c65c5c9f5242d : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1884526066d946c58f34cc0fa8bc96d8 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf1ee008824b4b45a65e624cd6ab7fb3 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f32a7d9de224afdb7c5b7d7fec736be : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fed4727faaf14ad4ac0d49764a1193f9 : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a352c03df80849df9ea8c4b1e5644608 : SUCCESS in 5m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7fa78d22b88646c8a1286a6e1efca501 : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e207fe9f22694e8dbc90d7cf3ad0ef4d : SUCCESS in 9m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cc5cc23574a14f009f668835ddaaf8e5 : SUCCESS in 3m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a36a770c6ae4402282cf196559e6e41d : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/625648aefdef47b9862762392b320fcb : SUCCESS in 14m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f53731dbb1214417bd87a7dbe19a8a10 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0066afb83b364c8c8e025e09dc1cf420 : SUCCESS in 7m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6bbf7df288de46098fb3e76ae135d01a : SUCCESS in 33m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/654064de85034945bdf304717ef3b799 : SUCCESS in 33m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/15f70e589fea44a9b17ff8b45b65d172 : SUCCESS in 26m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/30c45c11df374d4295072d866ee988ee : SUCCESS in 30m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e9b4d63c5f2d4ac7b86c55c939db30d9 : SUCCESS in 13m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8d6908dfc7774f5990659c38bca75976 : SUCCESS in 12m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/436fd758c6a54eeaa996cb24321cfc53 : SUCCESS in 12m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45a743b279f64a7d8f925cd5e5cbf38b : SUCCESS in 22m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1afb869c161747a28aea1efe7ff4ae20 : SUCCESS in 5m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af3bdf15a7094e859b8fc50975cd6aee : SUCCESS in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e5798eab1f34645a08a5e51cc485f9c : SUCCESS in 9m 21s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/492c740f696143dabd82ba2e73ae0470 : SUCCESS in 1h 21m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba95be0a3ac84c428f8fa8b8271d5a20 : SUCCESS in 1h 17m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2dd125b9f2d941f38ae8886ffd91dbd8 : SUCCESS in 1h 37m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4162231123b447979d95adf278321252 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9f20eaa06e88b929df907578b25dbc2a52a1a64d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 21:02:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9eb1f2d53d6fcf093320efa9023b3c8e37d0e616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 21:04:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5e226c56ff4a24bf8f4a8d40aa6482 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8111951024621bddd9921b8d4a47c93d860385ab","revisions":{"8111951024621bddd9921b8d4a47c93d860385ab":{"kind":"REWORK","_number":1,"created":"2019-09-04 14:34:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/680019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/680019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/680019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/680019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/680019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/680019/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:34:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:34:46.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 5.16.1","message":"update constraint for oslo.rootwrap to new release 5.16.1\n\nChange-Id: Ida2615cf33a326b24eb410596ea97dc59f28bf64\nmeta:version: 5.16.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Iae495750fe9d18b3dc8bce3e34e8ccd57851fe83\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8111951024621bddd9921b8d4a47c93d860385ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8111951024621bddd9921b8d4a47c93d860385ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
