)]}'
{"id":"openstack%2Frequirements~680023","triplet_id":"openstack%2Frequirements~master~I087c0822285b61277309380cf94ba1d6d06c5607","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I087c0822285b61277309380cf94ba1d6d06c5607","subject":"update constraint for oslo.versionedobjects to new release 1.36.1","status":"MERGED","created":"2019-09-04 14:36:13.000000000","updated":"2019-09-05 01:05:00.000000000","submitted":"2019-09-05 01:01:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680023-1567645305712-05248e82","meta_rev_id":"71eafca849e2dcccc70e9657bfc8a86a7e34227f","_number":680023,"virtual_id_number":680023,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-05 01:01:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-04 22:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-04 22:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-04 22:27:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 22:10:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-04 22:27:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-05 01:01:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5e9f684760b589efc5aa4ea0ef4777eff4e7b00","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 14:36:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb17fb0657a2402cece997d55234d382059e076f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb53e44813342d4bc3587b49196ed73 : SUCCESS in 5m 25s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a9a64d4be07d42f49cbb11bf745280f5 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b83b1a312674c29a5f9c19c639335bd : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f5e65fb356148bc8bfaae7ccce3d99b : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21b0aed2ae0044149f2f3bf8da532d56 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/692598ba205a41aaadf961b52a9dde3e : SUCCESS in 7m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5aa70bcaab374ccf96389d2230d3866e : SUCCESS in 7m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9a5371c08074c5da8169a406cd5f348 : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/31b57d17f63b4209a9ae2831d89ee262 : SUCCESS in 7m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/186fefdc17d94f5b9360b5e71d6b16d0 : SUCCESS in 7m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ed7677f8da6d4764ade4fbaa6f72f865 : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb11f24a747b49d290f8541da2cf5de3 : SUCCESS in 10m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/691d2b10ce004ba39e6d7b177b975bb9 : SUCCESS in 6m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32c69839ac704cf1b35e0b3822dc603f : SUCCESS in 6m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/52728d54130d4cfc818691ee7f0b3cb0 : SUCCESS in 11m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cc07c4be84604d9e940f7102f04f634f : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5643d4443e274788b32996f9330755ee : SUCCESS in 8m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9a6cf2daa83b43c4ae19ee26c1efaca9 : SUCCESS in 41m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a0e948d9feb14531a39669860eccbb27 : SUCCESS in 38m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0c2ed3b40c244c648b12c30fe8520556 : SUCCESS in 30m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c54c9a0ecab54feba75c4f5646627231 : SUCCESS in 27m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/53d68a76ca0e4c12a4ab4510190630db : SUCCESS in 14m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac6d48265ac046adbffebfb0a9b073d1 : SUCCESS in 15m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d8a7afd9704b449285e7e495c30dc508 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fdc2f115f1fc491086d8868dd491e237 : SUCCESS in 24m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df27e2c51f1c424f990817009a56575b : SUCCESS in 6m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8a5c8dac4654300982e5544b815fcd7 : SUCCESS in 10m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee34a8613fa940f8bc87a3e0896ae784 : SUCCESS in 9m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3214312d8a7f4bbcbff44f56a3a7f4fa : SUCCESS in 1h 08m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f8d82b6c50646a7b20b5d54ff23fe1c : SUCCESS in 1h 08m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/14f96075dbd6410aa7e31eff7a08daa8 : SUCCESS in 1h 59m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6c5d9789881e47bba130a285b2c357e9 : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"367109c76a3473e43a1693bf48b60d5bc6fa5cd6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-04 22:10:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a64fd8ea8cfaeb75756bf9dcdc673fdc3a1bbc53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-04 22:27:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"372f474abaf2b23ebe5966fa79e19a8d57cab678","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 22:28:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39343885d54b9f8fa1468b65578b16570150d586","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:01:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14d2d569ad8244268a46b58c5216a3be6659f387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:01:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c46b98541c646ba9a472ffd2d3ac251 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da270190e6ec479e8742af25db2192de : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d96ccf0b9d342659a15c0dc0d716016 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cb58f636c1b4642811a6c3c7cdc7b6d : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d773ab04bbdd4d69b740f9c391d9a3b8 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f5ab0371a6d4584ba6c132cc5adcd2f : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4608d20f18641149b2ef521b55bc0bd : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/851aad6a69c34cc097cbc8cf800c458e : SUCCESS in 7m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5a8dc5ba72e43ce91baeb4d5919e129 : SUCCESS in 6m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/88c5ae4be34947a8ba180f4d23f674c1 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/890208e66b064295ba531e74f089a0ec : SUCCESS in 12m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bb18f607b4814cb0a37c564bd8a2c27b : SUCCESS in 6m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d26145ddb7eb406c9e014e0e5bba6646 : SUCCESS in 4m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5af6dec9d1d04344bcbd5cebcbb59479 : SUCCESS in 10m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/943ac23d62294224a99ca174b994d628 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a46ca1a586a04d38bae73054c9cd7023 : SUCCESS in 7m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c03044f956314f5ab729f28609c544c7 : SUCCESS in 29m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0d576ca92f94057adfd282359c9d9c5 : SUCCESS in 29m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/76236c4bf4614c118b65f7c04651d27e : SUCCESS in 27m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d65dd3d2a7b46a7a83ae215202213e0 : SUCCESS in 22m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dabc09eacdcb446eab3264e896db247c : SUCCESS in 14m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f724b8a8c6294471bbe7d06ec5e986ed : SUCCESS in 13m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8f1bdcc22306434595e694a1c35c2868 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ef490da3fd9406b95555e7e1e6bbb0c : SUCCESS in 19m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fc699f0032443aa95e21124268bb871 : SUCCESS in 7m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2a934707fa444dcaad031b065583b19a : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d09835c082a24e93bce4a01417dbbc8e : SUCCESS in 11m 19s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/eef98f274e904ef7904134b91e813f2a : SUCCESS in 1h 00m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0dc8bd6305944e19077662f3a97cea9 : SUCCESS in 1h 10m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8efd89fe8ac41fea660181251692036 : SUCCESS in 1h 35m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2cdb2602722a4a6c98f275df56e31759 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0e03007f2bb8a79d656b1af770a4b17fa631ae58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 01:05:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2e7e8055764b5280523765c161a4c5 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f171c684791889cfd5bd8e3c0108f77da8d2bcc","revisions":{"6f171c684791889cfd5bd8e3c0108f77da8d2bcc":{"kind":"REWORK","_number":1,"created":"2019-09-04 14:36:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/680023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/680023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/680023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/680023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/680023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/680023/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:36:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 14:36:11.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 1.36.1","message":"update constraint for oslo.versionedobjects to new release 1.36.1\n\nChange-Id: I087c0822285b61277309380cf94ba1d6d06c5607\nmeta:version: 1.36.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Iae495750fe9d18b3dc8bce3e34e8ccd57851fe83\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f171c684791889cfd5bd8e3c0108f77da8d2bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f171c684791889cfd5bd8e3c0108f77da8d2bcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
