)]}'
{"id":"openstack%2Frequirements~680221","triplet_id":"openstack%2Frequirements~master~I444f8d141b0af9443b197ec65a900f049e53d5a5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I444f8d141b0af9443b197ec65a900f049e53d5a5","subject":"update constraint for metalsmith to new release 0.15.0","status":"MERGED","created":"2019-09-04 21:15:33.000000000","updated":"2019-09-05 17:54:40.000000000","submitted":"2019-09-05 17:49:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680221-1567705787009-f70f030a","meta_rev_id":"a5448e34db4a1e6ec1f3a6c2b97a770abb08f244","_number":680221,"virtual_id_number":680221,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-05 17:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-05 13:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-05 13:56:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 13:56:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-05 17:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ecbd98bf24124e4d33866f2e91c047ecce7c29e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 21:15:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68743c2b7f0acfbcf1e9df5466489eeb6c8b61ee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 07:08:00.000000000","message":"Patch Set 1:\n\nrecheck\n\ngate stuck?","accounts_in_message":[],"_revision_number":1},{"id":"a3d4a40d3506e0c1b550f1f6ec298ce535bb8aaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 13:19:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985334f61c5f4dabb8c057d913bbdb5f : SUCCESS in 5m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1f6f9504359249598c5e476e70962af8 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6742708ad944a3f991dc385f30a1d75 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b977411e7f6e4893b889cf2cd142a9a4 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a4cf646d364d36ba5ff4d6e7fd6588 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0334a594ddc446ec89fac6fca0e3bad7 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee23e702803a467eaed7f881a44e411f : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd8133642b874f3e9cf500ad04d9f4cf : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/556d89b9e7fd4f5e89fdf380d0c231bc : SUCCESS in 14m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/044c0b907f204d3c9c1e72ced374d0ea : SUCCESS in 7m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2a83cd5343e24f8e96dba31d5ada164e : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85d7a3c09438480a9c5494f67408e2f0 : SUCCESS in 10m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/68d9e48ff8e548f1b32dbda1a7c040a2 : SUCCESS in 6m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d04559940e5242159feca4a3946382b7 : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c1e9d15de11d45cfbd63c0784c304827 : SUCCESS in 10m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/436ba87498ce4631a88f823237964c3a : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8be7df6d52bf486c8e1c12b2700e1c04 : SUCCESS in 9m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a09a77b07e114c438e3d1b6851994837 : SUCCESS in 39m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5f978983b608496bafa46d20274d9e76 : SUCCESS in 34m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/760ffc4ed07a469f86162747997828b4 : SUCCESS in 27m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f179d29d668542408713117cb546952b : SUCCESS in 25m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1f222b9e049c4a5f93e00707bf7fe2d2 : SUCCESS in 13m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c2d0263a54a45698e3d2ee0fe2056b9 : SUCCESS in 13m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ff27c5eb5ffd475ebb9000713acf8b06 : SUCCESS in 12m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba6a04e7b16b4dfb833e063a7f959d8f : SUCCESS in 21m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ce9546cd2ab49e6900329f930bb21f0 : SUCCESS in 6m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8608fc2017a64537ba3a8297d40ffd79 : SUCCESS in 10m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8bfd6c9d1b774aa5ae9d0ed5c7f4b565 : SUCCESS in 9m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c843f9c232e34a49976a29ad9153a231 : SUCCESS in 58m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dbea60fed00b42fd946ad2fa2b9bd1b4 : SUCCESS in 1h 18m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c5c0405aa5064a07ba26b4716fb4300a : SUCCESS in 1h 43m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b84c0484e70141bca41b0ce0bc77a6cb : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"22a5d90a32a87cd80d41cfa53776d319905f8b32","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 13:56:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b40f568a632fee439dedf9bf79c315968672b1e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 13:56:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9efdacc97dd9902c8b6fc50344d744e7c1ac7a78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:49:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d29d34e63274ad58adc6feb5593eeb3 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48f54bd7daa49b9b2345b77a932a591 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4dad09a397de4973839f24c83d6e8f8e : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa08f2dede2e4e3e80f7f3e3ece37017 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c133fd38727a44c390fa21cc6d32fc0a : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/301e7bfb6efd425290e372b2813cb669 : SUCCESS in 6m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/503d0269ff8e4223a74a9afb4380f290 : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6210cd775c754919a85e6d22f7615334 : SUCCESS in 14m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc0ecdf49e1849b7806ba8f86f170187 : SUCCESS in 14m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/73c45da0b4f243a190e07bf7efe58321 : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a8db01470fd4487b5f8228b2c25530f : SUCCESS in 8m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e0bb737fb24e4e3189c91d9a359faefd : SUCCESS in 7m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1faae256af684abd9c440d9138e3410b : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e6533b50fcab4b539887c6048f6690fe : SUCCESS in 11m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/44fa9d08bf37469290380bf892029b8f : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a6a7e016ee74b3985c2cd16cd05c122 : SUCCESS in 9m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a2473be069ea496386167a7602fc19d5 : SUCCESS in 32m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/172ecc42a0e54788be8dd14e0e7b0ef0 : SUCCESS in 31m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/28afde8d4d93446a9df4a478dbd5298d : SUCCESS in 25m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6a2630a5d2724bdc9c8da4214d2eae1d : SUCCESS in 21m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4caad893e0db4b8492036c7b27eac967 : SUCCESS in 16m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/90ef71a3b82d4f31a7ed2135373bb15a : SUCCESS in 12m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/96968da46b42478aa5464e1534ed3457 : SUCCESS in 19m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/066713d325454b7796250e6a81f493a4 : SUCCESS in 18m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/316da349816f4932bae4ae746a0d8ff1 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/16e05c9b14874894b8f4ef9006d6bd3b : SUCCESS in 10m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/59c36c4b40c44e1e9faeb96c4dc150b0 : SUCCESS in 11m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8060be504af6499585705bd4a9df0161 : SUCCESS in 1h 07m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/455060af51a04318b36da281d66009cf : SUCCESS in 1h 18m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb76cfb73c314332978a0383173b5810 : SUCCESS in 1h 39m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/64996750328b49559e0dd64b2de3bb5a : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"148f9402a4c8aaa0019d6b3b18be9cb722c0694f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:49:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cea4b355c839a230bcbb6aa94c627541c3f82bd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 17:54:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e97efd35c2400e9eeb23aaaf540c6b : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f42c32264711a7050ea200549ade34e5de1a696","revisions":{"6f42c32264711a7050ea200549ade34e5de1a696":{"kind":"REWORK","_number":1,"created":"2019-09-04 21:15:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/680221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/680221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/680221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/680221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/680221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/680221/1"}}},"commit":{"parents":[{"commit":"d5dc5ef4a3398295289a9e988bcb4fbb0904dabb","subject":"Merge \"Bump upper-constraint for python-novaclient to 15.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5dc5ef4a3398295289a9e988bcb4fbb0904dabb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 21:15:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 21:15:30.000000000","tz":0},"subject":"update constraint for metalsmith to new release 0.15.0","message":"update constraint for metalsmith to new release 0.15.0\n\nChange-Id: I444f8d141b0af9443b197ec65a900f049e53d5a5\nmeta:version: 0.15.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Ia5f9606fdfab8915957788ffb1723de8b53634f5\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f42c32264711a7050ea200549ade34e5de1a696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f42c32264711a7050ea200549ade34e5de1a696"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
