)]}'
{"id":"openstack%2Frequirements~680304","triplet_id":"openstack%2Frequirements~master~If1507f44eb88013cf08aea4b46891876fa754cf4","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"If1507f44eb88013cf08aea4b46891876fa754cf4","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-09-05 07:13:50.000000000","updated":"2019-09-07 12:24:23.000000000","submitted":"2019-09-07 12:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680304-1567858991804-82917982","meta_rev_id":"1e4ad27a808d5a48739e92ada0eebd0a3b394e05","_number":680304,"virtual_id_number":680304,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":2,"date":"2019-09-07 12:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 06:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 06:22:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 14:14:36.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-09-06 13:51:43.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-09-06 23:12:02.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2019-09-07 06:22:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 12:23:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9706d0a0af6bc89b3df9277ca17f0dfa3708b2b9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-05 07:13:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"048d9bcdda029c0b5cf61d4b9b2706674dd3f7bf","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-09-05 14:14:36.000000000","message":"Patch Set 1:\n\nThis seems to have a large amount of overlap with https://review.opendev.org/#/c/679916/","accounts_in_message":[],"_revision_number":1},{"id":"3cdef7c3ef42a3fadbb1a2933d4bff94cbd74778","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 16:14:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"126225efbf6bd187fb787595bc7f9755441d093e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 05:37:18.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/boclob","accounts_in_message":[],"_revision_number":2},{"id":"17a40af5f0d08d3cdcc1434b891b63719106362b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 05:37:25.000000000","message":"Topic changed from openstack/requirements/constraints/boclob to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"a445d0aeb5c38ccee381f437305d0d39433ebeee","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-06 07:38:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6f3bc8d90cb1ac101c87658e7dfd166cf04c2590","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-06 12:30:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d422ddf115102322d9abe1dbbf29f2e094c031b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 13:50:06.000000000","message":"Patch Set 3:\n\nwhy did you rebase?  was it not mergable?","accounts_in_message":[],"_revision_number":3},{"id":"11093e6ace117789bd14d62dee1771c5534318d7","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-06 13:50:44.000000000","message":"Patch Set 3:\n\nThere was a merge conflict IIRC","accounts_in_message":[],"_revision_number":3},{"id":"a4c8cdafaf1b8efafa629615e789f1e3511ef50f","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-06 13:51:43.000000000","message":"Patch Set 3:\n\n\u003e why did you rebase?  was it not mergable?\n\nYes, merge conflict.","accounts_in_message":[],"_revision_number":3},{"id":"5071100653e50db02f74ff48b451190b583ab2a8","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-06 18:26:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a445fad5017dbf1b8ae9c24579444566ec4325e0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-06 23:12:02.000000000","message":"Patch Set 3:\n\nwhy is Zuul still ignoring this?","accounts_in_message":[],"_revision_number":3},{"id":"a96b7888cc2ffb656c2b2d6f2bad06ab2d0976c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 23:52:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc6b6704c2b464ebdf03ba930178c51 : SUCCESS in 4m 59s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4d636cc5182e463198ab908cdbc8f161 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31dc71a80d8b41efa801aeacb69ab7c9 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c43a8e14bdb649c9aef4f36e51af8b14 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eeb38d421ee41ca80bdbbf93879c98e : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f23f3ca53fc4f7c9590ccd66624940e : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2859a1b84c07475bb10d200189463d7a : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8999f408cbd04bcead7d8e9ce229cde6 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f6e9ded9facd4a499613087bc6a294cc : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/329f4c1ab9ce47668389f5222f3bd10b : SUCCESS in 9m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f8ccc7aeb84040168e9e58f06ef0d3d5 : SUCCESS in 12m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64fd43fd237942c49d490e75428c9a64 : SUCCESS in 7m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b6951a8d950143af80c3688e7cf07121 : SUCCESS in 3m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d793b8a6bbb84412afdb5491ae310fe0 : SUCCESS in 8m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/509dff72e6a24c03854ab9a7005668f4 : SUCCESS in 10m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81762fa54d9342ad8b32b5e83a08d6e5 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3290f8a185f242cb9cce324761ad837e : SUCCESS in 9m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/09c3d058f9a445a98032f8fcca426f6d : SUCCESS in 35m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce96fde4e689423296f0dca69292b14c : SUCCESS in 33m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9678ee419e324cc38cb70aa2ab0561c2 : SUCCESS in 21m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6447992fe7f42c2aaf8a3f65feeb7cf : SUCCESS in 20m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8b250c8ecb484a4d8c0c4e4d20f2f21b : SUCCESS in 16m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dfc0d214b5c64c788f59ecbed1c7197d : SUCCESS in 12m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8c215cb565ee426abd07176cac335d41 : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/147132b5252845aca0b7083136ccde06 : SUCCESS in 16m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2790e0fd308d4445b2e0e5047a4cb7fd : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3fef7c98886b4a30b0b1a7e9a24e3977 : SUCCESS in 11m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3d27490785f4b13b4657fec6b19c3b9 : SUCCESS in 10m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a91748471f1845ff8fbfc9cab734cde6 : SUCCESS in 1h 17m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16e78baeeaef4105aa3ba89bb22c691a : SUCCESS in 1h 13m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b4d113e1816f4c7fa4fcdea49239e5b2 : SUCCESS in 1h 48m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c93b61efadd34b778afe2f452c00337c : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d7ce67a57924dddc3665b54d0b56e43d7041d87e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 00:47:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"abe9b16357be1008eaad588bc8cbd279485aeb89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 00:47:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"225441d2fba51975731a7d91e3adf7b383ab85fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:40:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cebee0c2302b42caad397a2e7a43b8de : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dc9814cdd343fdad3d2fcaa9c1821d : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c40af8f7a964b70b5c147708cdb9688 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c039c80550d14b0a84185636be7aa4ef : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2633434ff78c4acba169da9617b56116 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01951fabe855461e88560c5384e681ff : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acbb5d75832e4dffbc15820d81ba349a : SUCCESS in 2m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/22708a489dd442f9bdc257bcce0e3804 : SUCCESS in 7m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38891ecb84ea4b02afe93e397e071cf2 : SUCCESS in 7m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e4f2918f31e044258fba3fb972da8f02 : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bf42c07f34ad44c2990b7d84950697ab : SUCCESS in 8m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54ec3c95aeae4a4d92e9d97cea4a32c7 : SUCCESS in 4m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5a60aeb7e7c43b5b75341192387c7e5 : SUCCESS in 5m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fa8eb7e8b6254e9194908ae48430c385 : SUCCESS in 11m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80d869c4f09f4d8b84927cd812c00c64 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e41dad81ac74fef9304cf714d8bdbae : SUCCESS in 6m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/522135d60d8746beb690621b8e5dc810 : SUCCESS in 34m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/39dd1f975edb4e4d9fd98334e846c50a : SUCCESS in 33m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/946958a853454af88f3c69ddf293e8c8 : SUCCESS in 26m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6ee0eab63d74a6ba18825e225aae76f : SUCCESS in 35m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dd464f6f497c40ef86278ba62e8b5563 : SUCCESS in 22m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ca6b635bcc7422b998522416444f7ee : SUCCESS in 22m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bed580b5ce4e4431aa725dcabca7bc56 : SUCCESS in 25m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25d0fe9ed781447abe4cf56be2d686b1 : SUCCESS in 32m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c10e82893904240a1aef7d6d4b4da83 : SUCCESS in 6m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/03a711b94fd94037ae390a29b06fded9 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed7fa02b694c4a75a3082d851bdb9699 : FAILURE in 8m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/210d483f9a0b4000b19cd5a1eed3ac1b : SUCCESS in 56m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f451f3932a9e41729acfc0484d684d82 : SUCCESS in 1h 14m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/46229c74c4cf4e7cb61a989272857e8f : SUCCESS in 1h 44m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cf25e0cf61cd4fb58b26ca4ad3ea4e89 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9712e9118229753e7bbf624bfa654df8c9dda8f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 02:48:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"800d8b31d9152fbfa95d5f708036183d54224c7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 04:31:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59498f93f7144d3c89eb2bc61ceb5ef3 : SUCCESS in 3m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2ea8ed6d74f148549b4c5bbea1674a4e : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fab9af0710441d38f5600dc6e403ddc : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4f466972d7641c2aa8f3cb546a197fb : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/157212b475a6459f8fcee01656e7d978 : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/487b012098e24fde814d8c7877431ae1 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf090f3df459419393cd3c43c7f19c8a : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fd29d9c80474e5e9ef172702a047ef6 : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b3f726bbdeb14b2698dad56ce6217de2 : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5eb01e213a3a415698031b6708d727fd : SUCCESS in 5m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ae26a3086ec84153ba09c1379a096e4c : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/de8aa7d63bb84a3fba5d2eb557083443 : SUCCESS in 7m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c4ab96493796431ea33a9d5065a6d6ed : SUCCESS in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/479c1d3cf99b45b4bb6c5fb52ceb1c44 : SUCCESS in 4m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/07b04c895242453ab8236f492f1319fb : SUCCESS in 10m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c63b5a6564d54a84a516fa611afb4d9b : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b6f061b3eca4c3aab1512b4133ce49e : SUCCESS in 6m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f1697fde1de547dea407087878dd9511 : SUCCESS in 29m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7ec5e92bc4a477d8aa55c5301f899b2 : SUCCESS in 29m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3e9c95eb2de94a7eb771a2a179189b65 : SUCCESS in 24m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/491e1d4475254842a9eb0cebcec0eb0f : SUCCESS in 22m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d90ee9fbb1a84967b904cd8e7f11f97d : SUCCESS in 12m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/471e1d884e4a4f14a48b49859823ffd4 : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a1218d9b24e4498aa45a7985cd45cfe : SUCCESS in 11m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c3ff7e7d93e42a285dc49d8144138ee : SUCCESS in 18m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7dc8699bf2d54377bc141e224437611c : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4efed139e949421e97d6450da5167703 : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8865acaabf7941408656f1fdaf72ec4b : SUCCESS in 8m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3c3c481244c8478d9cc295ecdd7f2b81 : SUCCESS in 1h 06m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a547465524f46ba81e63ff2ca320d8d : SUCCESS in 1h 19m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1352d90d71814db4a0b3aa4dd2997306 : SUCCESS in 1h 37m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a6394d5a6cc74f4280112b52f510eb3e : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a6a76ccf640961078662be7785fd02279995991d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 06:01:53.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"32375402a408607568e628da0f532cf770d7e25e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 06:21:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"61d2dbf01100b7d1351be6b3b2073f172d2a9465","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 06:22:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8ee07529b3179b2b3ee2189a816cf8cffd9029d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 08:31:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6c9d99fb5d4ca7b121b0ba183e071e : SUCCESS in 4m 17s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c2849cb2e8364462927c0b4a8b77cc36 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8a81a468fa46df97bb8647d48b0f33 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcb5c9709d52403aa2eac905b06eb2a2 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b43d68fd4ff8432982e699f77ae549d8 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e91f3cd90554ab4a1352fea85a56cb7 : SUCCESS in 3m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d542fc371e0f4211860debfad83e6812 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df0bdf46db9740b68132d67ea3246bcf : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0cb002d491434b97895da262f7138ddb : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4352a4bcb9c24befae721b3da8749dfd : SUCCESS in 6m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0e0e3521eedc46b7bddd068a091cadcb : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/21d6c710f81f46b593d24c0c0669008b : SUCCESS in 7m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/37e93f6223894e319c3c9811f66b3c7c : SUCCESS in 4m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/213629c7f7934898873d7ae81e5e14dd : SUCCESS in 5m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9fa01a13341f48f4b638baef0578ac10 : SUCCESS in 10m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bfbe96cc48cc47cf968f292d621df269 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3cb85f0c38a47cc8a042281765def38 : SUCCESS in 7m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/89245ffd7b8844fc9eab8a8a2cefa082 : SUCCESS in 31m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6c51d80bc46144dea3727ebf678a520b : SUCCESS in 28m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/990131203bc748cfa5f42720cfebe053 : SUCCESS in 23m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/548198a3c30d4fd98c2af3cc4c12635b : SUCCESS in 22m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2b958085748841499e07fd936fc89935 : SUCCESS in 15m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e13c0e091bb4491e95d43d85e5dd098d : SUCCESS in 12m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7a4b18943b4c4aeaba6e50d54b2435e0 : SUCCESS in 12m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40bf3e1c7c664ae2ab3ee79a8607d4b9 : SUCCESS in 19m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24ee84128e3249dd9b6d349a02d4a45b : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac999d0ac9e749f9be8c4e19fd5d8bda : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7f45d5ae43e43eabc66b166ce878baa : SUCCESS in 8m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a03c67d09d3b4c3daef1998891db4746 : SUCCESS in 57m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36c6a8fe97434377bc1ac2cda60a302f : SUCCESS in 1h 11m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0345914e8f444888eb79233e108f44a : SUCCESS in 1h 33m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f24cd3e19ceb43709c5cd8737b782973 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b20a31b9a9ad55e5b1f8247fb2f32cb8b8b9cbbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 08:32:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2e5dcc4c4b812d9038ca0bfa0ca555815e91a884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 12:23:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e3af7f8a8b42859a573f7c1b6ec277 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1778eb96c0748109a1654fb39d9924b : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de053e97a0b64a24adeb97e46dc0b123 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f9702192e72422c982840e00a133d8a : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/003a0d95fa944ca8ac96c18ee39ab5ed : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/760ac2335ac64fe48e16b1afd4dd83cf : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/597efdcfd7fd40dfba0c7b8c443c9ec4 : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ee91d2fcb5224beab58f58c79f598448 : SUCCESS in 8m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/358e30db5a0a40a7a79b1a28c7e476fc : SUCCESS in 16m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8ddef8a655ac48458ccd7bab9e6d9df3 : SUCCESS in 7m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/686c328e1a57433b93733c83ac2e0bf1 : SUCCESS in 15m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f78c0c3bf5bb42f2902ac90a024b1754 : SUCCESS in 6m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/28c6f25ce6d94175b0da77c5f0c0d87d : SUCCESS in 7m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eff565456c4a456386f89493724176f2 : SUCCESS in 10m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9d7ffbab784a40c08f83458d09cd7461 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5259f4a326174d35938f3ff262896696 : SUCCESS in 8m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/65abfeea02d343588c3edb5349bc4f42 : SUCCESS in 39m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ad31a13aedfb44c9b7143d6524ffe937 : SUCCESS in 24m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ebfec3ccdb74474297c3748f1c747b3c : SUCCESS in 32m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/68ab3a84a3c84c93b14ebe65b64d2b85 : SUCCESS in 27m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cf6d581849f64bec8e2e52bc266458c4 : SUCCESS in 20m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc31909807db41ec8076ea3dadc16202 : SUCCESS in 14m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a20e90b106f3467e8e40598f08f02b1d : SUCCESS in 23m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75e09c5be4654148b47c96de56c5f9aa : SUCCESS in 21m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/227ffe0c5b954d21a4f7aadcb2184467 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/63c5bb8aba43404d9880ea4abbe27bb5 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1901031568ee4cfba83350ffb3d3b3bc : SUCCESS in 9m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0461ca7bb4b0488d88e5a28347ae8b12 : SUCCESS in 1h 20m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e6948fe1b5d4f478a05fc043d64359f : SUCCESS in 1h 35m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8273fd9a5f7047bb9939c0db223f3c00 : SUCCESS in 1h 41m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a3dd1bb54f504958868dbff2d4c57931 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":4},{"id":"db64402f91a7cdfdc6b9f623183fbe32fc889507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 12:23:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1e4ad27a808d5a48739e92ada0eebd0a3b394e05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 12:24:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635894ec36b44b6d9ad5e9b7a34113e4 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0b62855049b1b335f9902432ac6c5fc8c059f161","revisions":{"d13456db659310c701f69c0e5d8fe8af145957d4":{"kind":"REWORK","_number":1,"created":"2019-09-05 07:13:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/680304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/680304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/680304/1"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 07:13:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 07:13:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If1507f44eb88013cf08aea4b46891876fa754cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d13456db659310c701f69c0e5d8fe8af145957d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d13456db659310c701f69c0e5d8fe8af145957d4"}]},"branch":"refs/heads/master"},"843a45477918536a56c4fbf59ce2a88ab32d12cb":{"kind":"REWORK","_number":2,"created":"2019-09-05 16:14:28.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/680304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/680304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/680304/2"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 07:13:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:13:56.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhorizon is STILL broken\n-semantic-version\u003d\u003d\u003d2.8.1\n+semantic-version\u003d\u003d\u003d2.6.0\n\nGoing to just blacklist the [REDACTED] release\n-ansible-runner\u003d\u003d\u003d1.3.5\n+ansible-runner\u003d\u003d\u003d1.3.4\n\ntempest STILL has failures\n-stestr\u003d\u003d\u003d2.5.1\n+stestr\u003d\u003d\u003d2.4.0\n\nwaiting on https://review.opendev.org/679842 (team awol)\n-jsonschema\u003d\u003d\u003d3.0.2\n+jsonschema\u003d\u003d\u003d2.6.0\n\nkeystone STILL broken\n-oauthlib\u003d\u003d\u003d3.1.0\n+oauthlib\u003d\u003d\u003d3.0.2\n\nChange-Id: If1507f44eb88013cf08aea4b46891876fa754cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/843a45477918536a56c4fbf59ce2a88ab32d12cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/843a45477918536a56c4fbf59ce2a88ab32d12cb"}]},"branch":"refs/heads/master"},"3e4b42b48cce2c454835e68b5fe6b11a3c280cdb":{"kind":"REWORK","_number":3,"created":"2019-09-06 12:30:02.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/04/680304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/680304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/680304/3"}}},"commit":{"parents":[{"commit":"d39de7a0a4300222e11a8c2e30aab35f95e54d4e","subject":"Merge \"update constraint for octavia-lib to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d39de7a0a4300222e11a8c2e30aab35f95e54d4e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 07:13:47.000000000","tz":0},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-09-06 12:29:42.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhorizon is STILL broken\n-semantic-version\u003d\u003d\u003d2.8.1\n+semantic-version\u003d\u003d\u003d2.6.0\n\nGoing to just blacklist the [REDACTED] release\n-ansible-runner\u003d\u003d\u003d1.3.5\n+ansible-runner\u003d\u003d\u003d1.3.4\n\ntempest STILL has failures\n-stestr\u003d\u003d\u003d2.5.1\n+stestr\u003d\u003d\u003d2.4.0\n\nwaiting on https://review.opendev.org/679842 (team awol)\n-jsonschema\u003d\u003d\u003d3.0.2\n+jsonschema\u003d\u003d\u003d2.6.0\n\nkeystone STILL broken\n-oauthlib\u003d\u003d\u003d3.1.0\n+oauthlib\u003d\u003d\u003d3.0.2\n\nChange-Id: If1507f44eb88013cf08aea4b46891876fa754cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e4b42b48cce2c454835e68b5fe6b11a3c280cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e4b42b48cce2c454835e68b5fe6b11a3c280cdb"}]},"branch":"refs/heads/master"},"0b62855049b1b335f9902432ac6c5fc8c059f161":{"kind":"REWORK","_number":4,"created":"2019-09-07 06:21:55.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/680304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/680304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/680304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/680304/4"}}},"commit":{"parents":[{"commit":"b56e70a02dce3afde768f947d340a1abcc011a75","subject":"Merge \"update constraint for cliff to new release 2.16.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b56e70a02dce3afde768f947d340a1abcc011a75"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 07:13:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-07 06:21:39.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nhorizon is STILL broken\n-semantic-version\u003d\u003d\u003d2.8.1\n+semantic-version\u003d\u003d\u003d2.6.0\n\nGoing to just blacklist the [REDACTED] release\n-ansible-runner\u003d\u003d\u003d1.3.5\n+ansible-runner\u003d\u003d\u003d1.3.4\n\ntempest STILL has failures\n-stestr\u003d\u003d\u003d2.5.1\n+stestr\u003d\u003d\u003d2.4.0\n\nwaiting on https://review.opendev.org/679842 (team awol)\n-jsonschema\u003d\u003d\u003d3.0.2\n+jsonschema\u003d\u003d\u003d2.6.0\n\nkeystone STILL broken\n-oauthlib\u003d\u003d\u003d3.1.0\n+oauthlib\u003d\u003d\u003d3.0.2\n\nChange-Id: If1507f44eb88013cf08aea4b46891876fa754cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b62855049b1b335f9902432ac6c5fc8c059f161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b62855049b1b335f9902432ac6c5fc8c059f161"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
