)]}'
{"id":"openstack%2Frequirements~680339","triplet_id":"openstack%2Frequirements~master~I517484c9b8d34a2126c77e034f0b7172c6ecf60b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I517484c9b8d34a2126c77e034f0b7172c6ecf60b","subject":"update constraint for os-brick to new release 2.10.0","status":"MERGED","created":"2019-09-05 08:47:31.000000000","updated":"2019-09-07 20:37:58.000000000","submitted":"2019-09-07 20:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680339-1567888565210-71632aeb","meta_rev_id":"8f149728f2765b65f008f5a06931015acf3ceb65","_number":680339,"virtual_id_number":680339,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-07 20:36:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 16:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 16:44:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 21:53:10.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-07 16:44:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 20:36:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be6646258aab7acd9d557c63fb308750ff98f4ca","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-05 08:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fcffeabc8171d1b1a1a82965c785ba9b4ed7881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:08:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9e45529ee041ed877afe32f15ed380 : SUCCESS in 4m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/95c79b95ed2345ae9a57b0520abe9426 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1acb653d804c089c447696f5a093f7 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c277ca13ca6e454ba65eeb8fbab34651 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a264ce5b6454295bec6a2aa9059fd82 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a82c4572f50f4010904fb1de1b10e4c3 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffa9d9d3d40241f29a81ee161c400d22 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/237607a22a314a4485568ecdccdc6421 : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/df3ba561691242199265f69a606edaeb : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa09ab275bdd4c6a8100dd2fe2a6538e : SUCCESS in 16m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cfaea8d7b38041158e7f0e6d5c665702 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/184132daf2ee4896bf5199b6603dd7ef : SUCCESS in 7m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c0a548a2ff954740baf7553acde386ec : SUCCESS in 4m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f99ba56d68c148fe8d9f9d1c1f596b0b : SUCCESS in 4m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fe1c5630233c4edd8e5eb845717ddd63 : SUCCESS in 9m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/571213abb99c4ab896aafd27a6dbe004 : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bf212eb0e084bbc9f9bcecdc6bc04a9 : SUCCESS in 10m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9a74d0b78ea24bec8ed203ed5a466fe8 : SUCCESS in 30m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4036f30d4a644bb1bfc7f85203b88b1f : SUCCESS in 33m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/dd97b0019b6f4448bdb5e95cf61094c6 : SUCCESS in 24m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/643657995bd64576b0c0392bae1ff1eb : SUCCESS in 21m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7e1407cb1452417e9022d25560323d01 : FAILURE in 15m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8d4a57a64f914746858fb6e71df58238 : FAILURE in 11m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/329cc9d3d7274047a7a4737e71ae772e : FAILURE in 15m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/72c5d96ee89a458388b52a7e159a04f7 : SUCCESS in 19m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8329e1337ff478590af099eba1b48c7 : SUCCESS in 4m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0ac2040da1fb42de93d6f921cd71f2c8 : SUCCESS in 8m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d1aaa466b8034a00b2d60aa7e245b1b0 : SUCCESS in 8m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/210a82b536ab423faba002e4d06a79ca : SUCCESS in 1h 38m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b13fe7b6ab0f4b0e98c4eaf55a4bfd23 : SUCCESS in 1h 40m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f98a530345df4327bf6cac57b6f1ab20 : SUCCESS in 1h 43m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a09dc42e6823478b892d92eb2cf80031 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"68a41b24a733e78cf812d3c72d969053b4484a4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 03:32:44.000000000","message":"Patch Set 1:\n\nrecheck\n\nI\u0027m guessing that nova is just gonna fail though","accounts_in_message":[],"_revision_number":1},{"id":"703b5f9758bb071377064dc87c6124d4ad8b3e84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 19:08:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ea3cddb95b403aa610b4b4c3e52ad6 : SUCCESS in 3m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9ff24611728a4f7fbdb95e624560685d : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a08d17a40e3472cba68ea44e9650db8 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9397007f6f8d4c01a8afdd7d0138612f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23b5199027664e868a0d446a7f40d7c7 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f8557f3bcfb49cf9048fd4fc22fe8ec : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/048347ec87614675970669b0cd04c99b : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5bccb9e377284747b1b79a2f4b15e122 : SUCCESS in 2m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a57aa578ed9f4f089bc35ef29a0a2025 : SUCCESS in 6m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba6f06d6166447b793290b1c96389ff5 : SUCCESS in 6m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7a8aa19bf4e34be29bc9459219c7423a : SUCCESS in 8m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/653b59a230584e709978f9c4048af0b9 : SUCCESS in 7m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/20a48774d1d64ffab22c8e11929d3452 : SUCCESS in 5m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63b825c191384e01a1066bada25336e1 : SUCCESS in 4m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ab59023b55864cee97f893879b6cc53d : SUCCESS in 10m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cd4ad2d481ef4cd382f48ebfd7a5a347 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef074cdfc7db4926866a0d80142bb3d7 : SUCCESS in 10m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b6ef15c87c1a448b940b6c99f2c7ff9e : SUCCESS in 27m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/40d8f6a81ecd481986596189cd8e6003 : SUCCESS in 34m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c0ee558abcff47ba9004b21533eef3db : SUCCESS in 22m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/74b73956d01a464e8a5b7148e9970ae8 : SUCCESS in 21m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/67825572a8c147ac9eb6775284d3d80e : FAILURE in 12m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/165c3db54895427da33892ea3ec79614 : FAILURE in 13m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/268546fa935e41ecb66aa4a326185355 : FAILURE in 12m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac3265a455d74d5ca2ee58d3c06053f6 : SUCCESS in 20m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6bea718e3bdb4aa0aaa8e58e14a92604 : SUCCESS in 4m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b72fc4eb3c704b9eb1367cefe6f33e23 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b60876e3c3c149c8ad996104b68ea388 : SUCCESS in 9m 00s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0df92f13d9d94e9f8012a5613b049f5b : SUCCESS in 1h 17m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/256eedd9b4764844b63b73cf0884648b : SUCCESS in 1h 18m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e05fd5cbfcbb43d58485275426cfbc1a : SUCCESS in 1h 54m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/01146bf0cf304367b31fe777acae39be : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b33525afa7307b4bb08119043915dff8e16d7ed9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-06 21:53:10.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/680779 should fix this","accounts_in_message":[],"_revision_number":1},{"id":"75062d85c3115db3475d79ccbcec5ac18b697f9e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-06 21:57:55.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"03842186ab420315e31845e0ef62b895c912f452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:47:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69415bbf4f3747729063ccae9c0d1018 : SUCCESS in 4m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/dd5f9c368734425cac5d46de43d59ed6 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf252363b4b4e069ee6d6058e72633a : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1765b6f3c8f94611805ba58fa4a3b0c1 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7599caf75ab545a58042e6e3af98f31a : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a5435c4282b469e84568ab6cf38d640 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbe84e81d3024505b0020654762eabbf : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cb99687807f4b4baca9e58915cbf6cb : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/837a3c1f1af14a15955b33f2578e1a44 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f0b61ebe7f7435a852c3dc2a00463d1 : SUCCESS in 6m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/34677c4c15bd488195620e94ef17eab6 : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e6ca773c4a24c48bb11cb1ab9c14177 : SUCCESS in 11m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/99c1da756208414280cf8e82babdc77b : SUCCESS in 5m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f8258df6842943ba86409e5f9892ec52 : SUCCESS in 6m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/edbca37031294d6f90b83d921e974493 : SUCCESS in 12m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d1cd38159404981b5d1d1bf5da8ce18 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/848dbbe72d4048859b1c9a1da2b730ac : SUCCESS in 9m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b00c95e9955145cc9078c7bd4bec058b : SUCCESS in 33m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/025b59903b164c489ce70aeb03d6451b : SUCCESS in 37m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9b98a11bc66d4d7ca537b54aa0f4626e : SUCCESS in 27m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/809dc3b2d52945a5bd83ee6b1d9b3bde : SUCCESS in 23m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3239336d83294d22bc9403ed5e8db0cd : SUCCESS in 14m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b025927f7a04104bd1f8b84532bb96e : SUCCESS in 13m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1290acad283c4f198c00e11625de16a0 : SUCCESS in 14m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cebbb8d6b09e4b1b8de227742af7e308 : SUCCESS in 18m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e062ff3d9a14e1495fdd91350e2b654 : SUCCESS in 6m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a5c3ab9d7e3240ad9a1b7243e05ad4b1 : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e04f266c17f340eda52d6dee1bd5c12e : SUCCESS in 9m 29s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dd000329a9fa4d528a91d1d2b0d98f74 : SUCCESS in 58m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c597c277f038435193a48c5b4c4ac2e4 : SUCCESS in 1h 15m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0f92763632c4ecb8d79f4086de441a5 : SUCCESS in 1h 43m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a631621785814330bd72cfea46ee1874 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4a8340ea2e8d60287a371b1c9cf309e4beabde07","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 02:52:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"98dbb85af2fd323053286e963ff60278b0696eb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 12:23:19.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"d1f3bdf581010ae9873ce81934b4d17c2931b2b3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 16:43:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e947586b005acef4125e97961ff0c93a289fd966","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 16:44:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"00e2d028f59a2b83d4bc60c152a9b4c94ce310da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 18:50:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b66ff8f1cb41c8906028786c88b279 : SUCCESS in 2m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3f5ddd22b14b416f91d93437bb7a6a63 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c804e648b954de3a977d1d22fb7b7e6 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d54e499590674781be8342fe192d5006 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a8abad555843d4a3f3fb2e806da2fb : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af3aab7fa06d46c9b436f7be88ca51fe : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9b02ba404054daf8eed8b472fc9274a : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b211f4da4e443ccbe893e278ae276c7 : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cbc9e5e4c1a0437f85a8ba342c019e76 : SUCCESS in 13m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/533e943f98624a268d61d524cac0d956 : SUCCESS in 6m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7adf8c72972b4321984ead9e28067c7f : SUCCESS in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/acca48635dd4424ebcc03734b1c97710 : SUCCESS in 6m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/10b1739f9cae40cbb175eae14d9583ab : SUCCESS in 3m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea1f4f92311f4f7c896e29d5fdb14955 : SUCCESS in 4m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c12f3a86c123449bb1d477ed2a501379 : SUCCESS in 10m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ef704aa88114484d8d8b5a296510d2aa : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23690a79f4d846c19b091970ee6af35a : SUCCESS in 5m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b73db7458ab04297abda45a0bd806234 : SUCCESS in 28m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af8064aea783422e81eb5de64e819a1c : SUCCESS in 29m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/408ddd297e764208bf0104ecca149793 : SUCCESS in 25m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce40c55805204fa39f165b2e781a6b75 : SUCCESS in 20m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/987544964b6447ebb668f9b0be098098 : SUCCESS in 13m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/93eab4aadc734c03a45264d886b05b7b : SUCCESS in 11m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6b71d33bf419445daa9df1ad33b99d49 : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89185544a0f8438881a836c3f5f65761 : SUCCESS in 22m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e4cae96e7a14855bc93971010b42e1a : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af2caf38b1cf4fa6b6bbfafc8f4f5bac : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7de0bab636b4e9e9925728fa5016c15 : SUCCESS in 8m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a8524f06ee054c79bd3c931f85de3d20 : SUCCESS in 58m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc7a0e4ff7414086ba2189ca67b8251d : SUCCESS in 1h 11m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fe8bf60edd014ac3a1a9dca03ab22f4f : SUCCESS in 2h 00m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58fcf114cacc48fb900031261253a1bb : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":3},{"id":"de350b4c4e0c1318cd3209030611d9236a9d6bf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 18:50:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f8fb990833b1a349a0b1e2e3327480a69760d25b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 20:36:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6f871bf5f0432f80f0b92b2ea96a26 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f0fe126ccb41cbb3b8709edb657baf : SUCCESS in 2m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3359c17862ea4efe966f3e4f7a973404 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04820742621146a28b4563d7678dd11b : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/882f6fadf4294040b30415a9921be78d : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62a39cb0885f490cbc7d5986aef8568e : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3ae40adc04041f48454472dbebb7c60 : SUCCESS in 2m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/099776cd74d244eda825dc2ab154950b : SUCCESS in 5m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f47ed744686748398f9f758d936198be : SUCCESS in 5m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7dc9771f5c244ed183515cd289f80289 : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/79c6398159b24caa8fd8e825e106229d : SUCCESS in 8m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/877fe6801a044f93801e27f9de024f51 : SUCCESS in 3m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ece6e394bc7f424a999232748bdb70a9 : SUCCESS in 4m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/95b9286a5e6e458ca2372343d0cab921 : SUCCESS in 9m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6fbe54b35e44d53a960c84c8f875ac6 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5cda4a5b21e4704bbaa136a19e2f8c7 : SUCCESS in 6m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b73b9a29146340dfb5dfdcfb6c44985c : SUCCESS in 30m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/488d2d730a774b469f7fe5705304fc95 : SUCCESS in 26m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8837b3b38ae24e4bbdf97ab49af6d031 : SUCCESS in 22m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d2779d5a6724412ba069fa9fb45f3ea7 : SUCCESS in 21m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1f5dc4c6bfd64d4facbe6aeeeef39813 : SUCCESS in 12m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed69b46c14054144b1694c7479e97d2e : SUCCESS in 11m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a4657ebde0ba4c4e8351495da0837211 : SUCCESS in 11m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6685ccf919f45318f60111db8440c19 : SUCCESS in 22m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f15b6d1cad9d4dd483d17190cf301696 : SUCCESS in 4m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/92a9eb94c510443ba03a014fa44f21b6 : SUCCESS in 8m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec609674af864d4f8986c373723927d9 : SUCCESS in 9m 01s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c470b9adc0e44755bc312d52d23ddc9e : SUCCESS in 58m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59b13e8be2da4b27996898ef8d17531b : SUCCESS in 1h 08m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b03277a5d88b499ab22a7a285fe05750 : SUCCESS in 1h 32m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eca19a9e20c444d3956b511724b45462 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":3},{"id":"32339416936509596e9192655e2d73c72a9c6f5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 20:36:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"11a6a301d31ce4744ccb0d0041f913033f27ba41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 20:37:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b22646b56e4f2c829f4890f0d7fb1f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3925a8c2e8a9223db81ad6f8f49f4ae7cfee2a79","revisions":{"d232b60f37b603ce675ebef57986e95555649461":{"kind":"REWORK","_number":1,"created":"2019-09-05 08:47:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/680339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/680339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/680339/1"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 08:47:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 08:47:28.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.0","message":"update constraint for os-brick to new release 2.10.0\n\nChange-Id: I517484c9b8d34a2126c77e034f0b7172c6ecf60b\nmeta:version: 2.10.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I60841f7d4e572db3fb06cc58303dd611a6af891c\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+1: Walter A. Boring IV (hemna) \u003cwaboring@hemna.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d232b60f37b603ce675ebef57986e95555649461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d232b60f37b603ce675ebef57986e95555649461"}]},"branch":"refs/heads/master"},"e9224569ed6e3e9028ef03b4d50da063c2dc4a1b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-06 21:57:55.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/39/680339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/680339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/680339/2"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 08:47:28.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-06 21:57:53.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.0","message":"update constraint for os-brick to new release 2.10.0\n\nDepends-On: https://review.opendev.org/680779\nChange-Id: I517484c9b8d34a2126c77e034f0b7172c6ecf60b\nmeta:version: 2.10.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I60841f7d4e572db3fb06cc58303dd611a6af891c\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+1: Walter A. Boring IV (hemna) \u003cwaboring@hemna.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9224569ed6e3e9028ef03b4d50da063c2dc4a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9224569ed6e3e9028ef03b4d50da063c2dc4a1b"}]},"branch":"refs/heads/master"},"3925a8c2e8a9223db81ad6f8f49f4ae7cfee2a79":{"kind":"REWORK","_number":3,"created":"2019-09-07 16:43:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/39/680339/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/680339/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/680339/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/680339/3"}}},"commit":{"parents":[{"commit":"74220bae1e306be96010c2f527c3497cc2856d48","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74220bae1e306be96010c2f527c3497cc2856d48"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 08:47:28.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-07 16:43:06.000000000","tz":-300},"subject":"update constraint for os-brick to new release 2.10.0","message":"update constraint for os-brick to new release 2.10.0\n\nDepends-On: https://review.opendev.org/680779\nChange-Id: I517484c9b8d34a2126c77e034f0b7172c6ecf60b\nmeta:version: 2.10.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I60841f7d4e572db3fb06cc58303dd611a6af891c\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+1: Walter A. Boring IV (hemna) \u003cwaboring@hemna.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3925a8c2e8a9223db81ad6f8f49f4ae7cfee2a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3925a8c2e8a9223db81ad6f8f49f4ae7cfee2a79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
