)]}'
{"id":"openstack%2Frequirements~680343","triplet_id":"openstack%2Frequirements~master~Ie2fcf8f90e4aed45418ff4d879eaf7df18af260d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie2fcf8f90e4aed45418ff4d879eaf7df18af260d","subject":"update constraint for octavia-lib to new release 1.4.0","status":"MERGED","created":"2019-09-05 08:55:44.000000000","updated":"2019-09-06 05:30:23.000000000","submitted":"2019-09-06 05:28:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680343-1567747710643-76a6aaaf","meta_rev_id":"2cf2b26d58f059d71b45f8620b9576269911e605","_number":680343,"virtual_id_number":680343,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-06 05:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-06 03:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-06 03:32:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 03:32:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-06 05:28:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf253e7710fc286ed94037b69d5deead4aecfb3b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-05 08:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d9c781d472f5eaf53fd2ef5b8eeaa0ea758213f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:25:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0099717d8dae4aeab2147018be54fb32 : SUCCESS in 6m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a74c4585d2c34153a42b7f483d011fb6 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8229447b30df4d8dbae77559d64a3b85 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a77d80bffc2d44f08e7872ff4f6c27c3 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39606d7f3d6c4b1286882d54acbdc416 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5f02b8532994a7198284101a1a530b6 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f608ed7072e4b08ab99a5133e8acc36 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f48333ad126b4772ac3214127f7fa3d0 : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b736f5057dbf41d58f015372c49cf7cb : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b932a497aae2428da1d6da5d4b37b1fd : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ad349f39a3854a6cb4ededfd3c9bd156 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a40b14d61c0448ca6db10fe0a0dea49 : SUCCESS in 8m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/378497092cd74a40a830259481d5c293 : SUCCESS in 4m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/52c6b68b838f41a688bef136931dd2c0 : SUCCESS in 4m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b19f8043b4654bc4b8aaf0c4949f5506 : SUCCESS in 11m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ee354e1b50344e8b835ea44ec5c8c3a : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5cfb7ddf372f42c2aac4eb7db176c66c : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c5dcd2e6b7f74ad19511f81788812c77 : SUCCESS in 31m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2d21497484be462092d47d63a90b0ec7 : SUCCESS in 33m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0998a11442f646f1ba1f5d412de2de5b : SUCCESS in 29m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/83d56b41c1d844498194957d96ee517d : SUCCESS in 27m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a00ac14fd1f04d0b963e99938a12147b : SUCCESS in 13m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fdf370feaa4440b0a26034b61df0c79a : SUCCESS in 11m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6d3e76cbed9343679e1ee6ab84cef57e : SUCCESS in 12m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c978108771d84c64a66ae1475054d01b : SUCCESS in 23m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa24ba8629054d6f8879d3678c6f3d42 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f2a73adb531c48bfb16cadd97433b2a6 : SUCCESS in 9m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f24ccbf9d034509aaf04d7870a38293 : SUCCESS in 11m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a9e1bb688cbd4b0e81ff4175e4f675ed : SUCCESS in 1h 27m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c88997d688144774a7a1d5cf53452e39 : SUCCESS in 1h 12m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a7159cee149546b49bb99648a083af21 : SUCCESS in 1h 56m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2afe81124a84659b40b8329c7f00d9c : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f5d76e50a371d64e01a148cd4552d23533ce44b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 03:32:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e06a8522778533c6756222f20d49119177e103b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:32:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34de71ae712ec4ece11baa34eb729ccec1f60656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 05:28:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83581281269541c3a2c8ba5675aec0a5 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1661e541f33d4a2697e6918b8b4f88d3 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0e7aa3e3e8c485bb3b97d4830aec118 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7ed1e3f530f4890827686e292ac1a92 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7017aa96794447b86c9ccbe7eadea84 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/303b150a96da405fbee0cfcb62a41d0b : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fff82c49e4984f068cd581688ef62fbe : SUCCESS in 5m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a4230b87354d43c7a29c7e96aaa7c6f1 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e1088d957ded41c08e66aef8982f847b : SUCCESS in 7m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7f08e0af213142a3ab5353e2279f5a17 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/620eab00a1fc40fea1a246a3fb9099c8 : SUCCESS in 7m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/585a520ab8d445528748c6da44dc4296 : SUCCESS in 5m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/87a29c5973dc4cd7816fe0138c60633d : SUCCESS in 3m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c3d2cbd6393d4277acb236bce8d4dfc7 : SUCCESS in 10m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a0b7904b73d8418ca30b4b0b171cec0c : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0be5bcc998147feb3229133113a319e : SUCCESS in 9m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cbd789fa545c4a188e0310495758dd0b : SUCCESS in 32m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1aa05f1656c64605b399f62998511d96 : SUCCESS in 28m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/00ba33b8c0a445a785c781c45f7d9877 : SUCCESS in 26m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e4306d4c74e24a788264c99cb4917771 : SUCCESS in 21m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/65a55505c59c45eebd3e0a189a7f9f36 : SUCCESS in 14m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/076ce4d208c24dd5a118c5d57e55903f : SUCCESS in 14m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3f838a71da7e447daa4f573461dcdc97 : SUCCESS in 14m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47f4b83190134ad7a4373424100e08bc : SUCCESS in 20m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c120f77b03f4d1badb9213284bfac5b : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fe38b27dd49e41beaa92cabcce8459dd : SUCCESS in 7m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3853b04862424127bdd6f6c54fe4c7bd : SUCCESS in 8m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a373fb098efc4d9c9132d931ba303fff : SUCCESS in 1h 01m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d117abc079049729b772f380b11e7b0 : SUCCESS in 1h 16m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/29f7561a7f19461ebbd7f842a33e6d4b : SUCCESS in 1h 45m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2d7d19383ec34cfba819678e3e63451b : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d5d3838e5632dfb86b713864554099c42b30086a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 05:28:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1213a89d16eb840fc06491d29665d280b1e24de3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 05:30:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f011e35f6a2b4450befbf5e73aa4e899 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5044ede3f0a9460b0330dadd0142b8924e4471da","revisions":{"5044ede3f0a9460b0330dadd0142b8924e4471da":{"kind":"REWORK","_number":1,"created":"2019-09-05 08:55:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/680343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/680343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/680343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/680343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/680343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/680343/1"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 08:55:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 08:55:42.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 1.4.0","message":"update constraint for octavia-lib to new release 1.4.0\n\nChange-Id: Ie2fcf8f90e4aed45418ff4d879eaf7df18af260d\nmeta:version: 1.4.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Change-Id: I20b00b56419ea0fed841737f89cb8749c48a2155\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5044ede3f0a9460b0330dadd0142b8924e4471da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5044ede3f0a9460b0330dadd0142b8924e4471da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
