)]}'
{"id":"openstack%2Frequirements~680456","triplet_id":"openstack%2Frequirements~master~I3f4b55391fdd648639d73a49b764fb96afd33246","project":"openstack/requirements","branch":"master","topic":"block-ansible-runner-1.3.5","hashtags":[],"change_id":"I3f4b55391fdd648639d73a49b764fb96afd33246","subject":"block release that was removed from pypi","status":"MERGED","created":"2019-09-05 16:19:53.000000000","updated":"2019-09-06 15:43:34.000000000","submitted":"2019-09-06 15:42:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680456-1567784529087-72f0986f","meta_rev_id":"c34aa74c6be710367baf4e83f2ae11d7cefd27f0","_number":680456,"virtual_id_number":680456,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-06 15:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-05 16:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-06 13:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-06 13:50:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 16:52:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-06 15:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"345cd2a238d23d5e0c6f78e0806f999bb105e103","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 16:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2df0de6c294ea97466e649ec564ed0e410df18bf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-05 16:52:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11628dc2d2e4d5c0037dbf89da7041ef26ae285f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:13:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d067ab7b7063470c9cb8aff94e44a4a0 : SUCCESS in 4m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/33103b9b3669476581b8d75e06107366 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77489b1d4ea14ad0a7a4d82eb5c744a0 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7407bfc2f874a5ca63fe5b2dba413f2 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23357d67ed834839ad0fbc3f48f65b6f : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/763d63772965462f96c9992925ca2de3 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7b4a742181748dfa01442cef6adaac8 : SUCCESS in 3m 09s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/db499e32cef24b9eb284f5c47f283844 : SUCCESS in 1h 09m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56a852bb9aaf460aaa0827eb8310aca6 : SUCCESS in 1h 11m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/97c6edda27284423a5ea9044b0c2c11d : SUCCESS in 1h 41m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a7b574206ae74e3e8a13316f756d4428 : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"41ca948e751c0ccc5f24eafa9874edb1ae3c9bbc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 13:50:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0658f794bc63647424df8e696a398fcbea7a7b11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"571f62faf9202126d4231bf003d8786368059a17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:42:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7bbda075c24b06a34eebd0fad68119 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866d632197c349e386f6ead49164a957 : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a38b20ed7714b8dbd2d1f43160fa971 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/207c2600788b43f89fb677794aefa132 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fea882196c0461e9695a59fc556ba8e : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97c1582f56cb4096a041e40fe429ecb2 : SUCCESS in 3m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/22ef089a3c4d478a8a8ecea1ecee1c44 : SUCCESS in 1h 21m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e88238a24574b47adfbaa40088969cf : SUCCESS in 1h 21m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d90dd206f7e410a954b02f81d618902 : SUCCESS in 1h 44m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/be253e640e0e4044be656535a6b0b7e2 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ab29762a93266d8461ba8098a313bb64b4530710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:42:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f00263457ffe062195d9bf6e8fa8125ebd1646b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:43:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e8fc43c6b948b5af1fa7c266c1be1d : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87eefa80e6378a2b6bddc26a065fb56c2586fe6a","revisions":{"87eefa80e6378a2b6bddc26a065fb56c2586fe6a":{"kind":"REWORK","_number":1,"created":"2019-09-05 16:19:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/56/680456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/680456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/680456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/680456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/680456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/680456/1"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:19:22.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:19:22.000000000","tz":-300},"subject":"block release that was removed from pypi","message":"block release that was removed from pypi\n\nThis will stop it being proposed\n\nChange-Id: I3f4b55391fdd648639d73a49b764fb96afd33246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87eefa80e6378a2b6bddc26a065fb56c2586fe6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87eefa80e6378a2b6bddc26a065fb56c2586fe6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
