)]}'
{"id":"openstack%2Frequirements~680457","triplet_id":"openstack%2Frequirements~master~Iae4e8cd7ff8ad35f9db501258377731a585a0b6a","project":"openstack/requirements","branch":"master","topic":"update_semantic-version-2.8.1","hashtags":[],"change_id":"Iae4e8cd7ff8ad35f9db501258377731a585a0b6a","subject":"update semantic-version to 2.8.1","status":"ABANDONED","created":"2019-09-05 16:21:04.000000000","updated":"2019-09-18 21:53:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5491761aa3915ae04afb6e743964baf4a1eca1d","_number":680457,"virtual_id_number":680457,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-10 14:54:22.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-09-10 22:55:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 14:54:22.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-10 22:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c50d489f95044c73ae4d8afb2139952a2326a4c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 16:21:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"620650f14333b22b0aac909c5edd5c3b7a4d70ea","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 09:11:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b714431bfdb6e92f42204703c9b656ff2b4c7f22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:29:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6191e2a85d2c462f8197d72b04e4b1db : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c7cba91ff01e44828f9f7e47326604a4 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9740cf36935428db45620adebdb2c2a : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd4bb838f00c4d05af4e448c9d2622d8 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1ccedfc7e234ab5becadb9cd6a7800e : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66b17d6ec75440969a8056f0aff784de : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0a45bca62a94d09bd50bb7cc103affc : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfee58b34fa347a7882923e0e4f258be : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c33bc7afeed84ad1a089df8d7e659c76 : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/044b19ce07e240cb8a4b59ed9fb4202d : SUCCESS in 7m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ca8bb80cc7a5475cb00bf9e5a29a068b : FAILURE in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1a973a5cdfbf44928d8e74127fa22c74 : SUCCESS in 9m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a8a526c9bcb246188e4ca676598ad186 : SUCCESS in 4m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3b3f30a094349aeb4c0dbdbf9af101c : SUCCESS in 3m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e220f88f64ad4f0e85c639424754f242 : FAILURE in 9m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/599a02db69f247f5aa75e79e17237ffa : POST_FAILURE in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d68b76813fad43079c71126630c0254a : SUCCESS in 8m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/57e6383e5a184d24bcc7282c3b52393f : SUCCESS in 35m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0fc20795b94435b8eeab8938c3f7b70 : SUCCESS in 34m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/04599652881c48ae9a381cc248f7cdb7 : SUCCESS in 22m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea4f6eef2d6a4b019d4dd2ac71fb282b : SUCCESS in 21m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/86083dd53d6943b1b5cb4eed7e6c0c1c : SUCCESS in 13m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b078a5bd68424f939b172041c76ba2d1 : SUCCESS in 12m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/500de2f294694985866e949b381fc839 : SUCCESS in 12m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cbd133ac685b4cc18eb4e13d9851958f : SUCCESS in 24m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a879a6c92b54aa7aedf2eaf1949a27b : SUCCESS in 4m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c09c41a4694a4bd78531b3c1ddded41d : SUCCESS in 10m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/07bff8f454864c108226f82ce986808a : SUCCESS in 9m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9e5c642feb78428f8a873d7c497a467f : SUCCESS in 59m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73c6e7e4dbf3468081f2e3fe1218267c : SUCCESS in 1h 12m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/05f903b810fa45d5bb53c8da8264eb81 : SUCCESS in 1h 53m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e5b0a66f399041c58c0621b177e56635 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":2},{"id":"548a01e1998e843d71c16f4ef0dbf22d61219b35","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-09 06:45:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33aa5e474ca323fbeb61b6f5022bf28bab15ab73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 09:57:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abec3eb00bda4d40b46579c212aa6028 : SUCCESS in 3m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7c9c1e96d59a42fcaeb2c8406f621f66 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24414b114987462097d5f70364044421 : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0856ec4bd1d14a71bfc1797332f1b4c0 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42308323e86945ba876651eafe625251 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46d11c409e784fc5814e2d97e118a09d : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1146ed543874bb9aceab919ef04ec50 : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74855cb6467b4b38b2e4e34485e856ae : SUCCESS in 2m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bf9cf99a1457425499f81b19204d1e34 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e7cdb70c9e6741169566020779eef6ad : SUCCESS in 6m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1f8039e904b4481b8bda083d41182b57 : SUCCESS in 7m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a6dc691c6e8a49dc8a6592893b72feb9 : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8d63b59504204df2a8d7722681131299 : SUCCESS in 5m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1ec57b30a38d4f6e95a5016624ac4b52 : SUCCESS in 4m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5180a3357f6540b58e94cad95c49b234 : FAILURE in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/009ce1b546f84197b7f3b740d42c4570 : FAILURE in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06db8bd58ab341468c917a67e857a694 : SUCCESS in 12m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/027ebd48e63e4305a2da3f128d4e9971 : SUCCESS in 37m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b77827bb8aa24073a291ffa62d2a8dd6 : SUCCESS in 30m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/002454278acb49cda4469401948f5caf : SUCCESS in 21m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee5b0ece87ff419ea4a087cd5fd97837 : SUCCESS in 21m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/804119753ba8457cad824597742d8d3d : SUCCESS in 14m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f4c81f1ad70b414390a869a5d41ab747 : SUCCESS in 16m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d46f379a9fb84061a3d38876fe9192a3 : SUCCESS in 12m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbfd69421b3145589bcacab65169b406 : SUCCESS in 20m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b462694bf9c47b389d387b57420610a : SUCCESS in 5m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8c3dc2c287994c6f8ecaff5e89055f85 : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/17100d849359451f9c0730e3a37e3fff : SUCCESS in 8m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3bcbeee1703d4c1e841b910caac2e7b0 : SUCCESS in 58m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f13f576fb4746e7bb672545b4d2b013 : SUCCESS in 1h 09m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/953e626fc2aa4beaa03900eba9cbf9fc : SUCCESS in 1h 54m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27792602779644bcb777290978119a95 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9322973a91069cf202a8fb9126a26f8a568e497c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-10 14:50:18.000000000","message":"Patch Set 3:\n\nThis is ready to go.\nHorizon fix has landed. https://review.opendev.org/#/c/680631/","accounts_in_message":[],"_revision_number":3},{"id":"4c1149286a0d78a2ff98da6a99ac3d5e4965ce88","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-10 14:50:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d595069468a14753cf76705fdcd2d7488da6a489","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-10 14:54:22.000000000","message":"Patch Set 3:\n\nAh.. https://review.opendev.org/#/c/680914/ supersedes this.","accounts_in_message":[],"_revision_number":3},{"id":"f4dbb545d67b7e140578715257af315550db1a29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:55:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f455c6ca69514be1983b9d3a9b782518 : SUCCESS in 3m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2b7751c94f824235be62caec54f74130 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8790b9a996254e9f990408b67f54773e : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1cbb9b62ac04010a8cd9b723aa51976 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eae8ffa60604fe4bbaf90695909c8c4 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1003165fe0d8419685cf7ed2a56ceb0a : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e03bb2c841f24fe28399ab2f320c8e53 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3353c85525004c988a9e5a9f4e26367d : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/56bc5bbf58474f44861b5b299e3fa39a : SUCCESS in 6m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0016d655bb0040469e47ed0a17f0acce : SUCCESS in 6m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/84f6d00d5ab54bbab8bd33120bfcb873 : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c8458430be541f989f02797b6b22b4d : SUCCESS in 11m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/28066fc91d62427287e37440f9ea6541 : SUCCESS in 5m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2134fe0f8434bf2b584dc7c00e9ec5f : SUCCESS in 4m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2e6fe6b00fbe4e72bb570bd6d54c3699 : SUCCESS in 10m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/371db88fc78b4331a1616874d57c2f73 : SUCCESS in 7m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a4ce91b150944b895051fd1d80057d9 : SUCCESS in 9m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dc0cc4d0c0f9426c8fd09d5849d7e72a : SUCCESS in 33m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de3b7e7d5bb94c86a8ebef03882a30a6 : SUCCESS in 36m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9ed0ac10d6b846b2ab4a7d1f676f3b8a : SUCCESS in 23m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e5f9c118b5234d30b0bb1ec0918479c5 : SUCCESS in 24m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d86ac8f3a1164946810c7521d4f64552 : SUCCESS in 12m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/583c9e98db324e179ebc46e39986a284 : SUCCESS in 17m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/38ed8d95fae94a28b9c62525d4584153 : SUCCESS in 17m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c495a7d19fda417c8cdde64ce8d51cc6 : SUCCESS in 24m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f81fe5b2afb9426eb7819af811cb1ef1 : SUCCESS in 4m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ab0ae06ed7bd4032bff7bb2efa43909f : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f07ec9d25e7e452ebc47579e22a73cd6 : SUCCESS in 9m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/da9e578e59234b6ab8d4631e54fc18c5 : SUCCESS in 58m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc15fc4ee8e14fcdadb6064dca9982ab : SUCCESS in 1h 20m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9688019a83214f8ca32840d577aea2fb : SUCCESS in 1h 38m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/86f1d9369fd14c7f87ecda35375e6fe9 : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":3},{"id":"14af14d8152b1286255e3b8630b25bad6ca1b4ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-18 21:53:33.000000000","message":"Abandoned\n\nwe are at 2.8.2 now","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2373d6434cf046201ff15eb767e8aa1ddf78507c","revisions":{"3f7401e3e2eb5bf413abcfc2ee0baa262eaa036e":{"kind":"REWORK","_number":1,"created":"2019-09-05 16:21:04.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/57/680457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/680457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/680457/1"}}},"commit":{"parents":[{"commit":"80861a41533c44fd8ec4a92e6864997a9779e799","subject":"Merge \"update constraint for futurist to new release 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80861a41533c44fd8ec4a92e6864997a9779e799"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:20:42.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:20:42.000000000","tz":-300},"subject":"update semantic-version to 2.8.1","message":"update semantic-version to 2.8.1\n\nChange-Id: Iae4e8cd7ff8ad35f9db501258377731a585a0b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f7401e3e2eb5bf413abcfc2ee0baa262eaa036e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f7401e3e2eb5bf413abcfc2ee0baa262eaa036e"}]},"branch":"refs/heads/master"},"e83720a06150b95b6ba04adb2ddbdd7228ff1a2f":{"kind":"REWORK","_number":2,"created":"2019-09-06 09:11:19.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/57/680457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/680457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/680457/2"}}},"commit":{"parents":[{"commit":"d39de7a0a4300222e11a8c2e30aab35f95e54d4e","subject":"Merge \"update constraint for octavia-lib to new release 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d39de7a0a4300222e11a8c2e30aab35f95e54d4e"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:20:42.000000000","tz":-300},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-06 09:11:17.000000000","tz":540},"subject":"update semantic-version to 2.8.1","message":"update semantic-version to 2.8.1\n\nChange-Id: Iae4e8cd7ff8ad35f9db501258377731a585a0b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e83720a06150b95b6ba04adb2ddbdd7228ff1a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e83720a06150b95b6ba04adb2ddbdd7228ff1a2f"}]},"branch":"refs/heads/master"},"2373d6434cf046201ff15eb767e8aa1ddf78507c":{"kind":"REWORK","_number":3,"created":"2019-09-09 06:45:05.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/57/680457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/680457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/680457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/680457/3"}}},"commit":{"parents":[{"commit":"b0619f4ff4d60627de68d97e1bec77aed27388f2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0619f4ff4d60627de68d97e1bec77aed27388f2"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-09-05 16:20:42.000000000","tz":-300},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-09 06:44:55.000000000","tz":540},"subject":"update semantic-version to 2.8.1","message":"update semantic-version to 2.8.1\n\nChange-Id: Iae4e8cd7ff8ad35f9db501258377731a585a0b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2373d6434cf046201ff15eb767e8aa1ddf78507c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2373d6434cf046201ff15eb767e8aa1ddf78507c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
