)]}'
{"id":"openstack%2Frequirements~680497","triplet_id":"openstack%2Frequirements~master~I45e2011424c45ce41387164d29f7efd90efdfe63","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I45e2011424c45ce41387164d29f7efd90efdfe63","subject":"Cap requests-mock to 1.6.0","status":"MERGED","created":"2019-09-05 19:29:45.000000000","updated":"2019-09-19 01:04:57.000000000","submitted":"2019-09-19 01:03:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680497-1568855021141-fa3c3f25","meta_rev_id":"2db2dfa4bd0b47fa90a0ea815b34abd786f3a362","_number":680497,"virtual_id_number":680497,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-09-19 01:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-15 19:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-18 20:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-18 20:34:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2019-09-15 06:00:00.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 18:02:48.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-09-15 06:00:00.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-15 19:17:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-18 20:34:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-19 01:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68197bd21f1654b6767e760d9de4281eacfc16a8","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-05 19:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95bf7415104cfa5035a7dc5fb0da5ade5c6d1fc7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 20:16:06.000000000","message":"Patch Set 1:\n\nso this is a temporary thing?  if so, how temporary?  I ask because the freeze is coming up in a couple of days","accounts_in_message":[],"_revision_number":1},{"id":"2a49d52a147407da4d824ca6500ce6cbe3c0a0cb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-05 20:35:42.000000000","message":"Patch Set 1:\n\n\u003e so this is a temporary thing?  if so, how temporary?  I ask because\n \u003e the freeze is coming up in a couple of days\n\nThat\u0027s a good question. I expect fixing the issues to get 1.7.0 unblocked to be relatively trivial, but I haven\u0027t dug into it in depth. I\u0027m also not sure if I\u0027ll get time to do that refactor within the next couple of days.\n\nSo I guess it doesn\u0027t sound all that temporary for Train.","accounts_in_message":[],"_revision_number":1},{"id":"2edead0d5453c96c54bbfea587e575ad84110d9f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-05 20:53:12.000000000","message":"Patch Set 1:\n\n\u003e \u003e so this is a temporary thing?  if so, how temporary?  I ask\n \u003e because\n \u003e \u003e the freeze is coming up in a couple of days\n \u003e \n \u003e That\u0027s a good question. I expect fixing the issues to get 1.7.0\n \u003e unblocked to be relatively trivial, but I haven\u0027t dug into it in\n \u003e depth. I\u0027m also not sure if I\u0027ll get time to do that refactor\n \u003e within the next couple of days.\n \u003e \n \u003e So I guess it doesn\u0027t sound all that temporary for Train.\n\nOpened a bug against ksa to capture the context:\n\nhttps://bugs.launchpad.net/keystoneauth/+bug/1842978","accounts_in_message":[],"_revision_number":1},{"id":"08cacd184ca87355dd27bc346099e423dc3bb26e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-05 20:56:19.000000000","message":"Patch Set 1:\n\nok, in that case (longer term masking), can you mask the bad version in global-requirements as well?","accounts_in_message":[],"_revision_number":1},{"id":"85eb5b84bc70a930d5f750c5a8dd83f127b6b50d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:04:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c819436e39c4cde873a4fb48c3875e8 : SUCCESS in 3m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/99d1335c20d6417eb73f3bd2a9ee24e4 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de75bb111e04e12a66ac39ce033ab11 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bae0736448d448d58a0f4af7a4899e73 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98c82c0a8a9b4f7cb903345a0aaa0808 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95af1dfee90a4d70bfd0e8c48daf095c : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e95fd0898c6e4f08ab80d51fea75968e : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de072b9f13cc42539f91c16dfb93b71c : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e5975ecd26944a0c9133fd205017ce99 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb2c7d2e782c4286bc51ed6f58230e3d : SUCCESS in 9m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9e434ace79484d6a95edc4fbbd132420 : SUCCESS in 10m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef65ec6b7f5946e69e594b3595c5de2c : SUCCESS in 9m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c3db67b8d5a844fd90cbed55983ab27c : SUCCESS in 6m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2433747b0d374d7799bc693097033f0c : SUCCESS in 6m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4bdee7ec2ce3436fb67b3ea3e57136fe : SUCCESS in 10m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0fcf47ebd35b478db912dc5e284fc29f : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a957ebbad13549b4b6978607b4d37f0f : SUCCESS in 8m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/41d876de21d843cb800fdb560f24535b : SUCCESS in 32m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ebdd69bba592497496a6448739e9218b : SUCCESS in 30m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/34c6a66a628347f8ab17951692f56f6b : SUCCESS in 25m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7121d3b9e56644f98da96b7a7387377f : SUCCESS in 23m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4a487e58712d409b8fd10914dae40832 : SUCCESS in 12m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ebdd0675ae204410ab4e588cfdfb0292 : SUCCESS in 13m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d15601b872814ac187827aa634fb497c : SUCCESS in 13m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/abbc649b695b49f7b17e8cf0c30fc9dc : SUCCESS in 24m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2280622367a34a51b698cc1eb80d5776 : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6de5532838944ffe893cb23e369a08d2 : SUCCESS in 8m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00aac8a055d649048b3bcef7dcd74a26 : SUCCESS in 8m 52s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/47ec24d2b9ce4febbc8507f571e299b2 : SUCCESS in 1h 00m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa600cfa04414657997d04eb18884a4d : SUCCESS in 1h 15m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a77c3c35f09d438abcfd50220a4b6163 : SUCCESS in 1h 36m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/44b328343ca94809a7de5e3af3cf2a97 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f617a310f61e2f6c813f62a723797a6873e955a8","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-06 18:02:48.000000000","message":"Patch Set 1: Code-Review-1\n\nplease add the bugreference to the commit message and the exclude to g-r..","accounts_in_message":[],"_revision_number":1},{"id":"ee328d791b08ee14478966d7e8b75efcdf3089c2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-15 06:00:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e257682a7422b5bf8799c7e36acebf70beb2cc0d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-15 06:00:18.000000000","message":"Patch Set 1:\n\n\u003e  add the bugreference to the commit message and the exclude to g-r..\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"c7ba85525f566f54d5ed7260edfeac18ee66965d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 07:54:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f70b96edb134b6695e5b58b8c0940b3 : SUCCESS in 5m 36s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7bfa1c8a52c246f2bcac4e348c906bd1 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b9166cb612f423d836083007a5b3643 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82c8c936eb7a436eade8f889af9f45c6 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac388e2ec43b4481962a293e4da3e479 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0125dfff7452449ea6360413be05b74b : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af0504544bd94e1ea6497c4885145571 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d79ac754f6948709492032cfa3569de : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0a82e8df958948f0ac18a1ebaa095cf7 : SUCCESS in 12m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf5a05cee692417d97e1c0759406d321 : SUCCESS in 7m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a0433811ef7452282f0b343f7b24c19 : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/367c2bb0d83c48d0a40228f84b3a3a10 : SUCCESS in 10m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d217bf36bfc4a4eb53757758409b3e6 : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2f6197a684f4663a6826ede786fadf4 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/af1270d4a2f0408eae5e6d738023d329 : SUCCESS in 12m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9cbb0a09dfa440a082e44348826fbce4 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79f8b54e3c054f459b6ec32167bc9565 : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e3a92b55970b48d383ba8b342b920be0 : SUCCESS in 46m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1efb669dbc644aa48ba7ea4fa9386245 : SUCCESS in 38m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6dee0b9e30ac416bb42386c72442c4fe : SUCCESS in 30m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/07e00be3e80f444dab5f2cccbc2d3563 : SUCCESS in 27m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c03e6daf84d446b9be8ef348f60cd91c : SUCCESS in 12m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49ec1db3aec149a2a69b0bb34bdafcaf : SUCCESS in 12m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5f1efbbf2cd94f7fa09a9b3ef57c4cab : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a02a93f1bc45445d947935490c239df3 : SUCCESS in 19m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7211e046de264277967be731ce1a654a : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7535ccbfcafd4027a069e5c3d0a9fc50 : SUCCESS in 15m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04c84f88bff14b3488838c88584b32c0 : SUCCESS in 10m 05s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6e7d315664c149aab50dfdb27fef2fbc : SUCCESS in 57m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/963246f819184741bc9d764916d6a9fe : SUCCESS in 1h 28m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04c45919c2b34e5988d1f28265118372 : SUCCESS in 1h 45m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f12ce3ecdef7409e861823c22c53ea04 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":2},{"id":"77ff8d956ab244dcec723c48ed1dbcb15edd31c7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-15 19:17:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"75c7b6a6974c423f64ceaea378f2acc911eefd3b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-18 20:34:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a80b2bb7acebd722d5881d37703c24bed21974ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 20:34:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83f0233f7f2827551a0a874f2f7a2d5fc48836cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:51:11.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"499f070bf879bf0be3f080affbeae7a3784cb8a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 01:03:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04412ed7e424f8785a41886aa0378cb : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ee9171f0a84412eb64e7441f42e3389 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e6a94d69e2a4eb0b718eb35b7c2b75c : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f05efef4a68f4b92acb385fe7d6a8671 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aecea1ebf83748d9a76ccd9de6e6d9c2 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b1ed8d2dd3e4c98945f60b06d17fe45 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2656b0a99386402cb1b14fd5ef0d08fa : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ea329a518ae74861ba0dbac30289ac17 : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/26d564133a6b42b3b407fa158305a162 : SUCCESS in 6m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/44035dac960149c4a35181d82c666315 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/32abb8f142e44ff38f600111d2638be4 : SUCCESS in 8m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/55255ddf8c284de6bd2d9ff1c42b047c : SUCCESS in 4m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c2fdb4d3c3b1402db223f737664273c7 : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dd79e327d5704770b0d12918e9f8cba1 : SUCCESS in 11m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b2e28bcbfee41dbb389b4463245351c : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48424af84b5a4346956ae5d57b47c427 : SUCCESS in 6m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c290db61a49e434481fe74ca1e37a2de : SUCCESS in 16m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64bdbc9a49554f8d8261e5a515afabab : SUCCESS in 13m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4c3a901390284412a617c8af00fec497 : SUCCESS in 28m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5aa400814c544c9f96d2c259bd1a275f : SUCCESS in 23m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/06b66130b5da49a790cca590d09216b5 : SUCCESS in 14m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7ff0ca8294b54b6b890d057ce0e3d208 : SUCCESS in 13m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fbbc8541b54948c9ac2bf04c8143f958 : SUCCESS in 25m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4cec22fdabfd42f09c3d210d8b1e953d : SUCCESS in 18m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97a07b6c2bd84dd7b8839714d08ef52a : SUCCESS in 5m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/899f28f9ddeb4fc5ab3291d404ad98f2 : SUCCESS in 9m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/220b8daea6fd40e5a9af22bfac0d9348 : SUCCESS in 8m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/73ff5dab54df46b38a21c099047879ff : SUCCESS in 59m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d13a3feb8f914291a5e346bb10a840f0 : SUCCESS in 1h 15m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50bb80033dc54e5e8cfed7e9c39ae073 : SUCCESS in 1h 49m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b5cd662fe95349608cf31360b5add16b : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":2},{"id":"04c64b43e9c31b9c5fdb7feb27d1ba376ed4c985","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 01:03:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2db2dfa4bd0b47fa90a0ea815b34abd786f3a362","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 01:04:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a134a125f64c7cb55ed62ffd0fcdd1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b7fc4603ab83a31ff973542e9e212b6cffa7248e","revisions":{"fa0417f1c11036fa634f6cc5ed70bb15c14f56f3":{"kind":"REWORK","_number":1,"created":"2019-09-05 19:29:45.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/97/680497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/680497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/680497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/680497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/680497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/680497/1"}}},"commit":{"parents":[{"commit":"84fbbf9ff187f6cbe0d3d2bf897f320cd921cd5f","subject":"Merge \"update constraint for metalsmith to new release 0.15.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84fbbf9ff187f6cbe0d3d2bf897f320cd921cd5f"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-05 19:28:16.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-05 19:29:44.000000000","tz":0},"subject":"Cap requests-mock to 1.6.0","message":"Cap requests-mock to 1.6.0\n\nVersion 1.7.0 includes a fix that is breaking some keystoneauth tests:\n\n  https://github.com/jamielennox/requests-mock/pull/93\n\nCap at 1.6.0 until we can get keystoneauth refactored to use the new\nversion.\n\nChange-Id: I45e2011424c45ce41387164d29f7efd90efdfe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa0417f1c11036fa634f6cc5ed70bb15c14f56f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa0417f1c11036fa634f6cc5ed70bb15c14f56f3"}]},"branch":"refs/heads/master"},"b7fc4603ab83a31ff973542e9e212b6cffa7248e":{"kind":"REWORK","_number":2,"created":"2019-09-15 06:00:00.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/680497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/680497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/680497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/680497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/680497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/680497/2"}}},"commit":{"parents":[{"commit":"84fbbf9ff187f6cbe0d3d2bf897f320cd921cd5f","subject":"Merge \"update constraint for metalsmith to new release 0.15.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84fbbf9ff187f6cbe0d3d2bf897f320cd921cd5f"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-09-05 19:28:16.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-15 05:59:19.000000000","tz":120},"subject":"Cap requests-mock to 1.6.0","message":"Cap requests-mock to 1.6.0\n\nVersion 1.7.0 includes a fix that is breaking some keystoneauth tests:\n\n  https://github.com/jamielennox/requests-mock/pull/93\n\nCap at 1.6.0 until we can get keystoneauth refactored to use the new\nversion.\n\nRelated-Bug: #1842978\nChange-Id: I45e2011424c45ce41387164d29f7efd90efdfe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7fc4603ab83a31ff973542e9e212b6cffa7248e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7fc4603ab83a31ff973542e9e212b6cffa7248e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
