)]}'
{"id":"openstack%2Frequirements~680541","triplet_id":"openstack%2Frequirements~master~If5d65b6267635bcb2b9ae317c7eda46d4db3d7d7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If5d65b6267635bcb2b9ae317c7eda46d4db3d7d7","subject":"update constraint for os-ken to new release 0.4.1","status":"MERGED","created":"2019-09-05 23:48:57.000000000","updated":"2019-09-07 20:00:41.000000000","submitted":"2019-09-07 19:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680541-1567886386666-72f0986f","meta_rev_id":"9d37bc3e7ec8c2debf60289dcbe038a15db7c989","_number":680541,"virtual_id_number":680541,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-07 19:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 02:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 02:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 02:32:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 19:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac18075c3dc50e396f9cf7c6bf13f6107b6d25ca","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-05 23:48:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38f4dcfa92fc5acb61fa5bfd107e0fa791ba7826","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:34:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee158b0c688f430eb7b63c4ade393765 : SUCCESS in 3m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f5fd41cf8d144fb094581e25c11ca95e : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1368f8495a624c6b8a569fcf1cc27edc : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d8a4bcae65f4c34b4b1ccd43d40cb05 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6579109cfd1747e69556600243351034 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/224ef097b58a4266b7beb89a4b81893b : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/119c26bd660142c9901a9dd1bcd5e5ce : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6525f8faba142fdaece4f430031f86b : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3c7e908a70ef4383a1ef958985521473 : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3bf6494b1093429eb715ee04f02c9972 : SUCCESS in 9m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8d5a588e6aa948a8a1ca7ac71c5ef5ab : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1bf36ddbf83349f2bf7bed5c7c580113 : FAILURE in 11m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/31bc2201cc514d8188f0f12be3b32a81 : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d12bfeee60534ce5923d61c31d0042f5 : SUCCESS in 4m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/18d3bbc75a0d46718dbd5e199ec3d42a : SUCCESS in 10m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8958594a3a814862b0e7784488e80ff8 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/008f44f9167d498291a3edd184e0005f : SUCCESS in 7m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7691837139034430b19a416954f4b897 : SUCCESS in 33m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/81754f4a03ef4174b15828defc9a6d78 : SUCCESS in 30m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ebc33e3afe054429bb0cec6947e34a5f : SUCCESS in 22m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/324e516f295c4c5e89789f63c20bc3d7 : SUCCESS in 25m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bd78fd4d2d41437791bd8eea5689b476 : SUCCESS in 11m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7816f90a5269449d968c86d02d1e429b : SUCCESS in 12m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/849dec37b5af40908cdb174f4b4155f0 : SUCCESS in 16m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/563b3e6fc4794cf48342ed4b41c028b8 : SUCCESS in 20m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/154eda3a5055470eb88db6afcd5ed82f : SUCCESS in 6m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6bec9451c93e4116a0e0c2f39e0e9881 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f874959e1b54b1ebe09c2b02fb88afc : SUCCESS in 8m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/24572f42627942b1b1e4598f3a20e62a : SUCCESS in 1h 00m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bc87a9023314354bafb60bac1391f27 : SUCCESS in 1h 10m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8ecd0d0fe7e34c96877f4ca611e754aa : SUCCESS in 1h 42m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b9e02c287d8c4b77bd233b09079a2491 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"86513aa90d2a2ffb81f6e3543ad7ec61d1bb07b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 21:23:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff1dc54b77ca101d4677e7c0956aa33e3114117e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:26:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e91f94498ba4e7fbf14306f578d1782 : SUCCESS in 3m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/11705d4f38634b15ae01c2380c502e95 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b64462390d64e84a2095fa791490248 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f2f57b0ef054187b418e51fa98dede2 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc6e5f9826d4edcb75be790fb7228de : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6725ef910ab45ed856b22a6683996e5 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/095eaa5614434c9ebe0e3f15cb01be39 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cda0b5ac75647228229a044f7f6451b : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/13905083279b4bcca56b099e90629cd9 : SUCCESS in 7m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e244aef431e443b79e5437c71354fe2d : SUCCESS in 7m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/353a0509bc5d4673aba315e94be3fb23 : SUCCESS in 8m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/296a7270bf204140b384d82df24184ca : SUCCESS in 7m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/804f18a0349449ef8f8c2b30c37e05cd : SUCCESS in 5m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/deb1f775cbbf47eda47f0e222cb5ea0b : SUCCESS in 3m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3806c681d64b486085a3550c239d7369 : SUCCESS in 9m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b1f9005f358146198a300e9020af84e7 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33ca195811b74eb59897bcb4ba83d90c : SUCCESS in 6m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b06f9518883644d08783b1c3c3cc7d68 : FAILURE in 33m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90e6ff8a1cab4baf85e236174ba05b8e : SUCCESS in 39m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/06a0098c772b46e89afc882221f5296c : SUCCESS in 22m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5b9786b731db4a4e885cbeb15034be9c : SUCCESS in 23m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7ea8b70fb49e47be835fd8efd4ed5e65 : SUCCESS in 13m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c030bf9aaf584234b656ee7efbab4ad5 : SUCCESS in 11m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/16ce2cb4e7a04efc847706d06b05929c : SUCCESS in 12m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce25adc219ba4b22acc29b01cccf417e : SUCCESS in 19m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a859ef7bb8148ada91130af7a16e5b7 : SUCCESS in 5m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f794d5adc48b4cb7a0026ba7113e430e : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ec9b9a2423b41fe9e270bb0599abd79 : SUCCESS in 9m 29s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a4f0516f0a51463b925582ec964f1e2a : SUCCESS in 55m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/917afd3ccf5547c1a895e7d0d74ed285 : SUCCESS in 1h 09m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c32d91bc4f064bcd9331abbf1200aff1 : SUCCESS in 1h 55m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/064cb4bf1cd245d6972045317ead1d22 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"491a8481b657d095961398c65248269e9c4a1099","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 02:32:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"899fdbb480a34fd1378e162fd1db264f0a05975e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 02:32:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d8571e23d7569f4d22f5f57e998ea3f735afeb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 04:12:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3853d43fb90f4349b7d0980aff2cd8e5 : SUCCESS in 3m 39s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d8907055228f47f2a85e331d019f785c : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10d0b77aac942f6940715e913e8e395 : SUCCESS in 2m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dac5c567f9b4b54a104aa72eaaf4233 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6244d4dc516e4cc19f644a293f373f2b : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65b440442add44f3936832a84bc3b85e : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69c3b5d2978242daa1f8203ea00dc678 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf56325c2f214e8d9c139d5eea1280e0 : SUCCESS in 2m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/35b9b640eb3e4f3ca6c05cc3c9914434 : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a5ce57645cb4fd492dada70a3c5c4bf : SUCCESS in 6m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/38b6440de1914899921e302e8b9d7250 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f1a6086840e44d6c952f57afbbc492c9 : SUCCESS in 7m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/56c42dc08395468fb8dd64d5387234fc : SUCCESS in 4m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6582b1a38abb4bbd80d55fef11e1124a : SUCCESS in 3m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c84bbd3c59dd4467b887f0a8a774e2a7 : SUCCESS in 8m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89f162616a0b410e850b41cf0d32246f : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86b7b70686684b17aa094e672201b095 : SUCCESS in 7m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d85ed297dcaf47e88188bce031a65454 : SUCCESS in 30m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/38437300927640d5ae08c4f01a01eb3e : SUCCESS in 30m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a323c64855914632bb3f90d08dfb937a : SUCCESS in 22m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71acdc16ad624aa4a7300d3d329c4988 : SUCCESS in 22m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c12d2254b7c9490985ff25ab1da7c92e : SUCCESS in 13m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d5c7c1ac9f924597885619d880b54c71 : SUCCESS in 14m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3c994fe7342e48629ac3b32e9835647a : SUCCESS in 10m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b184600380db42c3a019efef0e2fde3f : SUCCESS in 19m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be42fd51caa94a349fe7af5a9c7e96ce : SUCCESS in 4m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/61c7eb2fe85d4d13b6b7a5aedf7ec880 : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c39ebfd032a54cc08255b15d7aa00bcf : SUCCESS in 8m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/6b9110e8be29405d9e097177234f2c60 : SUCCESS in 52m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a8fee949fa5489795ad597967541080 : SUCCESS in 1h 06m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dd7feac0b4d549429427a78894baab3d : SUCCESS in 1h 34m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6fc521be8bfe41f2a3bcc05722ba160a : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"eb86b850418efffd60a15da37b532f3e9318ef82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 04:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"675a7e0540bbb8f4691e9c9dacc540fdba8240dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 06:36:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc512e983f24afca89c692f86ed52f2 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f07885f7e7c44dd9b215a38455bd5f9 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea1b374369ad4fe5af53a685bfc0de5b : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b45d6641fd4f7aad4ff787ce5317bd : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f043431352fd45ab81c017b431243834 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1546f753a7934549a02d4ca9b0cb4744 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d09aabd21274b7eafb6f1425310bac5 : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1d0e82ff325c4206bd41dbf582abb1db : SUCCESS in 9m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/038a532a693841a7870cef96e7db36d5 : SUCCESS in 6m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6c26f958fc9b484c9ccf2c5f9e9835c2 : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3beae431a184aa088c4787e514ff7fa : SUCCESS in 9m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6fb885ec227146019e21d54537261520 : SUCCESS in 4m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/def5d23f67154c6cbe0b43ac781d53f4 : SUCCESS in 4m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c38ceeb8681346cb8bc5ab451d82f16e : SUCCESS in 14m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b342fbbcf0314efeb11947244e646c0e : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9e3dafd2e4c459babe1228c18437f98 : SUCCESS in 8m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/459f0246605943fba06acafed3564aeb : SUCCESS in 30m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c2899bdfb534200951b9d0c0eb99aba : SUCCESS in 28m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/df8c08c46015496b9c82d8764838e805 : TIMED_OUT in 40m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8e565b466034199a034b1b6d812da65 : SUCCESS in 20m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b74e4a6aa2184889987aad82645813a8 : SUCCESS in 30m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8631bff988bd407b94ce5225360297d5 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5807cf4d6674b30a21bfab1895325a7 : SUCCESS in 12m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50ab10d48c2341ecb89d49ed68f34c92 : SUCCESS in 38m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81cc1ad20d494bc7a47b1c69829ea1cc : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e66aba8cda7e4bb984343f31fb3c6b17 : SUCCESS in 13m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/99ba2d6c6a12438c802e2253641b6443 : SUCCESS in 9m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/5b09585ee9dd4eb981136ab6fdebb631 : SUCCESS in 1h 02m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b78ac85aef954ae98ae9ea04b4a093a8 : SUCCESS in 1h 13m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f500769208db4f049e7b0464033cf2a5 : SUCCESS in 1h 39m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2b4b43b4faff41a095db0695cead104b : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b200821cce87ca8548427c8f7255c17cce46777b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 16:44:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d781c65be98901b31dca1b6c42b08fac70293b19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 18:20:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5bd05d980e47e38bf3fa073408139a : SUCCESS in 3m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/586c2486853c4ec2ae1dd50c7ea60454 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dbaf566494e4d76983873dde72fd3f1 : SUCCESS in 2m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a17aaa1db2bc44d8b072c6ea6f4d89b8 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b8737df57ed4465a7466c536367890d : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df7ac0bac8bb4f75967eb4413966608c : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83e139dab2874055aa123aa2e4c4e766 : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a95f157f33ca4f849debd52d886b1a8c : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5f639df88b1c42e5b44afce455b9c79a : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4baef7d531a34457aa235ffd29b1c99e : SUCCESS in 6m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cb1fd38f841e4b7d8047a0d86e495ac6 : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0d439b01b8404ed39802f978266c840e : SUCCESS in 7m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/30757e7e8b644d43bf46db6127cf7e8d : SUCCESS in 3m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3add146a41141ac88e644e11904a1cd : SUCCESS in 3m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/edcc65b504174fef9f81d047c7c40244 : SUCCESS in 10m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/11717ff90e3e4c498653a70421bb4d54 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e8ebc6c8c61493298f87a677fca9c50 : SUCCESS in 7m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d2e9dbb814fe47708c3ac7ee19163cb0 : SUCCESS in 25m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3a5690944b774b6184e9626ed61c03d6 : SUCCESS in 22m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cd52011669f74ffca7a387b4dfb4ccce : SUCCESS in 24m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/46db2c2f6d5d4e788fd947e595ee667d : SUCCESS in 18m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c2a39fa6def1411487d0bfde74e6c16d : SUCCESS in 15m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f5c6addd1d34698bb8dc6b0c90abb5a : SUCCESS in 13m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2e0d45e60a42417ca25643064f421726 : SUCCESS in 11m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d31c8a566ae24450bf5853fcb799b85b : SUCCESS in 20m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02fe88066ce14d14aeeb3b54306b2070 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fedadaca934342cba7246fb8c1ace252 : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72b83c3da0a74946a1fc24cced2a8734 : SUCCESS in 8m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8668f7d08e044b2e9978c3f23f02c4e2 : SUCCESS in 56m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8977dc19dab84708af38433b4ac86069 : SUCCESS in 1h 07m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/44bdd702453a4075be4357036813e84c : SUCCESS in 1h 28m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b54a3d69185e4085847a0f93e4c6c984 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"811e82534bb907810544c950067639836c958653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 18:20:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f2217a7afc46235d97f7fcf4adc8b79eae2d8c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 19:59:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01c0ae5a2154f34afdb6e6a5f33d7a4 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943b85e415c240de940b4814fde943d3 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d45362cb0b9b462494ae70adfe23663e : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af26b63f4f2b437d877844b7e0c4f191 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed280c849cdf45ac80c48fe7687f0e56 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eda90e4f10dc4f199c920b31d12e7e9a : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f047c005f424d25aa69968cf6c126ce : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ed66cb3122124796b4d8248826ff9999 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/159b33143e3f400a9ce2db2a98fe4aa2 : SUCCESS in 6m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7561ebe872f94ac78ffddc087a3f037a : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d8b4fe9f7374365ac1a18429987dbcc : SUCCESS in 7m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/78c596462fbe4552897b94a3a786b2fa : SUCCESS in 3m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8910787bdbcd43a48096b7adcef81d61 : SUCCESS in 4m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/11c4eeb18f9f414b9c575eaebebf1bcb : SUCCESS in 8m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/676ef3071f9644bcb6d189dfefacc8cf : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cdfeb4f202f74466815470bde336a900 : SUCCESS in 5m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d2961ad96e5a4d9388889f131bfb3d9f : SUCCESS in 32m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6cbddaf3d4f045ae87cdcb493259eaaa : SUCCESS in 27m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bc1a96d05e5448dba34102e3b334a834 : SUCCESS in 22m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be52b2027336440aa5223d08e9aafb46 : SUCCESS in 25m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6f5ba1ad868a434aaf6d221e37b2ed9b : SUCCESS in 13m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68a8fa8da3c748d29f634252a016b16d : SUCCESS in 12m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ad49c00cd5374e42b6e3b649524f5e3e : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/135801831417447f82d6f0175c161568 : SUCCESS in 20m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf6325c5256f4cd89f615c48f76bb59e : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f2416c0233f4cf68dd8aa1df8244ac0 : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c427faa2aa042bf8af89d38a7e29281 : SUCCESS in 8m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/97dfc88a66ef478994042e0fe125b63b : SUCCESS in 55m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e34ec9beb4e4f6c9e231bdb9157a4a5 : SUCCESS in 1h 04m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/623d0725102344bd833e7ff3bfa1183a : SUCCESS in 1h 36m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/efec98bcb60b448496f44a0cb199df99 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e4bf722ccb55f41d8d48f560b30a84f3ed096f35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 19:59:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7313bc828c38ab0fc79278e673f42ddda7be3792","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 20:00:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0463dd9395d44b209b950a21a1aad9a1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f504bf169558aa07a036a5d3c9f6d6acf280672","revisions":{"0f504bf169558aa07a036a5d3c9f6d6acf280672":{"kind":"REWORK","_number":1,"created":"2019-09-05 23:48:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/680541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/680541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/680541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/680541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/680541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/680541/1"}}},"commit":{"parents":[{"commit":"fd7e2ca65ca276ee9b85c1ada0ba8413cdea25a9","subject":"Merge \"update constraint for python-zunclient to new release 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd7e2ca65ca276ee9b85c1ada0ba8413cdea25a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 23:48:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-05 23:48:56.000000000","tz":0},"subject":"update constraint for os-ken to new release 0.4.1","message":"update constraint for os-ken to new release 0.4.1\n\nChange-Id: If5d65b6267635bcb2b9ae317c7eda46d4db3d7d7\nmeta:version: 0.4.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I0a939b7278bb93d4f2bdfcefe778956c36f44bfb\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f504bf169558aa07a036a5d3c9f6d6acf280672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f504bf169558aa07a036a5d3c9f6d6acf280672"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
