)]}'
{"id":"openstack%2Frequirements~680549","triplet_id":"openstack%2Frequirements~master~If75e320477a7f53ee723478059ac002bb53df226","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If75e320477a7f53ee723478059ac002bb53df226","subject":"update constraint for monasca-statsd to new release 1.12.0","status":"MERGED","created":"2019-09-06 00:05:20.000000000","updated":"2019-09-06 22:55:48.000000000","submitted":"2019-09-06 22:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680549-1567810433913-72f0986f","meta_rev_id":"88408ae6baa1a439360f396c9197a204ebfef39e","_number":680549,"virtual_id_number":680549,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-09-06 22:53:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-06 19:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-06 18:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-06 19:27:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 18:03:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-09-06 19:27:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-06 22:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0418be047f3dd7af00ecb23e7ee99b7565a8517e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-06 00:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c9e4ea7ca086419b4be7a873f81d0b79b2ce2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 16:54:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f74a8c4166c4b6e875f73b01c067845 : SUCCESS in 4m 19s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/01c38316094f4c8d885b2cf3061f92b7 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a6b6ba668f48219f30153462035c37 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c2b3cb1d2ad46b29b53170e33f7020a : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cab4d3ea096f404ba3badb8dc63b5832 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef0fa75b3fdc4f24814e7747df12360c : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3679dfd7597442a9a8f4f02485966161 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e02b8760d06b46dd8b094f93df6a5757 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c7f702881a9b40758eba6c9c697a3e97 : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f8b9b200376b43e68a0f3af524e8006f : SUCCESS in 5m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d3eed76938794eefb1f87057f199628e : SUCCESS in 10m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db302320295341799649447ad7dd8c1e : SUCCESS in 10m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3a1f4514ae604efc905bf80fce808188 : SUCCESS in 5m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/605403cd16ce490b8019eb892bbce45f : SUCCESS in 5m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a3d82a26e68449e1afaf13444d3ca449 : SUCCESS in 11m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0770f3ed54c74094bee2769d1f33ffbc : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a43600c82e7b4c1ba44ddbeec850d174 : SUCCESS in 9m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8cb63c7db8464f28a20a10ef22387735 : SUCCESS in 36m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4412f563939461584e51a801fc03365 : SUCCESS in 34m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d1fc1ab43ba44ade963479ac3dec42f1 : SUCCESS in 22m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b6103463ab64a8b9cf1584ebf91c2eb : SUCCESS in 26m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/62ac7d3c2d78463ab2e5cd8c0299cc47 : SUCCESS in 13m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a951d192d13c4edebe9cac057f34201b : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c7301d3f86184e4ea145444ce979e556 : SUCCESS in 15m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1bb3f1a424e34c0887d0e5324799370b : SUCCESS in 21m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/106e5955db6a4bdebe1b2783c7132c74 : SUCCESS in 5m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/167ee1ae3d29479e98526d67ce634c07 : SUCCESS in 9m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7bea4e542adb42b78ce2c99bd1b5a08f : SUCCESS in 12m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/44da4174bf684b7a8ca67bb3e29fe9a6 : SUCCESS in 1h 12m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c944125ff4c4045aa3bff2b84b8af22 : SUCCESS in 1h 12m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a3128e929d64a06a6db36933162df06 : SUCCESS in 1h 46m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/28eb228dc47c4b5f879f2eeb17c237e3 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1d4a56d4266770120095fabc9939e603926cf737","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-06 18:03:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c29334ef840a3854b306e0039867f2703983e2b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 19:27:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73a994d3c71efe8d08839778faecf7f10836eedc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 19:27:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e72e4fb646f69661622a33aef5e1bcd0f52e15d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:53:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f3e1e852df2ef225695013ecaf5f89d02c916ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:53:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64dfefaf78df41ffbe5c75889f8fc10d : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e2d73a0d4a4bc5a89179261d3d9b89 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b43d44b176d74b52805ba2d50b537b3b : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2982ac7c8d4b1f9f19c946a883f8cc : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ac66441f9d949eeae443a60dfb1dd6a : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4c082d2f24d4b0f9cc54d3bbe70c9d6 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d778336a949b477e81f3b1871d9860d0 : SUCCESS in 4m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aa5f4a01739345ea99b3f55ca7648875 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/146d140f1e914fef935529be6c0e826f : SUCCESS in 9m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/05e991bfb95049eeb5bbf4ac4693e2ad : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/21afd90c39b7418aabe2b4b7affa4f7b : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c51d04ff506f4dc2bb9c8313871cc215 : SUCCESS in 3m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7afa2168d3f44074a921de576a0a4fa5 : SUCCESS in 4m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bb2c8bc375e645b299c487a9f7978978 : SUCCESS in 10m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ee412e96d8f4988b38a08041a5b8a89 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95e907bc412440608cd2cedc86324c09 : SUCCESS in 7m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fa56ef2c42804d81877f14471510752e : SUCCESS in 37m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/51bef508ad38431eaf0c6b59bf52c5a2 : SUCCESS in 33m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/924ebcd67b254c7ba53b1093ae19a1cb : SUCCESS in 25m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f7e233fda4a4d9b8535656776abe260 : SUCCESS in 21m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/14e355f8deaa437ca56aa9bc512ae049 : SUCCESS in 14m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e4cbe417bfd04e8a8a240050366fecd3 : SUCCESS in 13m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3624a66fbf234ffb816f443cffaf13e9 : SUCCESS in 17m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/439ed0585b074b59841c3d55241ce8a5 : SUCCESS in 17m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6680d0e3ef7f498899bd64cd330d0b1c : SUCCESS in 4m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c8182b96c5994640b67216903c0c480a : SUCCESS in 11m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c6ec847dbe746dfb9acf458eb11f6d0 : SUCCESS in 9m 12s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/134621d09f804a82a99750591fb086fe : SUCCESS in 1h 00m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ece780ac13f64c9db1befa9480680eb5 : SUCCESS in 1h 09m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/58b3a0cffccb438e9303ca4129345d16 : SUCCESS in 1h 38m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e1d5599e53b24f9eb1533eaae9f14fbc : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"163ff044e4821c581a5966acd21eafe8894fa018","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:55:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179ad01e5ab74ad5966424ab4d074da4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f88cb1b686af0d265d9d42bf4785eded2d2fcc4","revisions":{"1f88cb1b686af0d265d9d42bf4785eded2d2fcc4":{"kind":"REWORK","_number":1,"created":"2019-09-06 00:05:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/680549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/680549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/680549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/680549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/680549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/680549/1"}}},"commit":{"parents":[{"commit":"fd7e2ca65ca276ee9b85c1ada0ba8413cdea25a9","subject":"Merge \"update constraint for python-zunclient to new release 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd7e2ca65ca276ee9b85c1ada0ba8413cdea25a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 00:05:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 00:05:14.000000000","tz":0},"subject":"update constraint for monasca-statsd to new release 1.12.0","message":"update constraint for monasca-statsd to new release 1.12.0\n\nChange-Id: If75e320477a7f53ee723478059ac002bb53df226\nmeta:version: 1.12.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I395f3148b051eb2bd0a7bfafab6db0e34a05614c\nmeta:release:Code-Review+1: Witold Bedyk \u003cwitold.bedyk@suse.com\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f88cb1b686af0d265d9d42bf4785eded2d2fcc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f88cb1b686af0d265d9d42bf4785eded2d2fcc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
