)]}'
{"id":"openstack%2Frequirements~680552","triplet_id":"openstack%2Frequirements~master~I9fabcb9ecba95a2d94994d31ae4e9e37c58c3dd2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9fabcb9ecba95a2d94994d31ae4e9e37c58c3dd2","subject":"update constraint for sushy to new release 2.0.0","status":"MERGED","created":"2019-09-06 00:08:22.000000000","updated":"2019-09-06 22:57:00.000000000","submitted":"2019-09-06 22:53:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680552-1567810435208-b79a56ac","meta_rev_id":"8c6ec29977fff1be11e7a434acd08e6d754deefb","_number":680552,"virtual_id_number":680552,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-09-06 22:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-06 19:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-06 18:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-06 19:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 18:03:17.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-09-06 19:27:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-06 22:53:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bebd2d348ee1d875d9674626a56c31375f36e70","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-06 00:08:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db495dcc23056069552f2e1b62b477e143bdbe26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 17:33:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54ee73840304846a93dd12da61359c8 : SUCCESS in 3m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5a1203bda2204301838fc2c361bbed3d : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7e94f6582a45f692e7f1af433dcdd4 : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26a6bf39658a49069cf4a088a120ed40 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64e4e2887e51445087c8f4d09f4a50c9 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fb3aedc314945eb93621ff0ea85fbdf : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a41586dd97fb409f96c8c7661e51cf8b : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/daab139201844e7a80595922fd1c130e : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/721f33dd26054a239b9dddde7da8bedc : SUCCESS in 7m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/22a47da7d4164b75b771e290f92240b4 : SUCCESS in 6m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/72e5df7efc18441e9c23d95314afca01 : SUCCESS in 10m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7bc9330a54549afa816a02cc79df125 : SUCCESS in 8m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7f33b40b713f433c89b15e039c116bec : SUCCESS in 4m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ea05b255c474658940b499c81800658 : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8024e9992aa64bd6bcbec01721eaacf9 : SUCCESS in 13m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db5451ff38cd49b4b8f5474915b64aa2 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9a76ac6644c244569a51fa008abe3204 : SUCCESS in 7m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/248d1e1a4bb242f39a2f1b11e33b8338 : SUCCESS in 29m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90fa56f338be477fa706be5a189791e9 : SUCCESS in 36m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8aa31fe03d5a4a1587e12d0691f9f411 : SUCCESS in 25m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1568bfceb3504fe8a230c3f520ad3808 : SUCCESS in 22m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5a33a1700c98494291d9648376ce4df0 : SUCCESS in 17m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4003dd2939c5499cb9825cf0d918994b : SUCCESS in 10m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f37de13104b044a1a3c548d60b0fe83f : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48e6664032a74cacb9a4de121a649d11 : SUCCESS in 19m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60094e620c114809bf92caa7d1339bc7 : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a421d41771c34b5b9799b40e19980f72 : SUCCESS in 8m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65e8ef2142d7410c947169c4f2cf911b : SUCCESS in 9m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/93b26ca3c01d4bed988aa43d16c1c4b9 : SUCCESS in 58m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b20458bcbf194ba5a73ebd6e0807f837 : SUCCESS in 1h 14m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/add71fb37e284816b6381ec1633c060f : SUCCESS in 1h 59m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/69dbd6fc29934f2dbff5c4407baf1209 : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":1},{"id":"60fb01cffec7182d3ad9ab5ec274ba18d8c00f48","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-09-06 18:03:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56947fd4ca99e7e447255ea5691abcf93a67b40a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-06 19:27:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d8277bf890a348765d377e51ad1338f324f5016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 19:27:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d48f06834cd0f745c285f8f6272a7bde34489b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:53:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d511df18ae034db8a29b2e964256e385 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf024d925b654329af04c71f545bfaef : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a37dc74a23224105932c19f9ec74185a : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef3453c5b0f74403910a3247b2890669 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e99ec447a61146c795a10209c5a34fa8 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ddd7dffeda84e50b9a283708973d6c2 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cef909abc724c6bab0942e8979d4cad : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d689e0393c47417488ae20f2866da111 : SUCCESS in 9m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59e9940a7a164cc6b06151607a3b5154 : SUCCESS in 6m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/501fe8a834ca44d888e792ea1cd404cd : SUCCESS in 11m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0e5827cbf2b749d7b3499668dcb47c04 : SUCCESS in 8m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c0faf5346f6c427494bcb1d4786acb5d : SUCCESS in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14095e7361d449e494e9ca5a2e277932 : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b8732f989cea4a57b675f629a96b6869 : SUCCESS in 12m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f21a2c0c64347ff90cfcc088e8256be : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa74aa57f8f0495faf85d7c72d50b28b : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/48951b50d6f94baf8f5033a3597e98d6 : SUCCESS in 31m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb600c4a92b446739cfa8d607a300dad : SUCCESS in 32m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c52650309ade4660b794e4aed7a14238 : SUCCESS in 20m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe549920f7974afb8de87b65bc6324a8 : SUCCESS in 25m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3d81fa0a0adb4b528c58c51f413ffb44 : SUCCESS in 11m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0ba88c132f77427e9fa6235c25c98c7e : SUCCESS in 14m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bd2ded534c394eefbaffddbe9c527bd5 : SUCCESS in 12m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dcbb08740e534363aa177121bbae09aa : SUCCESS in 21m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de56a8fbf7fc4db4a59f33b84d365ca5 : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96c737b507b84f6f94c8aba65ddf8e2f : SUCCESS in 8m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed9c285f1cfb47a79c34f535cf806fc9 : SUCCESS in 8m 29s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a3859160c8de44bf8f916b40e5c27877 : SUCCESS in 1h 02m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31cbc97d257f48f6b98e72094dc1a7e3 : SUCCESS in 1h 02m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/540468b0fbfc45f98dbe33d5d8ca58cc : SUCCESS in 1h 27m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fc2c96210eea4f76a3f494fb670852b7 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4af1a99281cc6528c41ae73d6bd7b8f2a14bf7c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:53:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b24a27a92156faf0498232c8c982f226a1e9f301","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:57:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de92048c6b7e4320a6415bc40c203156 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81cffb6b3fb3926fd437e567dbdfc29daacfc402","revisions":{"81cffb6b3fb3926fd437e567dbdfc29daacfc402":{"kind":"REWORK","_number":1,"created":"2019-09-06 00:08:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/680552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/680552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/680552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/680552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/680552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/680552/1"}}},"commit":{"parents":[{"commit":"fd7e2ca65ca276ee9b85c1ada0ba8413cdea25a9","subject":"Merge \"update constraint for python-zunclient to new release 3.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd7e2ca65ca276ee9b85c1ada0ba8413cdea25a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 00:08:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 00:08:21.000000000","tz":0},"subject":"update constraint for sushy to new release 2.0.0","message":"update constraint for sushy to new release 2.0.0\n\nChange-Id: I9fabcb9ecba95a2d94994d31ae4e9e37c58c3dd2\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: I133050f292a4191c2264428e4d140f6525cbaefe\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81cffb6b3fb3926fd437e567dbdfc29daacfc402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81cffb6b3fb3926fd437e567dbdfc29daacfc402"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
