)]}'
{"id":"openstack%2Frequirements~680746","triplet_id":"openstack%2Frequirements~master~I420816eabde41a71e47810072678e5072fe63c0d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I420816eabde41a71e47810072678e5072fe63c0d","subject":"update constraint for keystonemiddleware to new release 7.0.1","status":"MERGED","created":"2019-09-06 17:09:10.000000000","updated":"2019-09-07 08:57:28.000000000","submitted":"2019-09-07 08:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680746-1567846515760-72eb98b0","meta_rev_id":"06df3aaf0a3a48a7c6b159b9905e9a4c61c62f6c","_number":680746,"virtual_id_number":680746,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-07 08:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 04:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 04:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 04:24:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 08:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffdb15f4dff3dda24a7016088d022aaeb1dbde5d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-06 17:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49533775b6867c585eb3eb0b3ff9d01dfb918c5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:34:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6980f87824d438aa2841e0b6ec9ea9b : SUCCESS in 4m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f5354dd32b824844888188890cb8c5cb : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f47c9e120394762a3f90d87cbacc61a : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75fdc79e81814bc4b3a44b7f83ab046d : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b73bf1b57149e2b728acf153277394 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50542d696df34c519ba20ac4e8b1a474 : SUCCESS in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1665d12660aa477d9e8f4c125edf1371 : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2aa5b005647e4c2baa510dfc6161b560 : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2e2d387b628f4412bba5c4915ef137ab : SUCCESS in 11m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/298d7a0e431348eb915d876f9375c546 : SUCCESS in 5m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0a4a9275eccf434b9f2016eb118248fc : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/067c7fa44b0e44e7b8ad109bc26aee89 : SUCCESS in 9m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4ab6b384e0b14993b9d015657828fa9a : SUCCESS in 6m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bcbebfd0619447f29a1ff260211b7a1b : SUCCESS in 6m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/568d68f6dc5b4105bee28588f6eaebae : SUCCESS in 10m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54fb3f120b8a4455945122f2ee127704 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62cd63244a0d4b43953f47e31200c5b6 : SUCCESS in 6m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8f0a0d1f2bae47839bb701e1b39f7e5b : TIMED_OUT in 40m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a090e1623e6248f789b88d5d1510c32d : SUCCESS in 29m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/79dd7e418e8e4fe18c9838da4cf70f04 : SUCCESS in 20m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8ffbca495774928948960ac65f85fe4 : SUCCESS in 20m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ea5f6d4c903e466286271e4a75632122 : SUCCESS in 14m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97d7db3cacc54014a4025f361fb56c73 : SUCCESS in 14m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d85968ad5fd64130a93c3ea330682bbd : SUCCESS in 13m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2d4488087344e44b91e8d70a60a180c : SUCCESS in 18m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c54e7ff5ace74085be2f8ebe6b496905 : SUCCESS in 5m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a7dfcc1a1b8a49018f48bfc685128bdb : SUCCESS in 7m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22c27d04d00d4860aa3f3a78ed30d988 : SUCCESS in 9m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f0a91cc05e304b3e97f6bff3b1c36f09 : SUCCESS in 57m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e4f3e840b824ac5a606212262424ab8 : SUCCESS in 1h 16m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e81102c7e5d43e6babe603b6b3814bd : SUCCESS in 1h 38m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b44e9c741943439b81b7b181452a979e : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c9ef31311d49ee37681ce89e5f8337625fb40c14","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 04:24:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"500c5bf2fec9f6ce04c0e53503d67949c9c49fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 06:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10339c7a49b4afba4bbe8c8e81329ab : SUCCESS in 4m 26s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7aa362b165fd4db0a52c12436ff30538 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14befa5d001d483dbbacb534bdbe65bc : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ba5e975776141f58df4ae15109d9221 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e10035e6adb4432aa74efa2f8e7272de : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a06ccf579e1a49ada79fa67bd0a15210 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d05d9fc23e94481a4751676741aa3f2 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d64b75fbbdea46be9486a1cf5f8f2408 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7f74aa1bf1674e26bcbee43952b97808 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc26f9e88ff944a2823f380b7697baa7 : SUCCESS in 6m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/420dc25f690b4b84ac9b9614e3559006 : SUCCESS in 9m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/694acb6deede41278e85f233df4e4df2 : SUCCESS in 10m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/89793343342e488cb7137a05f3d61799 : SUCCESS in 4m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/273bb33ebb5941abb5a112df8f36bd02 : SUCCESS in 6m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/33522d8d3c474c3b9f2b48763afeb320 : SUCCESS in 12m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/51062183daef407b8000681d9f7db3b5 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aba7ea914d034c208041ed495b19dd30 : SUCCESS in 6m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a491d2a2104d41f4a901d8073bdb25f1 : SUCCESS in 31m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5f4690fa30d14ed183b1546cf1e31da9 : SUCCESS in 40m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/64e9057814424a67b4952c661ca1f706 : SUCCESS in 31m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/032b4c99425c420298fae21fd9dc856c : SUCCESS in 28m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bff0a824871e4ca99fdc391c379ed6c2 : SUCCESS in 15m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/110a0729ac5a4dc6b1aaf9cf1ad9a1e2 : SUCCESS in 17m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/90ec28a3209c48d6949f77d62cf7373b : SUCCESS in 19m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3718671eb6947c8a09f76126dd7e97c : SUCCESS in 20m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5d91bc4fb90467c9ce57303aaf49392 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/130b3df282eb439d9414f5866b11eb85 : SUCCESS in 12m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5abe56d454ea407aa0d8b85be64aef30 : SUCCESS in 8m 57s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/45fbaa852c8348cd9f0ac7fcba134b7f : SUCCESS in 1h 05m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60298245e2734ae682cab0dc704cb4f6 : SUCCESS in 1h 13m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4c3b6402cfe14f8f9f7cb31d954981ea : SUCCESS in 1h 44m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4ea326f4e7294f54a4c06f3b5b5e1945 : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":1},{"id":"70030da3ec5b2e057991f5e9a98fda3a3d465ee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 06:16:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc3707c4eaf5d571fcd1ef3b7af709a531698e53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 08:55:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee3129ccc47f79d98ee460e9dd3d896650322b0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 08:55:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25fcc40f45542348300107492ac8e34 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812f684ccb824f458866aa977e824a34 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/344ce1b132854f9ab927e3313376506c : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad65c8bb1e454d7188969de94d65fdc9 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb2017b7ad954a4b8f4aaa2ad3f5fcbc : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e67a5abc9504c1daad16d3de8c437b2 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/127885a08fc94594a2fbd8c7ea39b4ca : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2db0fe3107b3478cb7583c95fc205ca3 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae03e4b249b747a49abece96fed65a2d : SUCCESS in 6m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a5fa58e2ab4e4f689c91de01b8c6e078 : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/50b0530db4694f8ea5e0cb52350e1445 : SUCCESS in 9m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/314ab7c596ef4749bcbcf1d46e2304a1 : SUCCESS in 4m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/26a9c4d8a2214106ba63979503828d81 : SUCCESS in 7m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bd2d0b9b749640ab8124935174a1ef48 : SUCCESS in 10m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fd96f67b0f80495691dff7b0a3bfaab9 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce065501be0749549c9ed33711c6ba0e : SUCCESS in 9m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/962624e3a78441cbb2a70968476e6b69 : SUCCESS in 31m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ea89104d27e5432e9107ad6d6b432df3 : SUCCESS in 32m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/68db6d2ec5bb4df7a86b1ddc158eca4c : SUCCESS in 25m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/76d399a9f6eb4737a16fca1d1857fbbd : SUCCESS in 24m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/74e29d21117443f9b675d2621fc4be91 : SUCCESS in 11m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/88d5368322e444de98c7033510efa918 : SUCCESS in 10m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/61433679836f4333b9b9c1f66b5b72fe : SUCCESS in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a83bb1c4c8424a66ba1f89efb159c36b : SUCCESS in 19m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/749dc3ce655c4440a1b8a54313fa361f : SUCCESS in 5m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b3fbc684560344698021f53743af1250 : SUCCESS in 9m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d3f434c7037c4528a29e9179d39c78ae : SUCCESS in 10m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3b5f3f19bbce4df28cbe5f6aa2ea1f9c : SUCCESS in 57m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d77d2ed4f8e4d51a0c494eab905345a : SUCCESS in 1h 19m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca499f76ba434a9a93532fe23b2cbc3b : SUCCESS in 1h 26m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/20c6a5dc131f482f906afa8e32a8cbba : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8e9fe757c5d94c652429bcef1c67ff04bf0b93a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 08:57:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cf350e57a04349b00798513354b4d8 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"031de178c2d5d7e5a315429bb67890590813b0fb","revisions":{"031de178c2d5d7e5a315429bb67890590813b0fb":{"kind":"REWORK","_number":1,"created":"2019-09-06 17:09:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/680746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/680746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/680746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/680746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/680746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/680746/1"}}},"commit":{"parents":[{"commit":"f73a9f655e1da7bdf21a319bbba8b980ab9e84a8","subject":"Merge \"block release that was removed from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f73a9f655e1da7bdf21a319bbba8b980ab9e84a8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 17:09:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 17:09:09.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 7.0.1","message":"update constraint for keystonemiddleware to new release 7.0.1\n\nChange-Id: I420816eabde41a71e47810072678e5072fe63c0d\nmeta:version: 7.0.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I4c182baedd9a44d1d1d520a1129856343f0ee404\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/031de178c2d5d7e5a315429bb67890590813b0fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/031de178c2d5d7e5a315429bb67890590813b0fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
