)]}'
{"id":"openstack%2Frequirements~680748","triplet_id":"openstack%2Frequirements~master~Idda5e45d6b8384728d16962dcfe64d0be7354719","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idda5e45d6b8384728d16962dcfe64d0be7354719","subject":"update constraint for pycadf to new release 2.10.0","status":"MERGED","created":"2019-09-06 17:26:42.000000000","updated":"2019-09-07 04:18:38.000000000","submitted":"2019-09-07 04:17:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680748-1567829845867-c9591a94","meta_rev_id":"32ab5f1b9b3cb9bcad5cafd0380287eba450e47f","_number":680748,"virtual_id_number":680748,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-07 04:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 02:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 02:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 02:02:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 04:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3340d42663bf12e5c32492f25a11a460bd39b90c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-06 17:26:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e51f89b9368eb3e09ff82f6c0ef25065fa47c6a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40539d5950824bcd8db7dbb4c2cd9b19 : SUCCESS in 3m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/48e430a6ab4641b0829a215773d59e71 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb11f6c6f8c4fa99643453446311684 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93d159c4295346069aa21e5b56f2bfcf : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24097220b11b45eabc5c51fedeed7733 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da46c8d662b4462eaede1be4d1e30ecc : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/367d95d1e0f2408584821f4c717589c1 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf98cd6311ba4ddf97930e0096824666 : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a1c3a411283c4f618f8f841468ba1048 : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/86a58941175a47b6992a6c14b9c36344 : SUCCESS in 7m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/63e2207c3b504af091e36795bf6d1bad : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d8cce6374cd74bba9730ad4b440d9974 : SUCCESS in 8m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7dfecf8d88b544c48ad0fa8e9c049fc8 : SUCCESS in 3m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f64d7c7084994110bdf03a09499e66ff : SUCCESS in 4m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ce7f7d93dc9944869fae452b94daede4 : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7abeff72b9d6400fb13a3c90816f49b5 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df39697040094cfbbf231313e69d5c5a : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6efe37c8a4d54ec8a3e1840361e4e24b : SUCCESS in 37m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c84d050d4aac4dc3865ffa5fcde3edfe : SUCCESS in 29m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1a57052843cd420a9ee9af05aa15502b : SUCCESS in 22m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e3f774af649643cbba3af43f26134bdb : SUCCESS in 25m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1711fe839b734f60a070a024233304c8 : SUCCESS in 13m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/266127b9b1d644888dd64252b77961b3 : SUCCESS in 13m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f0e3eac7caf148519d6c5411f2e247ad : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ba428a749474621b985e64e5d9c3bc9 : SUCCESS in 18m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/235483d986d14db3be52654741cc0657 : SUCCESS in 4m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/410f57878316442a8365fe540e4f9155 : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ae1fa5d8c62461a8cfbde13c6f3c79a : SUCCESS in 9m 00s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/71abf13706314981aa6e438368b89824 : SUCCESS in 1h 03m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bebd6b23b3124c2c9687e32b8606bc44 : SUCCESS in 1h 10m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f5d550c6648e4381adf9939ea7a421f4 : SUCCESS in 1h 44m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a25c0cfecf14ea7b37098cb7391b6cf : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c0e2e66f9140fb3114ce94ca105664f41160e468","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 02:02:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61fdb3e9d9dc2289e591294b9beafa2e43c62ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:02:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e045dc3a6030f8ed8e8eb093deef3fc7f4716126","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 04:17:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713f6e9cd6f84677a35e3b776b181765 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdc3a935a5c4d8f90951aecf5bec3e2 : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0601d59d3aca4ae194cfa2c44bfd83b3 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a94a8a20f42491ebfaa23574799ad79 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b5d0851a9804cc7ad3e74172b0f6fc8 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c595f3ae725044aca2eb93da737f7e48 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/712507d352994bbea0a469870593e93c : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b2524b00da8646d3ae87dc726bd06e14 : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4533cfbfbdaa40f1afc444f8e5cbf3c4 : SUCCESS in 10m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5dd071a0cbe7497490629dcad71634fe : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/11ddf5fcd7ba4ad1ba08eaf6de0a3a01 : SUCCESS in 9m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/613a42a4fc094dcebce29735f106a299 : SUCCESS in 4m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aeb0a7aa5988435988a2b5fcdb5350b6 : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d8c6359486e14c6787302f36dfdb81df : SUCCESS in 10m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/00468a1159d64510a1fa18fd20d305b0 : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/392b90237ff242dcb801641b01f80906 : SUCCESS in 6m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9e503c1d1d7042d6befc62605b518c51 : SUCCESS in 34m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd4eabd1a62c47448a24f3c375baeb73 : SUCCESS in 38m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3ce5a37810f3465fb127e68099f098b5 : SUCCESS in 21m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15e746e9de984167b9fe7c7eb8feed5f : SUCCESS in 29m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e0e3964ac1db4ff185faa010d1a5ca11 : SUCCESS in 13m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b3d9eca00734fdf8258502457d1dd5e : SUCCESS in 10m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7360fd5659644d2a84580dbf4c233a29 : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1b5a7a4725f4cf89b1c149a1041d3fc : SUCCESS in 18m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78ece5f55a594b97ac1a996187ef3c79 : SUCCESS in 5m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb28ff869cfb4fe8a7d5820285d6041a : SUCCESS in 8m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be0ec5c30a054d33a7194e7b16a9fa28 : SUCCESS in 9m 27s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e208257634f349cfa7aad12656227c41 : SUCCESS in 53m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1704781539844da94f355f5c23e5031 : SUCCESS in 1h 06m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9d74938c0e7d4791a6641e2e02916b1c : SUCCESS in 1h 42m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0ddf0fb9952240a2b75dbe9360a505d3 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"198bcc3009a63c113ab405fcab56f7366b8e158a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 04:17:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5dfba1bf82782af1bcb5b968e3fcc31b9ccf1275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 04:18:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57374be610b243c18d1c817f4824266f : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49c9af43344dc0f27832bdd903866a668bb565d6","revisions":{"49c9af43344dc0f27832bdd903866a668bb565d6":{"kind":"REWORK","_number":1,"created":"2019-09-06 17:26:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/680748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/680748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/680748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/680748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/680748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/680748/1"}}},"commit":{"parents":[{"commit":"f73a9f655e1da7bdf21a319bbba8b980ab9e84a8","subject":"Merge \"block release that was removed from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f73a9f655e1da7bdf21a319bbba8b980ab9e84a8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 17:26:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 17:26:40.000000000","tz":0},"subject":"update constraint for pycadf to new release 2.10.0","message":"update constraint for pycadf to new release 2.10.0\n\nChange-Id: Idda5e45d6b8384728d16962dcfe64d0be7354719\nmeta:version: 2.10.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I500f15a3ca661f0af4e8a878260b511fe7da4d43\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49c9af43344dc0f27832bdd903866a668bb565d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49c9af43344dc0f27832bdd903866a668bb565d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
