)]}'
{"id":"openstack%2Frequirements~680776","triplet_id":"openstack%2Frequirements~master~I992d8271632740fbf765b7bf076b74fc7a5b85b4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I992d8271632740fbf765b7bf076b74fc7a5b85b4","subject":"update constraint for cliff to new release 2.16.0","status":"MERGED","created":"2019-09-06 21:37:37.000000000","updated":"2019-09-07 05:21:00.000000000","submitted":"2019-09-07 05:19:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680776-1567833593821-813f3e3a","meta_rev_id":"6ac393aefca660a4fb821e81430c2e4574c34198","_number":680776,"virtual_id_number":680776,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-07 05:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 02:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 02:48:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 02:48:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 05:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdc1175ba626267760c5a31efc64442f973417e5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-06 21:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"276023153434058fcd9ca81db6b9cc8019387f0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:41:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ccfead839142c0abb259737a11b85b : SUCCESS in 4m 37s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/df49003f4e64435faa2121a92cd4c4ef : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/426af5c33e4247bfb1f5f2046e09e4bb : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36c4916e5aff41f2b3c3b05b1c7a767b : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92a58a9ed4b24c66aa8730e7e687c484 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab3324f360eb4bf19fc769be9d408da9 : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8aac9b1039c4e1bb2d9cf6c70b85910 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f56ecabf3ed9449eba9763972213b06f : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/77928d6a75fe4e06b9bcd2e4c9cd2c28 : SUCCESS in 7m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cf82946ac87f4636bf2f8d950e3b8206 : SUCCESS in 6m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/23087bd0708048738506fd0b7031154f : SUCCESS in 8m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e958bb2b4183423b9d7693ee793cd8cf : SUCCESS in 9m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/189106f7847447d88427686a60619c49 : SUCCESS in 4m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bdd595aaa3a6446787e9fed842c958fa : SUCCESS in 5m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7b9307c97e854f409fff4bb80ebfbd80 : SUCCESS in 9m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df060c19b28f463abdc1f049f9729190 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71eae9cc3cdc46d99c13aec3381b11b1 : SUCCESS in 6m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cea703a32f3743d082e18d8ad90a57af : SUCCESS in 35m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/225969ee19e84e9cae2db6d30671d068 : SUCCESS in 34m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4c1eae924d864a7ab468182fb2662d1c : SUCCESS in 22m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f99ab5d7395c4f61b318965a6a99eda7 : SUCCESS in 21m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b04e1900a0a541cc847ead8096941d80 : SUCCESS in 13m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/af915f5682814095826bdb07b2fcd728 : SUCCESS in 12m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/083f30bd81d34e23baaeac6b00bdf27b : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f457d8bb3d014ef5bc72d8594b7da143 : SUCCESS in 20m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab8ac0a0982e43d4bf845833efca63e4 : SUCCESS in 4m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/644ba794790146abbe6f4c3ae111bfeb : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c5f073998ffa4043ae4bd0672535f8fb : SUCCESS in 9m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4fa2d9aea9994a3188a27d8480adb926 : SUCCESS in 51m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0964ff8b203476cb38d6d147885705a : SUCCESS in 1h 14m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7562126e5c3f40aa83309a8d6f5d3f59 : SUCCESS in 2h 01m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9da70fe77e264d14a7aee3ae5230bf5c : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2dd1f78c7017c57520afa6cc6edb2f62136c809f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 02:48:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95f4212f4ba4fedfb10524073589a598bbfbe29d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:48:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3482e748a46c873e4ffbd84ae28f31a45f25c1c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 05:19:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312e96be1ab54cdd8b755b993fc4c6ea : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32880de805b44565b9d9f6c5b74877e5 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a98e2d26554e4a13a386aae8ef3e83f0 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa9a4323617c4a13943cc7c272ca3d12 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eee9c17c50b3467b82a87fcfd6cd435f : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/425f25214f6d469b978d4388e8766ce8 : SUCCESS in 6m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7dc8e270f5c4c9b945dd8d1ae210980 : SUCCESS in 4m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fbab0ef063aa4106a892998ef23f0031 : SUCCESS in 7m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/302b6ccf64fd4ce3ab9c462535850346 : SUCCESS in 13m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/354b52caa6e946b88d4aaf327a817ba7 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d37c2daf43db40b6a18045592409b3b9 : SUCCESS in 12m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5324b466326a4131a323ee62eec34fcb : SUCCESS in 5m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00ace41359bc4c2992ff54bf9039e024 : SUCCESS in 6m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3bc1e92cebe749df9d3e0a9cd9c4c247 : SUCCESS in 9m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3affdcf877474b008f54b2e788f36058 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d97d18730394132a2d92bd6b50cbbb2 : SUCCESS in 7m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b9ba1f698eea49fea077b1b556ff0b3d : SUCCESS in 30m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6b2c5a666b54b6d8b053e78a7474bb9 : SUCCESS in 35m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8576e7dbd2844ab9b8a2a86e13ac17d5 : SUCCESS in 23m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/abd6878edc2741c1a0859215d997cdbf : SUCCESS in 28m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9c49447eb62941e597e88ea92268b076 : SUCCESS in 17m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a796b2639ad49d6b074f35933a9a9b2 : SUCCESS in 12m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/946bc2a6cbca4f2e925fc20961784147 : SUCCESS in 15m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0a8fae6fee54941b2ba91ba0f32b139 : SUCCESS in 18m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8726284b276b45e7a791d9c3b66a2d11 : SUCCESS in 7m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/839b77bcc2cf426e9b2c0d833d92260f : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d16f9158a767429994a5f6336c920822 : SUCCESS in 9m 46s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/090cec0492764ae6bbedee7f14638624 : SUCCESS in 1h 06m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/18ee0a1a1ef04153b6efa6b0c8d1cdb5 : SUCCESS in 1h 21m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b04b554eb5a348feabece7eeec3fb6cc : SUCCESS in 1h 51m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e468d1b4a27e4db781f8d0043f8863d2 : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ee3b313e682253b6b43c2f8daa28e4e352395572","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 05:19:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"069c929fb4c2c330b35e566fb13094f9e955d1f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 05:21:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1268d27fb0e84fcd893f8ead412cd3a8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf44b8777443c275f6e32ed4608df40de8984e85","revisions":{"cf44b8777443c275f6e32ed4608df40de8984e85":{"kind":"REWORK","_number":1,"created":"2019-09-06 21:37:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/680776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/680776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/680776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/680776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/680776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/680776/1"}}},"commit":{"parents":[{"commit":"fe8005f811dfd5ccf92d755932cc55d0886930b3","subject":"Merge \"update constraint for oslo.rootwrap to new release 5.16.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe8005f811dfd5ccf92d755932cc55d0886930b3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 21:37:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-06 21:37:35.000000000","tz":0},"subject":"update constraint for cliff to new release 2.16.0","message":"update constraint for cliff to new release 2.16.0\n\nChange-Id: I992d8271632740fbf765b7bf076b74fc7a5b85b4\nmeta:version: 2.16.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Change-Id: I08f7278d93e44263984939a8872d1970c89c9b67\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf44b8777443c275f6e32ed4608df40de8984e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf44b8777443c275f6e32ed4608df40de8984e85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
