)]}'
{"id":"openstack%2Frequirements~680800","triplet_id":"openstack%2Frequirements~master~I853d1641bc99d8450c42ad6504fbaf93df1dc01c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I853d1641bc99d8450c42ad6504fbaf93df1dc01c","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2019-09-07 06:16:28.000000000","updated":"2019-09-08 15:23:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24da1795d1217d1f221e3ad0defcc793795930f4","_number":680800,"virtual_id_number":680800,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-07 08:38:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 08:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a4b4ec8a0ea71d1598dbf22c357864e5b287ece","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-07 06:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a0f8a36265e1454e9b607a5fdbf69aafe3f7e22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 08:38:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3677d6a31404ac7b96722c8055467a0 : SUCCESS in 4m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/aec3d6fc78944d88b30290e18abafdc0 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c65b782f27d440ab9ea62d985583e82e : SUCCESS in 2m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab35dc2c0dc446c3a1b0117408c00917 : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/603f0d5925914d5f863a2554d7e15e40 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e22ca72b2d1433eaae6ac488cb4eb33 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c563d3be842742ea9bdc4952ac0e44b4 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c763cfb055e54e97ab65a4024abc7400 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bcbf1c9fe6af47bca8c673d9b6d79bb4 : SUCCESS in 8m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3fdf4d922225403b977391a1df45ce9d : SUCCESS in 10m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ed719b038100445dae0756e250960e4c : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f6cd3118933144fa9cd4cff35b88cb17 : SUCCESS in 8m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/56773ba837b54838a9feae8f0cba06d1 : SUCCESS in 6m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ff20c1ae89042c08f72f1352f4326e8 : SUCCESS in 4m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e2b6b789037a42c1be662f69ff50a279 : FAILURE in 10m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/40b767d1784b4001a9e5339afe91717e : FAILURE in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a68ca573a5c94a27b8dce15c4f29fac0 : SUCCESS in 6m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/91ea1945283f4b9a9751887eecefc0d2 : SUCCESS in 32m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a03dbd3f96a4c098dd3f11afe54413a : SUCCESS in 38m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b4623c593a464cb6bad37d048496703b : SUCCESS in 24m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18a5de12de964ad38726f50fe28da536 : SUCCESS in 25m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cacc5b27fa3848d88a4c1d400a05ba2b : FAILURE in 14m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb399a7c03b1483d88fcb9879f1f9927 : FAILURE in 12m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b5575892dabf4735a1a6b112af2c504b : FAILURE in 10m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec57302431164cdc855458bc6b12606f : SUCCESS in 18m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddbb3c99e3fb49ec9ded8ba1eb2b7a51 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/53c00ab5497d4eaca51088ebfe9cee79 : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d7fbc64b03e4d71904cfad855d9bbdd : SUCCESS in 10m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4fabba86ac7b4fbb8e74d888540e363a : SUCCESS in 58m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6476c8b9a4aa41dc8595555f535184a4 : FAILURE in 1h 09m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/079145a0a90d4a60a07d983dda129797 : SUCCESS in 1h 43m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2b6d7356fde34176861ece1095f2b4d9 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ff576e60a80f7b5d52cae2eb475cf8975601f7b3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-08 15:23:46.000000000","message":"Abandoned\n\nbot didn\u0027t update","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eac1887648fa82631bfd7821ea85a0d1bd8e20cc","revisions":{"eac1887648fa82631bfd7821ea85a0d1bd8e20cc":{"kind":"REWORK","_number":1,"created":"2019-09-07 06:16:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/680800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/680800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/680800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/680800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/680800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/680800/1"}}},"commit":{"parents":[{"commit":"b56e70a02dce3afde768f947d340a1abcc011a75","subject":"Merge \"update constraint for cliff to new release 2.16.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b56e70a02dce3afde768f947d340a1abcc011a75"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-07 06:16:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-07 06:16:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I853d1641bc99d8450c42ad6504fbaf93df1dc01c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eac1887648fa82631bfd7821ea85a0d1bd8e20cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eac1887648fa82631bfd7821ea85a0d1bd8e20cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
