)]}'
{"id":"openstack%2Frequirements~680832","triplet_id":"openstack%2Frequirements~master~I561ace0db62450e380c962da2bdf783f82fe360f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I561ace0db62450e380c962da2bdf783f82fe360f","subject":"update constraint for heat-translator to new release 1.4.1","status":"MERGED","created":"2019-09-07 19:28:32.000000000","updated":"2019-09-07 23:09:21.000000000","submitted":"2019-09-07 23:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680832-1567897703579-82917982","meta_rev_id":"21aa9d9bafb83674c0787a014f016245e9453230","_number":680832,"virtual_id_number":680832,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-07 23:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-09-07 20:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-07 20:37:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-07 20:37:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-07 23:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f350c16917ef04d1b51f24fa50c5d4593c8070fe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-07 19:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"841d073b5cc8f8d56b041605bbb56b6650ca4ca7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-07 20:37:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f281b71e6c51385765f5c859111696bf6da93f1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 21:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9151f67f217d4518b85ec51949813582 : SUCCESS in 3m 11s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9d28017c89df4424a2ac63ada259f08c : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a9285232dd4e4aa52a73d776180e88 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa5f030ded674222a6ab208636569460 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04914c1ca6ce4790a6c80dfbfb8fd361 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad91535b2f584f3fbceb035baaf2cf8a : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e280ed015c174890a2d0765a8764218b : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/398c4545bb1f4e48b52fb43b02679cc4 : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f8aa43884ccf4c75bcb4e9dfd6870a5e : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b2ca8474aa0a4ba18f0a73dda786198d : SUCCESS in 6m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a7280426e85e4ae485a521fc31dda464 : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ada03446f2a141d79e3bcf1cbebbff98 : SUCCESS in 7m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3be2efbb442f4d89b8b71ed04dd2f329 : SUCCESS in 4m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/972f64e2fc2f4ae48ddb42e519d77066 : SUCCESS in 4m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/94edc1ee190644e9a3035e365a8c6266 : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2cefe06a8bc845bbb5bda3f3badc89bc : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0add0aa08f154f6896118301170b89ac : SUCCESS in 6m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4a95b911177445b1a84d55aa17e69113 : SUCCESS in 29m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5545112340234ef892dce9023fbf37fc : SUCCESS in 31m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d743d8d51a2540919ac93d53b5b3a36d : SUCCESS in 26m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/659c6c9361cf45028812d83d18c7e647 : SUCCESS in 21m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f3a570813036420c9b443a2bc12d65b5 : SUCCESS in 13m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3cb2a1b9a5f142118886a615bdcf04ad : SUCCESS in 12m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9bc544a589144a94a69fb6aa69c34003 : SUCCESS in 15m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70efc9f759eb45ce9dc4928586dbf06b : SUCCESS in 19m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa5753e84df54570baf27f831cfebf9a : SUCCESS in 4m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/26c97fbf2c1c4c57ad7302ebb6884038 : SUCCESS in 8m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e2d5bb851c384a2aa43c066da5f2ef88 : SUCCESS in 11m 11s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/fc96def4dd854f7a8cf4cad98e47dc60 : SUCCESS in 55m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b05b828dfc749e1b14573c8a2fbcf40 : SUCCESS in 1h 10m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/567c0ab1eb344ad584854daa377774da : SUCCESS in 1h 36m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5c12eda96c3b4163bcde05f0c11d3a2f : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8a92fc7e0b086201a354b8021d67d4d3db6c39d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 21:11:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8f8b522305ce01a56c833637610b6463e0e6da0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 23:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b965d83b3c34f53bed592e8e4e46654 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab1a6de59a04b99b5f91b227e9d4de7 : SUCCESS in 3m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e75eba917014b458628b08350585a05 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/525c660454ca41d0b3abc35579086e1f : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8a1e8d09aff4bb8a21fe4244fabd6e6 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d1085ac3f0d4b31bc721851ded1a155 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef2c46e37628493aa9e9a83c6175e629 : SUCCESS in 2m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5067776b393b4a229c8258f2b7d8ca14 : SUCCESS in 6m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4be17a4827541a6a3bbd63590952403 : SUCCESS in 6m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5893c7c9e4e248caa4732832c33aaf33 : SUCCESS in 8m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b26e649aee594b7a8d185f24dda5e0db : SUCCESS in 8m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/26cbe4e154f349f69b41ad7dd2e68aaa : SUCCESS in 4m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/87036f99f79b414f86fb8ae769274e74 : SUCCESS in 4m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1079ba385fda46fe8c41229ed41f3182 : SUCCESS in 10m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5139fc6979a6470987202f31b9917526 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3d65dad8c29423594fd59a98bcd6a42 : SUCCESS in 6m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fba7c5c454064221a85d33532e6c2c3a : SUCCESS in 29m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c399acad994543fea7dfcade81312d34 : SUCCESS in 31m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7bae3420999c4e30b5a2a9889505cab4 : SUCCESS in 24m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe874b631ca140298021766e0bcedc37 : SUCCESS in 22m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/122047231ad944a48ebb61cbf500c0e7 : SUCCESS in 12m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97848c448fd448f3b2e1313bfc23a34b : SUCCESS in 10m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/796e4c4628ad4300a6c292b36cda8883 : SUCCESS in 12m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb0aeacaa0f048d18415db1e2da497b9 : SUCCESS in 18m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcc5aa3825e84eca964240829f83f24a : SUCCESS in 4m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5df8e959693d46f7aae4bde82aa0053e : SUCCESS in 7m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b8878532cbb348e396d8c75ad8258c3d : SUCCESS in 8m 36s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4248257b5ece4e23a35c5ac27a6ec5c0 : SUCCESS in 57m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d31567e7bdd449a2a450defd21bd1b35 : SUCCESS in 1h 10m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04c6329d79634327afbc3b7fe21d8b7c : SUCCESS in 1h 32m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c13c63a2577f4bb6a34ea61f93db5b06 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"392f5e29012d0b1c3887f82fbbbe6ca1787c8513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 23:08:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5271cff79a7c4bf0b53c19e3f65e2144dd559a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 23:09:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef8867712254159a48965607fa4f53e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"293b8b1b235a73eebfeae99c0ad8d05b8d2fdfc2","revisions":{"293b8b1b235a73eebfeae99c0ad8d05b8d2fdfc2":{"kind":"REWORK","_number":1,"created":"2019-09-07 19:28:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/680832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/680832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/680832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/680832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/680832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/680832/1"}}},"commit":{"parents":[{"commit":"74220bae1e306be96010c2f527c3497cc2856d48","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74220bae1e306be96010c2f527c3497cc2856d48"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-07 19:28:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-07 19:28:30.000000000","tz":0},"subject":"update constraint for heat-translator to new release 1.4.1","message":"update constraint for heat-translator to new release 1.4.1\n\nChange-Id: I561ace0db62450e380c962da2bdf783f82fe360f\nmeta:version: 1.4.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If0490b885bf9c4086f79e1fafcf8b1cbfacb52e4\nmeta:release:Code-Review+1: Rico Lin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/293b8b1b235a73eebfeae99c0ad8d05b8d2fdfc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/293b8b1b235a73eebfeae99c0ad8d05b8d2fdfc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
