)]}'
{"id":"openstack%2Frequirements~681382","triplet_id":"openstack%2Frequirements~master~I0b2eae0e848416e05430278a5a934a256b0ee001","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I0b2eae0e848416e05430278a5a934a256b0ee001","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-09-11 06:16:40.000000000","updated":"2019-10-08 10:14:22.000000000","submitted":"2019-10-08 10:12:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681382-1570529573710-9cc037c6","meta_rev_id":"e3f96d04444ab501fe1606a9672a3d0ad3cf702b","_number":681382,"virtual_id_number":681382,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 10:12:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-10-08 00:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-10-08 00:06:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 07:58:41.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-02 08:59:36.000000000","updated_by":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"reviewer":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"state":"REVIEWER"},{"updated":"2019-10-04 06:29:08.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-08 00:06:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-10-08 10:12:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e401543333d384fb267461d4db925ebc5fca9f53","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-11 06:16:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bdac7176ae6054821e4779aa33ea1c7dd0af842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:33:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d09b9ecc564731bc8ba51b7df350cf : SUCCESS in 3m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1899e212b6fd4f6cb67b536c85144ba2 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0096b4ad6bca4471a3ae374103f22fc1 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/141a0e1519134d1ab37fe6e073d7e2d5 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a43711d2f54403ebdb720df4a10abc7 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0323cb9524824145b42ff369bd55eb83 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/543515499b8d47e6978c27c034fb08c6 : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bf62a8020234f9fadd599b80066fbf7 : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fb06c202f0344eba9a4e230103bd0b36 : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/664fcd49c97c4265a57b0bdb0d77b6fa : SUCCESS in 9m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/293f8cd6b4d2436fbad78c8bf6413a72 : SUCCESS in 11m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3595faa57c894bcb92dc51b976936c06 : SUCCESS in 11m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3529335d67e24f8b89cc18185f7b2c79 : SUCCESS in 4m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e262f947ff4d44329fa1704138557153 : SUCCESS in 4m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4f2de2adec00461a8cf9b458f8716c45 : SUCCESS in 10m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c8b8a11c3f24c2fae26956f0bd4a437 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/850408304f954542a2aa69967a75ddf2 : SUCCESS in 6m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/989366b3916e443e822d6ed00259bbbc : SUCCESS in 41m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c89b1ff333b448e4896826504d9ee591 : SUCCESS in 31m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d16beb2642ac4ae5b4f173efc8431683 : FAILURE in 26m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5abc34b698264ce294e911fb5fc11c4e : FAILURE in 22m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bb6978e93e1543c682d4f41f279c20cf : SUCCESS in 13m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0a76d0cdffc74aceabfc53b32a03cffb : SUCCESS in 13m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dd66c6d1336945ca87b0c337c9b8314b : SUCCESS in 16m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d2fe0291d2e4c3c8d9ddd7ffe58b2cd : SUCCESS in 17m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/562a0891497a40b2bb45e05dbd90fc7e : SUCCESS in 5m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e99b590c857f405a8b1e764bdd239685 : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24ecd7ec72b74513ba5e6fdce1270393 : SUCCESS in 21m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dbcd95f637c74483b0a75caaab216e99 : SUCCESS in 58m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59770c1176dc48b5be3033cc87ad4bf5 : FAILURE in 1h 14m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdd35be71f6d40a895dfb06e2273d6b9 : SUCCESS in 2h 02m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6a2fe2604dd94f1fa0b7c3ce440e2c7e : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"cd41df5cfd6c190700e2b059b15fad73e8034e1b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-11 13:46:03.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\nfreeze","accounts_in_message":[],"_revision_number":1},{"id":"148c9c88e45723b28e705340b7bdb23b9f324ef6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-12 06:14:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90d851d608793b94bf2ec85def30690c20969054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:16:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2034bfe7e5ec4eb6b16a187164271071 : SUCCESS in 3m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ccafca5ef5504b2d8b7af8e1f9ae0106 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64bd86c327ed4a62aa0cf209b5926519 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e819654db5d4e0b8f40f7710ecb83a1 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeca76fd5c764dc89bac35a001398011 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0941ec5b30864dc6b9edd10f6088d664 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4136adcaaf0442c1a0f94190c56b80ee : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76ee18e55b8b467893132a66e160e1e9 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b24290dd1bec4be5b38b025b6d55c718 : SUCCESS in 8m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6b4f5e458203480d82d5e895ac63ec54 : SUCCESS in 7m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e0ffcb48725946238271c1827273e120 : SUCCESS in 11m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d47ce797eafb458eada8627b5da50daf : SUCCESS in 10m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/05c4a084b46c4851857a756930e16900 : SUCCESS in 5m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/05f0c65796a246f5a60ce22529e50cd3 : SUCCESS in 3m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c1aade8dc0024f18878670b75ab6f526 : SUCCESS in 10m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fb0a39a6bbc641fca1b72691e5aaac9b : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba4ccaaaf21243c58ab5a7fdd67952c3 : SUCCESS in 7m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3be7560d4dad45d8abcf27b31ef7c52a : SUCCESS in 31m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9bf5563b77044a4d9e587a6a34805d25 : SUCCESS in 35m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9e571d688a4c41fe8139c0981609b014 : FAILURE in 24m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d9a9b7173b34c3cbb7e8c398bb6e072 : FAILURE in 26m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7e42739ccec34140978e92a0f6048570 : SUCCESS in 14m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f6ce87e4c26a4795aeadd5de36020ef4 : SUCCESS in 12m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fee0d5977060455a98b18c50aedbf82f : SUCCESS in 12m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f87c8b407be439bb6d1bf20bfabeb6d : FAILURE in 20m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3416d871a9f42b497b75267cb4e6d90 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5aa7ce89885e4001adadefa1d0f02d83 : SUCCESS in 9m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/78e20d3b7c204478a926ed2033621372 : SUCCESS in 10m 09s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dc28cf5b7d1e49e9ae2d268650a3f015 : SUCCESS in 59m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c17158b51814c8f8e763b75008a3848 : FAILURE in 1h 13m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b2cb8622b1fb4ebcbe60198efe8b2bb9 : SUCCESS in 2h 05m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/04f489596e90422a8ce55a3e28b654c0 : FAILURE in 2h 19m 14s","accounts_in_message":[],"_revision_number":2},{"id":"cb73f438b108b85bdd0d05abe2eb5cb7c192eeb8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-13 06:34:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81b9c6e34c4f7df781aa0ff3041f5979a8b428d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:21:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac75f043ca0477ea8a069137a3679c0 : SUCCESS in 3m 43s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d53124c31f774b0a89ba56e68fd7915a : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab2d9fd6b84493ca30cb51328cc6cf1 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1526b1b272d46398540f656fddfdbbc : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74f85f3d23ef4860a67c55e0857fe801 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12f846476e7f4702922bbc82140511c1 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85461a77bfcb4324af8515ab138976ab : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccfabfe62cae4ef197ed05154a778d34 : SUCCESS in 2m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b74b04927f81463bb45bc0cc03269581 : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf54aa0264a24d70810f516d9438cb30 : SUCCESS in 6m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b67c85c689f6472bb55d552ec2a61f84 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7899ff9294e74aadba623443c897e7c4 : SUCCESS in 8m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1e9eb14611bc48bb83e972315a79b0eb : SUCCESS in 6m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/92ec41c9f55e43cc953b2c06d73377e2 : SUCCESS in 5m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/89a054af22304d129b598c0f5e56e467 : SUCCESS in 9m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bcd0422d3ecd4f6c8c26490924ba8df4 : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b758214af7b491eb1f96b06d6b1fa22 : SUCCESS in 8m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2a0f2a2b200842d9a124785927d2c9a0 : SUCCESS in 40m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b55116b66e03421a80b7e80b4d649bc3 : SUCCESS in 35m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/06d7036fd13142c9a4104e0d75e87a32 : FAILURE in 26m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/143933baf81d47f19af226b81ec3c4e6 : FAILURE in 22m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1aa2ec13c5954717aa5d36e74f398a8e : SUCCESS in 14m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ce9a86c28634edcaf29bd1528ce1d48 : SUCCESS in 12m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/236e7c3deee24885a0e88e08f0533e1d : SUCCESS in 15m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8859dba30d99455f988e0367113cc7a2 : SUCCESS in 20m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c160d47fbe14d4c99ed57b750adf2b1 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2596a6de6c174028b20a06610ab3e858 : SUCCESS in 10m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c14247b062de42068c1e276d7b91072e : SUCCESS in 9m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/474b729f4fda4dc69ffbe120d1915563 : SUCCESS in 1h 05m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2740481ba1af4a23b2c182fd2073bd17 : FAILURE in 1h 12m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6578fff3394441ceae90956e0e1c32cc : SUCCESS in 1h 48m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7d13f6796e1c400283a05c6a3007de36 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":3},{"id":"44a47a6582acda3a510598957a6077cb5ad719e9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-14 06:11:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d1bcdafffebe84601ee68b37db000404e06e2732","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 09:14:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3a50505a3842deb7fa7ecaa98bc3c4 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/95f22716ebff4706ba4ec4224c2e8509 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a78b5dfaef54330af9da246546ed358 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a802f29d1d0402e85a73d9aa7e7e52d : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c75f738751ce4927ace977c3d5cc06d9 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5456f528ec1349c3b78c632d885d333c : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dfd74c7ea00406f9ae5c336b78a997d : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ff488c8188b46b8aa1a37aa58c4094d : SUCCESS in 3m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d396b5efd07b4766aa2137622dc637ee : SUCCESS in 6m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d7d15839f2fe436faa1a37c4a83e9270 : FAILURE in 3m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9775feb69fef453c8cd962f49d9fd499 : SUCCESS in 9m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b478cf24f0240d987a239bf15ef4155 : SUCCESS in 8m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e6b30d2d9df74250b780d327f1330c5e : SUCCESS in 4m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e4b271a1b004405b29452694bba08ee : SUCCESS in 6m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5b9e898159774eaa9d4aeada12a97660 : SUCCESS in 10m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d1159aa7b4874e20b23241927cb71dc1 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5f2bda3f33a44bf9495dbba6543a0ad : SUCCESS in 9m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/34f16c10f838403281a70e0b0d05c48f : SUCCESS in 44m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d9e836815604e0693f9ffd7cf12db61 : SUCCESS in 40m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f1ff8d944b754d068ea83287c40fc3dc : FAILURE in 29m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7c3d9e5953ba41a0b3b8bed02df6f138 : FAILURE in 23m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/37b7319857d748f5abd6b1693b4aba91 : SUCCESS in 14m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/563cec70977746c59ee260132a3e846d : SUCCESS in 14m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/590b74da0c3b4376b299b848804d96b5 : SUCCESS in 18m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08081a8f59d14c66a2fa5f126698e134 : SUCCESS in 26m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa801a97f3974360a1b1fa597d707598 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8eeae6da94fb45dcae820c8c2436e7ef : SUCCESS in 10m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4250141069cf446d8444b270a4b14742 : SUCCESS in 10m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d3e59aa95a0241cf9179708ffcbbe720 : SUCCESS in 58m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f14b69b3ea7644af9e7db0c00ea4322f : FAILURE in 1h 03m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3293005ad8c04006b9eeb0b8251081f7 : SUCCESS in 1h 29m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f441ed8c258249c9a3bf212f6d08ab6a : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0184a8f1cbf4e0ea2422ae894458a036e7cbef07","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-15 06:14:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ac6585feac1c8bc70274a0df1c49aadfb14dbf23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 08:56:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef3e1890ec74b3d8f59b7ae21f3d30b : SUCCESS in 4m 25s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3db5bb49987f4826b9f3bbaaa61a4fda : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c320884910448d693ad87efb5760f51 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d124809da2474ec090c87f7614fe573d : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2390d3a72305417c95152f4b3f674e1b : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88df39953fdf47cd845fd616fe3dab1f : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c5b0ed984f44ae883c5fe820e51d4ef : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70b0bfe2c7b848ad8a90d69b9ce1be92 : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c705a23bcc9f4574b0a086e41cda6cc6 : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/30948cc9997c4663abb1877fc541087d : SUCCESS in 7m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5872bd0201ff4d4dba59a9198f26918a : SUCCESS in 10m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7db704428667486582ea58f295681945 : SUCCESS in 7m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/37a3ab1a1da546779ffc8c8bdc7ebf4e : SUCCESS in 4m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d783354bf14549bfa5db76ba56d79d0e : SUCCESS in 4m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9ab8a7311254434e918c7d6e36f8fe8a : SUCCESS in 10m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/caae2d2c9dc243f284e8514927a7cd53 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/241818a3194942909245e93b893c8399 : SUCCESS in 7m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/82f50e9beeac486da846008fde0aa075 : SUCCESS in 57m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3afe025c32346f2a54f199aa530abe5 : SUCCESS in 49m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/82c4c58adda84f58ae5330a98e598c2a : FAILURE in 31m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a9d95945c3f4a92ba6c15d27fe7a0e1 : FAILURE in 21m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7454042a562a4f9783d1b96ec5d825d5 : SUCCESS in 14m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b19ee33f33a4d15acf85cf4eae6b765 : SUCCESS in 15m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/301ff7abb1094751ac39437de91f932b : SUCCESS in 15m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a69a5a2fe8bc4fcda791219249454c21 : SUCCESS in 19m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bde3db6820fe40848d80928101aa6de7 : SUCCESS in 4m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d6eaa75a355a4579b3c1db1bbaa90b98 : SUCCESS in 9m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/01192db5f6764c078e48a647007e52c6 : SUCCESS in 9m 33s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7c8ec63a61264d2b99210f96742d6922 : SUCCESS in 1h 06m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/856966b1ff7b49a49e7142fade12a4fa : FAILURE in 1h 18m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/33431d09475047b28cb17a96ec5cae16 : SUCCESS in 1h 33m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc70dc0f9b2b4b3f8f007bf0307a1d2b : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":5},{"id":"bef368e44fb029efd6ae7612939eba865fdb0173","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-16 06:26:09.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"14968966b514f3a82204ceaa0bfc3227db6be70b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 08:40:25.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad55f220d1474196af52ab0d949fa110 : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3b696c68a6a14e36bcb857fbf26a8212 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c2b56ccda54be586fb8d6331c28966 : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a06c69a32cd042049d43fe6b157c5bc8 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc767e7d9bc74ab083e8850ecc82c9e5 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/308347ca1b10461994bececd09bc717e : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d526973c6ab74fab9c96da1ed34036d3 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/680c3f87563b4fb6997fe29d7893823d : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/63c3ca8d334b4d13ac6a5bba44109d7c : SUCCESS in 7m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4284a162c29745df95926454e5871c50 : SUCCESS in 7m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8334ae84122444e19f83d116792dfaf9 : SUCCESS in 14m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a2c5aa91ba7543fe92d06c3403a5f7e8 : SUCCESS in 8m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a9da78b296714770b26736d877220608 : SUCCESS in 4m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb8f13d9e02d416483b72e60f4f9798c : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1a47d86dcef34919b00640d6339019cf : SUCCESS in 15m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5064796eb6e6470d9991e1b32052ef56 : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02324a804bb6479787a85830fbc118e4 : SUCCESS in 10m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a4484e4221864d4ea06b84a146670899 : SUCCESS in 50m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d4b04009fae44b3a8d39e7db8825571a : SUCCESS in 38m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/79784e4a3e2249bdaa967de87204074d : FAILURE in 33m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4b1245e9274240f18250c7536a9e8b98 : FAILURE in 32m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/17ebb6aba9c84977afa31a532b171582 : SUCCESS in 12m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6bd873cac0ac435eae2f3809fa13e74e : SUCCESS in 16m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c2808fbc2338428396ec756b98485ab4 : SUCCESS in 21m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22f6214eeef2434b8dae1356a1e26596 : SUCCESS in 18m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9fd0c0335de245ff8f191b9afe37c86e : SUCCESS in 4m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a70a0cb741a44920b32595418d4a8919 : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7997d78f03f54017b463a6ad89b90852 : SUCCESS in 9m 06s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dc06b43df2644c5ea3c21b8054fc755c : SUCCESS in 1h 01m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e02bfbf5ef645639bb3f0e26bcd42df : FAILURE in 1h 17m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8ca059206fe546a09d4054fff5493fbb : SUCCESS in 1h 41m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3af19328d00a49b19548139af2a8870f : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":6},{"id":"62c0c19f64dc0d9beedf141ab098e042180d8795","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-17 06:38:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"90dea4b347f8eb79b6663594036052cfaa429e58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:01:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ba0bde4ea94ccfa94f251955d9e3c0 : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c9ac7a5da6a84cc0b92acd8f4689fccb : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7a2f2d41f54117933d895350c97e84 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf329307ba3549c5a24b9deab81c63df : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b160500224949a6988ed5667c7382c1 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52f5e99a146d497bad51edd1c8292759 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/941670afd8a84748b5375e6d5f8f7a71 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2b26d817a8b48d78799f427bc45dbb1 : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e053eb128f034db6b18b058157057431 : SUCCESS in 7m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2abd0355e8dc4030958a714f1f58c43e : SUCCESS in 10m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/495fa6fb659e43668953c6423efb179d : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b631764c7f74e9b92a3f482aa576b4c : SUCCESS in 9m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c6aab1614b3946d1b746c7a07ff975f2 : SUCCESS in 6m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/827d37eac6104ef7af6ca423b457f7b2 : SUCCESS in 5m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/035e49fd4dff4a9591ca5facf41365ed : SUCCESS in 11m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0de0b2a55d2482fb3c1449b1aef54bb : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09aab330595a4627a67a0d143fa56dcc : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4c623ff827394c6fafac4b89889ce070 : SUCCESS in 46m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd799df804134e92aa66a35ba62e512e : SUCCESS in 46m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2f24fbeb2d6c458983be651f7a909fb1 : FAILURE in 28m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf98a377159840e894cc030cef95e7cd : FAILURE in 25m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/22b0d1ce7def43a1a3018458decc53ad : SUCCESS in 13m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee4cc5ebc44042e2857a14a5e4328c31 : SUCCESS in 14m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ae9eef955ad24d2891df37471428b717 : SUCCESS in 17m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d31177b61e63441a8c0ada255715913d : SUCCESS in 18m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0765c9400cdb4c8ba72891721b58e17a : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ae7a63f57f994adda76304ebcc2dcaab : SUCCESS in 10m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc832b7e467441ba919781c6e28b3f97 : SUCCESS in 10m 59s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9cc06b1938f74f28a3b6f2ad8c3cf510 : SUCCESS in 1h 03m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b74708b295c4d2db9cfd2e2af971886 : FAILURE in 1h 08m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9b01383dbb0f43469e64e8256fb600c3 : SUCCESS in 1h 49m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5079b79eaa024ce3aca9777d43f84cc2 : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":7},{"id":"329f92b51e9377820e8ab3127677211487f1fcd3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-18 06:17:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"95e2c870c4e4fbdd5c3e0cd40eabd32b38e16eb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 11:15:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19bcf5fa3dd2413ea0d5bf638d1e2858 : SUCCESS in 3m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9ad1f2a4c39843389de4fc6f466eef88 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c073bfe8fc4a03988dd2c17755ee3f : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2e7ef2d536640049dc8c4460ad17faa : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4dfbd4d1f2c4709afa94bfd3654c9dd : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf2f1086cbf84779829be1ab12f8cc82 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af7b1baca7554fccae9e86d13a2773e6 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9e0621b93634ec9b405a1928b81753f : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/86cd2a2c94524ca88fde5f9e1556f551 : SUCCESS in 17m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/30ec0f7f800849109a760025f3c13f25 : SUCCESS in 18m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0e5a64852f0e498eaeb0d649f180c5a6 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d97342183686455f800d96660e21d14c : SUCCESS in 7m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f60f5e4ba8c04d788bbad3912db5858d : SUCCESS in 4m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/10f45b88d0e24a1b907887ce0296cb09 : SUCCESS in 6m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3f402e57b9c1489fa122be07c1185861 : SUCCESS in 11m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a5b6619166f41e597e1529b7ca535e5 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1a4df3027204fd68211edeea3105a22 : SUCCESS in 8m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d2d737267b0242ee89297613af334624 : SUCCESS in 14m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb00980cf6f64e53a42a0d5ec572a979 : SUCCESS in 15m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/17b1c5c7f3044afd975d774828ca72b9 : FAILURE in 22m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1bb0ff20c0b455db45ae8cac20a3469 : FAILURE in 26m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9bdd708294a34943995612a2b2cfad26 : SUCCESS in 27m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a24be18ab0c9468b846442ad2d034fbd : SUCCESS in 26m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ea30587493cc47089454862b3b47d84f : SUCCESS in 26m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1b94385cb6840c084dc29df7475b923 : SUCCESS in 25m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd7189d5f2954d548c71c3404aa2a7c5 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/082dc11ffd504ce185d9bb4aaca53245 : SUCCESS in 9m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7cc9f7ed17a9489e88fbfb2d2a10327e : SUCCESS in 11m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/43a43b219b134ec8bac088f6ea11e811 : SUCCESS in 56m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e585674e5dcf42ff8c000539e2588ddc : FAILURE in 1h 23m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/394b29f124004db9ac269b740cfe0aa5 : SUCCESS in 1h 42m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/298d85e0af67475bb0b0a051ae88e999 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":8},{"id":"9654758b9315415f77f0b9cce37b6f43eced4fbc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-19 06:47:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"041b2c33eaa87137bd6b45c78efa96dcc4235174","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 14:21:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b533ddd2216242fa9375f78e85877973 : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/12e193df23ce4575a8dc07bf107f0d44 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f055ced280ed4aba8f594b00b71ca8d7 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cb2544030ea4d5b8f4036651b0d225f : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbfd0990054d4bd686d3bc8eddd96a5b : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1fe928966424ec198fb3f23e9bd1bcd : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16a27e5f494d41d08e81b258e2a8f139 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9a0466437cd4ffeb33f6402bf5794dd : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/327105ad9b8a466397026951a5cf344a : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5af664b4f1ec443c8728216fd5ed9693 : SUCCESS in 6m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f6bad2ee3fb44578827b24b07496a3d8 : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b88aa26cb5f947a386c74f00f9ec01f3 : SUCCESS in 7m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d7ec840444434c0f88d19730d80c6d9f : SUCCESS in 7m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df22bf353ee44044a32d77c8b8c52e8e : SUCCESS in 3m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/18f5274ff9b24c58b24782acf236c03c : SUCCESS in 10m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89fba250aacc4388bbbd389b634b65ed : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/617c8aa595c44f7a94157ab476836caf : SUCCESS in 7m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/834f33aa22404c36be561e2dc9fab9f8 : SUCCESS in 13m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8bb97a98e43745db8e235fe203a4065e : SUCCESS in 14m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/236a16d274e54cb4b5a9085d56110eba : SUCCESS in 28m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1617d74e8b7d46e6a5632c5dd99e5620 : SUCCESS in 22m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0a5876ac8ca34fc9ae1bf4c8250f8608 : SUCCESS in 15m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/18ec7f808bc94ba5b67c85ad5f1ddc02 : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fd500143240b49ac97f975f1515f311a : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eefc63e375c4db4b88e2a4d4aaf600f : SUCCESS in 19m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18b4e5a6be9b4c2e831a3336a27f68f6 : SUCCESS in 7m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eab205e42db34260a8199e4834e6775d : SUCCESS in 12m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3649c87e0d9444c9e455c81247a9056 : SUCCESS in 12m 48s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3aebbdab128045f1a8a5736dd70d1265 : SUCCESS in 1h 11m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6bd9ba7055e14a6b9193a52151c4ad82 : FAILURE in 1h 10m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1c9e0a4a7e744974a1c0d483d8960859 : SUCCESS in 1h 46m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e354e72d24d34d4192ccb34e7ac1d886 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":9},{"id":"ebb2c376892d8cef12bb818bbc3d4ee186a39bb1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-20 06:27:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f33eb9fca9a80bbb37f594951804672ddec6b231","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 17:45:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57ae5d5f59c4831a1129bcc5ec900ec : SUCCESS in 3m 56s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e123c0b30401411c8a5bfa2a60fedd01 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208e6c2d4b044acab21a996d8d06a5fa : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30365227350c47d7adf98bb7828c2674 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/165572f1e9464cb19e394a2bf426c671 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e063d4eadd7c497b9a76e682e090f269 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f45f9e515e944482805cb13b94afdd59 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78f1eb93444647daa51b1411962f6873 : SUCCESS in 6m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/320e270fcf4247af8ad7bbe8162ffb22 : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/475ddf964ce34878a46948f07738585f : SUCCESS in 5m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3c559cbc675e4f10906a0a86a742482a : SUCCESS in 14m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/17c6f7d592d242c0bf13a663d4a20916 : SUCCESS in 8m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c1e0912cfb1842738ae93d6088641597 : SUCCESS in 7m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/88ed120f530d46e2a157d1c5c22d4d70 : SUCCESS in 6m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d49e8c437df4449bba078bccabb91739 : SUCCESS in 11m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c345b74c550944e3892b8f2329f2553a : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ea61850a9fe45fd9d1ac4346c48219c : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7c8e0e2641ca4789af7239d101299435 : SUCCESS in 16m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/61e8a1a581c54b42a743e78f04aed703 : SUCCESS in 19m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b9577ba2c6ee43b69a84c85a4ef51e4c : SUCCESS in 22m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0cd97901695d4991ba374c7c704d19f3 : SUCCESS in 29m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/234512cadb174ee0891a5c5c6220e93c : SUCCESS in 20m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2af7a0ac140345fab9c25d678c2d2036 : SUCCESS in 12m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/74cbf54bb6f64c6aac7ba24c331f439d : SUCCESS in 12m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f04c1ccb1497497980f2d86168901077 : SUCCESS in 18m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e345e5effebe485aa1015c7b3bfe886c : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8a72b8db321427d937aba12e006567d : SUCCESS in 12m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7a5091c8f10f4a7ab378fd3375eca2fe : FAILURE in 11m 00s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d29bfb8a18fd448aa1d9c02ab1c872ca : SUCCESS in 1h 06m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cce08f65db14b458ccd91ad2f215e9e : FAILURE in 1h 15m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/88990c40a89a4c6db0773575e6219cab : SUCCESS in 1h 36m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/609fa25d6a8546bfa53112be0ea1f4e1 : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":10},{"id":"1f428f10e2962c43c9b7dcfe4b3f769cad5e643f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-21 06:36:11.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"885928d1c185587d64b0ce918861d03a6fbc0604","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 09:22:49.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9197a4ee376344fbb936da18d6a59b9f : SUCCESS in 3m 13s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f9bac3f8e71142a5b3d5a9835eb08f22 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f85eb4ae24e4e0494629c543e14e0aa : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5661f267909e48c98e9cc34147642f6f : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae8821f484349f38725192aa9ae103c : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/017ca13876f548dba7254af3ae070667 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/225e86fb209a44219d30f37f608a20b6 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/968bee3332eb4b9eb01fd725de27380c : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/31f78a4ca49a44aa91ad1f7a06948f11 : SUCCESS in 19m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0f5527543924bd49d29d3fec66bdbb6 : SUCCESS in 18m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/85367f1bbd094e849a8c504d05c3ea2a : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1031b34f0e1245cba402c46543691478 : SUCCESS in 10m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5d1bb4a4e1fd436599cb94f167fdc24a : SUCCESS in 6m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cafb71980adf449c80dd027b20e27415 : SUCCESS in 6m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7cce1ee85cd444c2bfd62279af31a2e8 : SUCCESS in 11m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b083a754b9b4fecba3ac176ad3a2a94 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/475d95f93aee4a2eadfad16b22b02cc4 : SUCCESS in 8m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/389b23a8951c45a6913f5a44fe61bb85 : SUCCESS in 30m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/08e945b115de4a9e99a46d92a22759aa : SUCCESS in 28m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/74450185f6364c76a896a9ccbc0c1d1f : SUCCESS in 25m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c27bf5adfbde4241880eeca95a467948 : SUCCESS in 26m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/670b7dac940d4384936dc8772385ed0c : SUCCESS in 24m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0648738b7d144d959d773e031d8a5635 : SUCCESS in 10m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/01bb2b9a5c374f1bb610b09ca9d13143 : SUCCESS in 14m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba3ef6afc23c4852923c1b18a83e1377 : SUCCESS in 22m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b01f1698f7884fffa85bf506a29db1d8 : SUCCESS in 4m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce7b62813a554e40a0cc3c31012988c2 : SUCCESS in 7m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e108359725a4c64b140e28d87a1cdd8 : SUCCESS in 10m 04s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/20a7866034244fda87df5cce720aa4a5 : SUCCESS in 58m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c571edcbac7d402299b40dd6709f2782 : FAILURE in 1h 12m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff660d5eb98043a5a15696853af4eb66 : SUCCESS in 2h 02m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/46f9f857a16d463dbbb2f7881a03b7a7 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":11},{"id":"3201f62bec180d03d918f62991ebd03bcf69c8c0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-22 06:22:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"710305eddc276cda3529be4cd9cdb5c99c6d58f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 07:57:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9048485a03554402a3c849277db47558 : SUCCESS in 4m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c4e40eb671be48d89a34d8b963a2f0e9 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726b0baeabff49f89fe7101c776f26e2 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3639af1c335c4d88973625064de0f89d : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52924475e95c460388e8bfaf9919835e : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae948a332f3f497d90f741eb1206b1a2 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d1e977196d746aea33ab43745daf71a : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d77423bf1f644d1bd57c8f268940d5a : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8a7879ebe15e4190b6a56949becbd12e : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/97c0e55fbc81479885bfc5c73755a0d9 : SUCCESS in 7m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/530cfb233bda449abf44feb9cc437014 : SUCCESS in 11m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f4ec2dd8ddd54caa8fe50b0535ac0bef : SUCCESS in 8m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d934b33ae6384d8cab4e01925674d95f : SUCCESS in 4m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e9d3d30f37bb4d38b9b863ab28974f8b : SUCCESS in 4m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/04a0854af3da4366b12044a2512ce4be : SUCCESS in 9m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/864ac742d270404caab69be0a4efe5e2 : SUCCESS in 7m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62631b3f3ef04e829fe537fb67331481 : SUCCESS in 9m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e0e56ae42075417597dca600002110c1 : SUCCESS in 14m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0c43a178382243589d7c9c36b9954a75 : SUCCESS in 12m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/68746d275bce401286b9a549c5a15555 : SUCCESS in 29m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/950b956bd0d344b7b483a176d9e7f5d3 : SUCCESS in 24m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6c50fef19e1e4866b3b31f650010884c : SUCCESS in 13m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/17f171d590204d21a0c9157039a4cbef : SUCCESS in 15m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7844620f305c4aa8a28eead42073561a : SUCCESS in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d19f0888941a402c916331e75d0b6fb0 : SUCCESS in 20m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0a6af2ae0fc4ac08fd6896efd529f19 : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5fce2100a78c479c94b3f8cc9b6af658 : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8386721d974c422ca2525b67a047f71f : SUCCESS in 9m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a46553260ade44399ce7687927935f40 : SUCCESS in 1h 02m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d89c226592fc472e849ad18038f0e78e : FAILURE in 1h 15m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d007cee79024241b00fe08ab0f05aa1 : SUCCESS in 1h 28m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/34bc974201494db1bcdb78f23a9e0f0c : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":12},{"id":"36765ee94989c49b7d066dfc12ef55ffc3ab7640","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-23 06:29:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a0fd39be87cc470ba052487976f1c7434647b44d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 09:00:03.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1acbd88da4345d98fa2a2c0a070b570 : SUCCESS in 3m 23s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f2af6c449a9e473c8a7e1a644890eabe : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c80d8357ea47e6a558f01b904c1a1e : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33d007569f7040fe8b062d317ee06a8c : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30c64513cc9744cab32ab510cf3464b8 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/175e3848e48b4b589e8261ea23541e22 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54b6906d53a54ed8846424262ed7a046 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccce9eef0baf4095aba709da82774d44 : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9d66a4078a9c4a0fbf3a4be62b4dc572 : SUCCESS in 8m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f834812f2aa4bc6b5969e9319fb3980 : SUCCESS in 13m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c8344a14c6a447e4bbd859b32e4c62a7 : SUCCESS in 12m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b986a95c4334d63afa3a372d8c3ef0e : SUCCESS in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a79ad20bd6074673b4c9406332344a44 : SUCCESS in 4m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0bcf5518e8d04211a4f86b0ce2d847a4 : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d669ea66c6dd4ee7b5b419ab8410ab72 : SUCCESS in 10m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cc3fe075ee484fc480852ccccd9c9df7 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cdf2922943c445b9bf99541a528af26 : SUCCESS in 6m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/10191c6bf7d64639ad7d82d828cb282e : SUCCESS in 11m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e0510cce1a845b8b2e4b440c61a2870 : SUCCESS in 22m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fbf3ada2c1154e9a8b2e1805a548f999 : SUCCESS in 35m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8bdf989e38ea4787b87438efba060f07 : SUCCESS in 34m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bc941bccf99d4661b8372867c957c0b7 : SUCCESS in 14m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d220566667b45faa15c8b55d6cd4d29 : SUCCESS in 11m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e355f69df07440b485cf5c507a35c10f : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e423245fc8584b79afabac4934c9caf2 : SUCCESS in 23m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73bdddaf7b534e119c756f306194d744 : SUCCESS in 5m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d40921aaefe6477da8660146526c1368 : SUCCESS in 11m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ce1d7aa454f348e797b5e0cb069c9f3c : SUCCESS in 10m 31s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ef769640f21a4f0283f2194a5693747b : SUCCESS in 1h 37m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef7d02a8be4748ecaa2a4b79cd264fa6 : FAILURE in 1h 26m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/baa51869834b4d06bf82b27fa63f8ebe : SUCCESS in 1h 59m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27e2c37fce8d4868b40bbdc4dd30e990 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":13},{"id":"353a026a3846867525a776767f77c212625bf9ea","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-24 06:15:22.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"af31da22379046a91618ccd8f3c946449d4397ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:55:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b087ae9bd8b41dcb1453c849676106d : SUCCESS in 3m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ebf64a7270844328ef3129805b80c31 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791df646f3454424b20daa708791d9fc : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/689fe46754294cfabf68b0d9935419fc : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f012e65d084d25b185c67d1e95e729 : SUCCESS in 11m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14f80c3522764e3f925a7012b98454d8 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f3ce20216eb42659ef49bb558b5fa24 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79b587697888452c9ce0a75b5fb2c7e9 : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5301a2bd81a6458c9b2d63d102d6829a : SUCCESS in 8m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f773cdf64b04a17bea48517cd9b4298 : SUCCESS in 5m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f72e7693f67444cfa18cb80acef6c2a8 : SUCCESS in 11m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/75cb5b69a9ae478daadcb12fd13083bf : SUCCESS in 9m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1544d9bd4c91490784ce4a07e0b0816e : SUCCESS in 6m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b65840892b3e425195cc081628fab924 : SUCCESS in 5m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4b14e2f0070c4b768ac0913baaaf4d1d : SUCCESS in 11m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55c74db4b7fc49edb0ab68367bcd9c48 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4bc97a1f54df468db732308a077b9221 : SUCCESS in 9m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e102f993a55f47df8d40160b2105d31a : SUCCESS in 14m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9aff45b1f8a74bf7b17db390ddb5df9a : SUCCESS in 17m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/40d6f12856bf49bc8a0e26c246493e4b : SUCCESS in 25m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/31f0092ccdc343ff9adab0a0765912b8 : SUCCESS in 22m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f5b6b02e26d24d079f1d1f640c315456 : SUCCESS in 21m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db4170f7bc33488e87340c71b80180fa : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/397caf04457d4af794d9310f684f2afb : SUCCESS in 15m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c9702c750e94860970cbc4bfd98e5a7 : SUCCESS in 22m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/011d17e80aa347e9a0b3b8aca525c9ce : SUCCESS in 8m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/921fdd7be0c046bdac38096d188c279b : SUCCESS in 11m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc56c8355900491c9780a960da947298 : SUCCESS in 15m 25s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8fcfd050eba941dd8fe3e7852c38dbe7 : SUCCESS in 1h 12m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a10bfc6a8e12434eb95e69f5cbbc6cc4 : SUCCESS in 1h 24m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2841fcabc43d4c1b8bb8e24cadb9b5ee : SUCCESS in 2h 01m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d1c31eef66e44806a9e4a5ff4eeb7c7b : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":14},{"id":"5873697f94299ee1293fd1770ce86d02815e45b7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-25 06:20:43.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a1d42f0f33a370b7b72a88d3cc6eb55dc882c373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:36:12.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b57667a796f450991c1998f68ae1276 : SUCCESS in 3m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/01ed1b153f3144f9809d6332cec9ca95 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e69424c03fc4e19b78aa121064a1daf : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a25fad3d1433427089832bb00ed9c105 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80fac67a1ab44d28b714004b1bbc5447 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5618e9e002dd4867b2570941abf5bc0d : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da282fe379c7400c85511d618b735987 : SUCCESS in 2m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fb5524cbac2440ca07d69dae161f796 : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cdf2bf112ca143419289261901bd0a45 : SUCCESS in 11m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2eb45f93f6d491283168efb6283a91c : SUCCESS in 6m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0f1d3683082345198c090900d76b494c : SUCCESS in 9m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/668bbf29cfce44b0ad439bc791c9b958 : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8026f900904145c6989b4bb4257734d9 : SUCCESS in 6m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b90cd357dc9640648ab67fc7a56cd840 : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8d97772a93cb4519ab6477e65b352c15 : SUCCESS in 8m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfcb25ea52184c4eb45c7885df830c80 : SUCCESS in 7m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3bdc922f4b64836bd02099779d6622f : SUCCESS in 7m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e60e47947b0b4924a2cdecdfc0036059 : SUCCESS in 17m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/94eabc79920e486182d1296e172f992e : SUCCESS in 13m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6a48687533074670a9b76f3c99c3e46b : SUCCESS in 21m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c97e295893eb4307ae6f445263f922aa : SUCCESS in 25m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f5db072987774c5ebeb243ee03f1f14d : SUCCESS in 11m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/de030fa66b424e6f813f529ae3973455 : SUCCESS in 13m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2b31c16e535c41909add16ec529499f9 : SUCCESS in 14m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26fffc78c3a146c394d732e66b72691c : SUCCESS in 19m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37c62d322a684a36824885025c90385f : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e5b02b4c45104bc4ad7be41845143d81 : SUCCESS in 8m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38815a5187914c50a5555f5f8b2d4b6d : SUCCESS in 9m 03s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/edc2556031504c3f9b775bf823a3a05d : SUCCESS in 1h 00m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc2043b172844ba18fe879417847432c : SUCCESS in 1h 09m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d3b850239a241d99df40fa44564cdba : FAILURE in 1h 32m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f33ed47ed0f5462c8b5c8dfd961510df : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":15},{"id":"b5e3b839d116e11e2798b1d23e693861c81b4c66","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-26 06:32:53.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e5895fc407b355d431035abd5d4316fa99077fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 10:55:40.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3efec868585840b8a5fc4e5f6eb1bab4 : SUCCESS in 3m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f318835a59dc45b0af979d798c0f68f6 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e86e83869af4388b24ee8f59f5f0700 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eaf8f3b38e6240bd91ca2ea696450369 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ba3a910b7b941668d963b37d2e7d3e1 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df49f461b5e74077bb217b87b0dc941b : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/586d5965f2fd4c608baa8a0124d7cd0a : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25a563c726de40b9bd8e3fd5595f9e58 : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ae6284a7f06a45c99dd09950b38161c6 : FAILURE in 9m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff1f3d7923584bad9bcdd85c033fb99b : FAILURE in 8m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6769198ac46a479a9c31d6df5eca75bc : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a383d1c0402424ab0c3b9f6b72c7aa1 : SUCCESS in 8m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d2b8109e62c84c78b05eb12bbac27c63 : SUCCESS in 6m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a9898889a2a0463f8ec4f44726ba8d56 : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/427072e85ed34de1a897357f221a87dc : SUCCESS in 13m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8981f5b477c496bae534eccffb87601 : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bdfad687537842ba895384754969c0de : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/39b8aee5bdad45c29049793b6446f6aa : SUCCESS in 17m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4935b10849eb4006934f664e4a55c974 : SUCCESS in 23m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f55e9d9de45b4def858971301d247283 : SUCCESS in 26m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ad42d9632334cfba60ee81e0a5a3be4 : SUCCESS in 32m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8834a537fa8942d1bdf1ca718bae3f47 : SUCCESS in 15m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f66cb8dbfd234058aa1aea39bbc62b2b : SUCCESS in 17m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/035b9062e7034c54890bb80724f7c2bc : SUCCESS in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d57ba24dd1b46448f8bb55fe3df3407 : SUCCESS in 22m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d40a66f9c32f4dd5878eeac82a84ebda : SUCCESS in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8062d450733c4c6cadf416acab7fdb6e : SUCCESS in 8m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6d2c754e0ee64e56b011b999fdb41061 : SUCCESS in 8m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c876bb680dc240c38daf4f258d058d5f : SUCCESS in 1h 02m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d855e99a307440979e1b8306c20b1e5d : SUCCESS in 1h 18m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2b4f51b281aa493d8afa946247917487 : FAILURE in 1h 46m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/167d9b3943f344f7bea7a4bc08fa1dfc : SUCCESS in 1h 25m 17s","accounts_in_message":[],"_revision_number":16},{"id":"12e7553b4c7353599473422992d4827f25243fa2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-27 06:17:16.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"723fd5db0230f7c65f9a97d3469fbc94cab6566f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:01:31.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668780ca39f44fb2bef4c9c6a4465738 : SUCCESS in 4m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8e4632d1a904435ba528f6ce5aabf1a6 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/456be2deb85f47fa93c50882ccadb779 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf88abd4b8de4763953b753fb7ac0312 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75eb4b1cabd4e7291bd7055427b21ee : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f6c1b0c3271492dae4adabadd25c282 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d863163a9094595a1e5e2feeab115f8 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc2c544b66c741e8808e1a4d7816cdc3 : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a196c26d68724c7787f39230e0c33045 : FAILURE in 11m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/485f10ff15654a86875d37b019aa101f : FAILURE in 6m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/be360b7b446f440087f3a640f5412fd0 : SUCCESS in 11m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dfd5dfe8fae949fdadc996ae93284011 : SUCCESS in 8m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7975f80944604622ab3ad9023b54ba8e : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/12b872960ac04138ada7b04d682f4559 : SUCCESS in 5m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/93a2d467a8fe4cfeb1cffb8928e85910 : SUCCESS in 10m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ceafaa489e84dbf9a22f6ebaf3fe872 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6b727836c0f48b6acf75f1419a0f482 : SUCCESS in 8m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e51fa960c3f54fbe92b81c1c1ef6ae6f : SUCCESS in 15m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e063b28752134d358d3c5201d26b2f72 : SUCCESS in 14m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f8891a288ec34811ad3568f0c70a2e22 : SUCCESS in 22m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7a77b045d3842408519e51a22fcea48 : SUCCESS in 26m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d7e2d71293f14229af7eefbc8599d435 : SUCCESS in 12m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e27baaa31c7b44a297211e5cf1802f7b : SUCCESS in 13m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9081b3d2725f4dd4902dfe64ee09a980 : SUCCESS in 13m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/894a4dfab90547ef9198f0ce95d9443c : SUCCESS in 23m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/532e886fea6944789e0f2263a81d23ef : SUCCESS in 5m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc996aead04143a7af9872cd4e4ede52 : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f6e81383f2974bf78878058073b9d0b3 : SUCCESS in 11m 15s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a062a427092c40ad953adb9a56ca167c : SUCCESS in 1h 02m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23010a3d4f4347e09e8d07a81b784506 : SUCCESS in 1h 15m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ac490994e672478f9ca5f79cf98d1a99 : SUCCESS in 1h 49m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ca418d1858a4fa783300f3ee4dee09e : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":17},{"id":"60e16dc347f6b2263d43c92b5522e45897eeda23","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-28 06:18:43.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"80705e794eb070d7e723a4bb0b373e8ba0cd10ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 08:29:38.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3032376a382540caab9e32ffcbef586d : SUCCESS in 4m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a8f68746878e46db8f204c1659fb8fca : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118f5d3472a04bdf994631e79073d91a : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ade802e767d142b38cb7d09fc7e391e8 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6676519793f44a2cb9201cc97bfb38c8 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c941352b24248a09b893e2c24e43610 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bc7aae00e9c4732a998ead089192920 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e8b8b873f4742adbd06b8d6a8293d1b : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b7903e2c079f42688f8ba155749ea29b : FAILURE in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c452a3ea6d3c421fa5066fdaa89218c4 : FAILURE in 6m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/15b90c62fee24f348e1dfbacdb4f664f : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0b96215a7b14133bbec3cc5bdc085d3 : SUCCESS in 8m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/337e78b6ecd24b2f9008dbc9f9379170 : SUCCESS in 4m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eae31d74fee54a71bdd573d8e0f60d12 : SUCCESS in 5m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/56137a84efaf42d4a2d8dd3296f5ff16 : SUCCESS in 10m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/02465ef1bb6e4da49c5e02e2a547b6c6 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cedced22fb443b2a6a5d2beb0cebc2e : SUCCESS in 8m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/149616952988410b8559af7c846648a7 : SUCCESS in 18m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9e401bddd4c43baa4c779bbed0b7d28 : SUCCESS in 13m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3d792de336b64404b690bf35e103d137 : SUCCESS in 23m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ab209bc71a8c4b1cb193fb45ce2cf12c : SUCCESS in 24m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/640be70e6352485b99ed4a38c94764ba : SUCCESS in 13m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a922bf674a8945a6b7ce4462dd683bd6 : SUCCESS in 13m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e4be1530cc17459cac56eb028dec69cf : SUCCESS in 14m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/400bd06508bc4506b65dcf6d4309ece9 : SUCCESS in 23m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91cb811f4be74f86a34f26ecf03ec921 : SUCCESS in 5m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f9de312b75cb4c4daf19f44c58eec011 : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7664316247fc4ebdab78a44af4ce6ca8 : SUCCESS in 11m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e7735e25de6f4a51bd621dac596f05d4 : SUCCESS in 56m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8f3af4f2e6d401abf72d37c68df8c9e : SUCCESS in 1h 05m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3be17c2c75514426bd7490542dbbbb3f : SUCCESS in 1h 42m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eba98a7ec80b409a99f7ff07248bee11 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":18},{"id":"f99c944b2979252f6e80430c8736b081764e9434","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-29 06:45:58.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"809cbc6afb56229375d4da38c5e1e1f833d5d884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 08:25:23.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbeb75dbfd564f4a8bef4aa1d923d2ae : SUCCESS in 4m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a916ef0301224f3f80de42ab717db0f6 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3531fce6d26e4a59bdcec652b0240a83 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3de4c32f9ac4004a6c8155035c0dfae : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e765a9375ec4cef846ed1e7670575df : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c603c26c17074089a28f8952cf95ae43 : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0372c5df55c3468099ebb9684df56d65 : SUCCESS in 6m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51d452a042284acf97f2337ef7814036 : SUCCESS in 6m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/583ced640cd742608fe8b39e1c957890 : FAILURE in 24m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f44a788c11846dfa6ee8eef6c1a701d : FAILURE in 7m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/59ad41b745c74ea49c1426903fc92acd : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a8416dc42b2447b3aa43c37df14a3b9f : SUCCESS in 8m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1111e73c82a84b89ba1228ab24b54888 : SUCCESS in 4m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3374389e970840509a46b2177e531771 : SUCCESS in 6m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/38cf6d53fe814a61a46e349b2d84cc0c : SUCCESS in 10m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dc9e93833e934c5189507b93fd06dfb4 : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4c065d634ca4a35b8e8ac567c65c145 : SUCCESS in 7m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/71bf92c73e6948cfa917635d3bc3812c : SUCCESS in 19m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c2064557d5b24f41a6cc7754039d8c56 : SUCCESS in 20m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4f9f8ec4205a48af9b8671c618769e6a : SUCCESS in 23m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2ea5c28ae18b45b38604ecdf795dbe06 : SUCCESS in 20m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fb6504cef1764bf49b404187e9f4579d : SUCCESS in 13m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/105af2f03f464e3aaf6ea41b7e64a93b : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1740ccba5e814a938c54945683f72926 : SUCCESS in 16m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4f90fa4838d4684bc85dc3351d8033d : SUCCESS in 18m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d183aa412fd24541875ca1f49e45741d : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e3fcf07ee8cd4889a5ecaf2f26f0f70c : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9beeb543b9264934a043113f1ffcc122 : SUCCESS in 9m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0277f508f2e144bc887d49483684a2d0 : SUCCESS in 54m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6310ca636a843dda9d3f43ec4ac3546 : SUCCESS in 1h 06m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a381e2cd40d44e6886aefb6ab88927d5 : SUCCESS in 1h 34m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/781e437597a84ae1ae76260e87cecc8d : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":19},{"id":"5b13cb5d9f0c37c809123171e521c0069ab1fdbd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-30 06:18:42.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"57ee1d1ea3064bc6a2f6d202f8136b8b2c80f13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:04:17.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d04454e471649d28d2860aabb5a53c1 : SUCCESS in 4m 20s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/aca573bd10a744f785f2d3b101309575 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd64838951d143d8a747b60686c1daf9 : SUCCESS in 20m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/341b048041e440bab2cbfb570fc7462f : SUCCESS in 21m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/281085778f9f4d2aab5982d3f47a65a8 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/172e96caeb4947078d6f11b0680c0612 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3e386aaecb142d1aa03827158a3560c : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/850857f64270478cb86c45bbadd83b12 : SUCCESS in 2m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0a44d0e60aac475fa4deb2da503f897c : FAILURE in 20m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0b1fa204cbcb4389ae6c61a95c488cdb : FAILURE in 20m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4123adfc242b4882babbe00b10e39994 : SUCCESS in 14m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/28c5d34ec6ea4efd994b9689037dc6da : SUCCESS in 7m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/155e90c359e544bbb65dd45b7ae1a94f : SUCCESS in 4m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e763e973bcc4a18a2d73404de33e665 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f5681adeb00344fdaaf030ada9d5fbca : SUCCESS in 10m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3cb7deca298648ccb96044d54d485fee : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f393843bcce44178aebcd884a90e0309 : SUCCESS in 7m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0b334ea79375438e8b81f59948471dd5 : SUCCESS in 12m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d31d7bbcf5854385b1a42e88748c919d : SUCCESS in 14m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/af95d7f597da4ee9bfb77dc4dab9264a : SUCCESS in 21m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9e62de6b791041ee8a95566f7063f3d6 : SUCCESS in 26m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/841705f6cafb4d43ac634ccf1a5cdda2 : SUCCESS in 20m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d55a4db0701f4bbe8f5d5f1088ec8c72 : SUCCESS in 19m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/130987a0b163408ab5f60489a7d6be61 : SUCCESS in 20m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94769b4924f84defa240c18c20e50196 : SUCCESS in 17m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/861c305d161748388695dd3b3848381e : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/47cd760f56e347a9aca08763284523b1 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/239cc79895a341ff87b82b2bf7800df5 : SUCCESS in 8m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/23b80f6fdbbc4beaa010947d15105de0 : SUCCESS in 1h 03m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2431117385c944f3a4b6881ad96f6648 : SUCCESS in 1h 15m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1277e653d0ea471790aafeba161ee08a : SUCCESS in 1h 53m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f866ddc3c7ee4785ac6edfbf5b609e3d : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":20},{"id":"3fc08c31331b1d72341bf619b7d9cd6e7fb6917a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-01 06:19:47.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"22633a84ea983a3143d1bef8a8ef37db25a7da90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:04:05.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68bf37a0326c4c8096bd16cc1227053f : SUCCESS in 4m 16s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/99cce0e2a0644f99ace3fd4215c77946 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02f96f7b6ef4322b16ffee4aa8d0ead : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d048ef499cb14286bfdc64e69a8ffaf6 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/034d23e5f09049eb9d4e29d35d65cf8f : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/688e6beb40ca49829d53fa133bad052c : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00ca9ff83a944cd092375ee1ed0798da : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df3155e1db9c4b84aec59079705e40e5 : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6471e334e3be4dedbe757b7903cac378 : FAILURE in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88830567c78b487a9e0f66cc23c760a1 : FAILURE in 6m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/72cfbe7af8ea4cf282a647b8bca4d361 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a3a9cd7224c442981ea3cc3545955b9 : SUCCESS in 7m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8b66bc2c57b64e38bc6f2da878ae0000 : SUCCESS in 10m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3fdb30c900d4ff890a704de6d157b8a : FAILURE in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/20b0ffa85b604f82a4d1f0d52557e201 : SUCCESS in 10m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c54e0697c47742f08f9a03df98a1a0c5 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62df64af47934a23abc7941dd686eb45 : SUCCESS in 5m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/36d99d64d70f4a6fb0728dad8f8d969b : SUCCESS in 13m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d20d5866b21d45fb96295ee0ce2d58fd : SUCCESS in 11m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c2742534bf0545e790c984337a8387bc : SUCCESS in 29m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/17c3cb132ae44dbea98254dd598c4296 : SUCCESS in 21m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/07a6b265768446469946f51430342147 : SUCCESS in 23m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79fe830f184d4bb4b358845e4f4f21ba : SUCCESS in 23m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/946d4aea568342f7a10bb5eff745edca : SUCCESS in 27m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21e500682741494380b5386a08b9bb7b : SUCCESS in 32m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e27065944ed43e68a3a80536acc5eac : SUCCESS in 10m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d65bbbc745aa4d65ab2866e976fd3618 : SUCCESS in 13m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b92a76a339c14bedbf3f612e678de615 : SUCCESS in 14m 26s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/f070a4e8ef914c739d197961443f6883 : SUCCESS in 1h 01m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/77dbcc870eca45b6ac1f6427893234e6 : SUCCESS in 1h 10m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b1cf1c03538845a78ac65ebdd14d0666 : SUCCESS in 1h 52m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d0b82a4576eb4b7c94e5357196cfb354 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":21},{"id":"0a8110fa33592624e8c5838dd722895eeb989ebe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-02 06:22:00.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"2b4c6e1a701e97ca075fee4e1a39f1103c82e277","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-02 07:58:41.000000000","message":"Patch Set 22:\n\nW/o this change our \"docs\" job is failing: https://zuul.opendev.org/t/openstack/build/6fe7c7d3e60b40458d2a98f3a293f412/log/job-output.txt#840\n\nThis is because we need 2.5.2 for amqp in upper-constraints.txt. Any ideas how we need to fix it?","accounts_in_message":[],"_revision_number":22},{"id":"abf93c67ae505ebc912759653d0509abd884a3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 08:21:48.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e623885977ce4b49a9b3bc2326a07757 : SUCCESS in 5m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2ad79bcb3e32436f80e48282493d229e : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8105ecbcf74b8095ddc1482a0b549a : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ca454a1af8041379374a0043a7ab29b : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91c8ebb7fe264931b38b2075b407e243 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/495c54c74d4c44deb80b301b5bfae1cc : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93a59954024a4eac830be463894687be : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12aafa22d9be493595499d9ed3e40eb0 : SUCCESS in 3m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d1bc8528fcf54af0a00312b51f52cf25 : FAILURE in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eda3bebb2efd4417b2dbce691fefbee6 : FAILURE in 7m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f46983acee0146d2a93028c0f1d0af92 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a2e36bda8fad43fc9e219aae31ed0c4a : SUCCESS in 9m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b924da3da53e4069bab41cd12a81fb55 : SUCCESS in 5m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5dc9c82e7074f8ca2f816112059889b : SUCCESS in 5m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/91e6493bc577405ca510adbd88caba26 : SUCCESS in 10m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7647d88732794110a34515ae71d89fed : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d98e2701f31b44a589958fa2e2019f68 : SUCCESS in 7m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2b2a38523e32492ca678c1254b56e3f2 : SUCCESS in 13m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3519c3f671244b0aa746f3eb85126a40 : SUCCESS in 12m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e18d75188508436bb477b4d6c7056102 : SUCCESS in 22m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e375d2e94bc04e67a13802e6852be0fc : SUCCESS in 20m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/77a685caab634365ab76a4647c472b37 : SUCCESS in 13m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b88ba1e8710142c88dfbe3102958971f : SUCCESS in 11m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5fd1a96d1314d0992daaadce8cc1499 : SUCCESS in 12m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/654a0cec98f34df5991d9a7c4b3f6ee4 : SUCCESS in 20m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce2b16aa8c804c6c80e165dacee46ca7 : SUCCESS in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5a2a61a3ed1c4ea48a9fa50738de0c33 : SUCCESS in 9m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09dd78ece4cc4737a7b7f9d06fdf5537 : SUCCESS in 9m 33s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/dd32c2bd8afe49f9bed0046cf47f39e4 : SUCCESS in 58m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a04fb3ebcd648aa95415634258bf922 : SUCCESS in 1h 08m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d8e87487851a48bf8a1ff22f06efb883 : SUCCESS in 1h 52m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1d31461d71274b17a802ee6bb9a04a54 : SUCCESS in 1h 24m 51s","accounts_in_message":[],"_revision_number":22},{"id":"3f46a0a6d54d4107403e1ed9020af7da8e5f78f8","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-10-02 08:59:36.000000000","message":"Patch Set 22:\n\nSame issue with the vmware-nsx project","accounts_in_message":[],"_revision_number":22},{"id":"9d23890b3d4c11388280a758a73a79d9e4d823ad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-02 16:08:35.000000000","message":"Patch Set 22:\n\nif you NEED some lib to be updated then please submit a review for JUST what needs to be updated.  You\u0027ll also need to submit a email with the [ffe][requirements] subject with reasoning behind why the change is needed (as we are in the freeze period still)","accounts_in_message":[],"_revision_number":22},{"id":"6a33533bf73e31eeb5a81fbc6cef38fb7e52e708","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-03 06:14:16.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"d3590428d2a969f337a3ee55d6a22b3e73141b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 08:57:15.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203eeec8b5a441a5b897432660c81467 : SUCCESS in 6m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/abe71c474a3c4373828527bfde598799 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de1cebf860d48ddae11207b54f4d7f6 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89270c89f5e24bd791ba3e5579cbfb49 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb5af76bcdd04b22a82f78541e4088fc : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1593f4e61c5e42049dd49062db46df90 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bca3f3962614bd8b91ba7f7b26218c0 : SUCCESS in 6m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2924d91edb47467d9d060b19427a457b : SUCCESS in 2m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/eb22ef60320b44518f8ebdafcdb466d7 : FAILURE in 17m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2f4c111b15ba4db6a520f15f240a5cf2 : FAILURE in 6m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ee24a70f01c945fe9cd51b930c3c0e08 : SUCCESS in 15m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b06c6023831540b79a59a5ada22c6a09 : SUCCESS in 13m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/850778ae87864833a51b96155d0d6422 : SUCCESS in 7m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/abd7660a4e814a13bbc0a5118e3ce7de : SUCCESS in 3m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1d8d183e3a124ae5b93246cc2f8e2b15 : SUCCESS in 19m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4cf3378f315f4914a13636bbb4dd2ce5 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d51f310ea0c148069920be1d3a40e4ef : SUCCESS in 11m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6a38ff5ae55844218ad61e44f7c5528a : SUCCESS in 22m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/956bfb28bc0b4cdb805553ad9387b6fc : SUCCESS in 12m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/487cfe7cd83849d9a93d92c75f1da986 : SUCCESS in 23m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a1a5fbeedc64d859f0b7556f04e8ec5 : SUCCESS in 21m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7982b1ac41f64ee9b7934112a07e736c : SUCCESS in 16m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a989daaa0804b9baec317ab96764fad : SUCCESS in 13m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a56c35fff184a8c87624e37e66627b6 : SUCCESS in 11m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e824a3349964603a5d543555cdfde44 : SUCCESS in 20m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b50a44da7094b729eac34d722b89d79 : SUCCESS in 5m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/189a59dc34034cd2a9e3674fb4252755 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/566ef2f17278496b8cd09f49f244e631 : SUCCESS in 9m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3bf6e2afb03f4d9581cef1c5dcc255cb : SUCCESS in 58m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf1d77dc175942439ceaadc5fcdd7e5f : SUCCESS in 1h 07m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9dba0895e33a451d9acd623c12c37554 : TIMED_OUT in 2h 11m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/647653c099de48cf86837fcff0016125 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":23},{"id":"2760be96ea57a8e3d923e56f27cae2e2f271b0c4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-04 06:12:27.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"b8affd8bb7c390b8b9cb0d6b0d9e673c445522f1","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-10-04 06:29:08.000000000","message":"Patch Set 24: Code-Review-1","accounts_in_message":[],"_revision_number":24},{"id":"6cbc1a6c3075ee5e6130ce9e834f2dbf1f0edf52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:40:52.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8ce5f14b714e67a4042cb8a522de9e : SUCCESS in 5m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8f84a7821cd14184828274b0dcd3c8eb : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3cc0691f9ba4688b6b98cbb514d8ca2 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7135ecea1868406caffb67cab3f0ad99 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8002bc1a391a42debae84423f719d8c8 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b483f055501a47078c502bdaeb02ab1f : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/734200b5f36f409db6715bff22d1b12b : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b4b9a7417b44400bdf030199dfdf3e8 : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8eec980b75cd49ae93f16ca71c1c8fdb : FAILURE in 9m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/24fcd00686e84e3f88c2bafd032e7b07 : FAILURE in 9m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/035e6a4f00df4212a984577d03b416c1 : SUCCESS in 8m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/297f3de271594b9cb99a9b9dd665b1c5 : SUCCESS in 8m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/821ce5122bb14c98bfbea3f4ce8863f2 : SUCCESS in 5m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/55109bbf880248b59feb27337a0edd92 : SUCCESS in 4m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e2f8207bd2e14cd5b15ff92b01b00947 : SUCCESS in 10m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/59faf438ea15458789fbfbc821420d85 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b86e89f1e044014b7ba0fc67a8d4afa : SUCCESS in 9m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/380a1d0727d3403db564f847c177def0 : SUCCESS in 24m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b13478d79724486695e0bef55fa4e68c : SUCCESS in 17m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/156a7e56a901489eb0c0c889df5f2230 : SUCCESS in 23m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/67184e4162934bcfbafc64e7e2582f96 : SUCCESS in 25m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/babce1a96fc74917be7ec32f10c311a6 : SUCCESS in 13m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9266ac83d5764b0f960225794e9946cf : SUCCESS in 12m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3f1bfdb9cf2643839e407d0e51fa2131 : SUCCESS in 18m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/440841fc9b6346f189cb56ba6f1d7705 : SUCCESS in 20m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be9c31355ad64a4292b27e45d56c40b0 : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b7a6bdff4c3742a6a61d227e6be72e34 : SUCCESS in 12m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/82676f70d7944940be7d695eb421b50f : SUCCESS in 10m 43s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/83510e8c12e44a1abdf24702aec8bc5b : SUCCESS in 1h 01m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7b932a5f1c648e9b6f986eb9b8eb3fc : SUCCESS in 1h 05m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/995d90fc90bd4189b148a577f29cd079 : SUCCESS in 1h 47m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2ec80afa1544436b833736c583952e5 : FAILURE in 59m 37s","accounts_in_message":[],"_revision_number":24},{"id":"aa6c219376d8de60953d9b35ef0096426aa572f8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-05 06:53:07.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"74caad51dd41c190f44ba8e5c8f6f1f6e8f9f0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 08:39:31.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c456bcb8a584e0ea59a1bc6ab78dc25 : SUCCESS in 6m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b05a6991a1214c019cfc75909e19597d : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748c59d028c74c1b80a9b0d3e0db437b : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ccffa8644ba543b78a45a6ef16a68de2 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8870216a1c794f4a9e31f68c5cb81ae4 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49d8151f71234d5baaa830739b35759e : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/106bb7848ac14f25a5d1472db876db7d : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a10fd442c0f54ef4b88d89ca5fee3a7f : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c92d19418be9417db8c83de874b70d65 : FAILURE in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b146be0f828341948554b209d8bcafdb : FAILURE in 7m 35s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e87064697d5546108aaab39c4270a52d : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a7c600f68ca48e8aa4d035928a8a823 : SUCCESS in 8m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b46cf107fb914710aa7ca70c51f3e270 : SUCCESS in 4m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/466b61a6641546499299f68d92ea7a21 : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5ccc5f52fb3544c69c299bd1220db3d0 : SUCCESS in 9m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ee396a3a4324c778b2e6de668082d66 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55603d24335d4aac95244373321bf2d3 : SUCCESS in 6m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f3f5ce11c6c34974bc6f693eb79e8f89 : SUCCESS in 12m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a63e5cedbe5a4a9287593f2aaa18c3f3 : SUCCESS in 19m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/27ac9444529449518985830e9099a535 : SUCCESS in 26m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0e36c47cfd4e43fda813417369db8c00 : SUCCESS in 22m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/772d75fbf7e04e82ae87fada1f0b1d5a : SUCCESS in 15m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b976f60d4b76442a997710ef2e8fb4e9 : SUCCESS in 13m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b4f224960f534fbfa85cc971a8656b8d : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d24811ddd2254e1692c09bafd4f6276d : SUCCESS in 17m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/080497b90c564163865597bf26077a30 : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/50e89673079d44cabbc201b94660cf2f : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7ec661b227af4f2b87067d7db377728e : SUCCESS in 8m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/697d964b797b40458776f922b0c10244 : SUCCESS in 50m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ab988fbbcbc4364aac0bbc470207240 : SUCCESS in 1h 07m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f9c9791b0f1b47ce91d3a4fc80777128 : SUCCESS in 1h 33m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9a811d7421ff48d7902cf3646f58274c : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":25},{"id":"31fc51e7c988d85aea787dd67bb19c2de32af98f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-07 06:11:51.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"249f3143abc856b57d42882b1c30d244980fbb4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 08:16:20.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c025fa5b674b76bc865c39f8da3389 : SUCCESS in 4m 06s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cceabd4bc85145c781178e9f9d79c712 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080d786dc935411bb4eb235aecd521cc : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48c57d6674434b8889d74713387b8368 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c3fa00500024433992e39c9190eb249 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fd239bec0f94e98ab3db3f0c835a180 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8e66e6ee1074a3189a112db3031cf37 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca0ce0a8f9cb43b98c7f79c6e269aee8 : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ba284e1ea9fd47ac9a19315b93edc701 : FAILURE in 7m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/34804ce2a90c4fefa17ff4299349a027 : FAILURE in 6m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d1cd41ed8f594c0da4b227a58cebe979 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6557409a645241668d9041251604cbbc : SUCCESS in 8m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fb9242ab17f54189bf836391b2099eab : SUCCESS in 5m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e33b8fdf43544d8b8f4cc70862ba5cfe : POST_FAILURE in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bcaa6c322234410299ac2a2b6f0666c6 : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d5cba458b1914779a14742abbd6fb448 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/712eb0788f74447482eaa65fa8a02647 : SUCCESS in 9m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9927fb54c61e486e8384ccefb4f60e8b : SUCCESS in 15m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9d7535228c9846c49c9d08f050e11afc : SUCCESS in 20m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/51d9671a82c341bbae428dd57d597357 : SUCCESS in 31m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3fb525bf1524c96b9be3de3a98d308a : SUCCESS in 21m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b7f1fdb7e2d54af78d17334d4c24085b : SUCCESS in 15m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5c732bde800142659ad3c9749eb2977b : SUCCESS in 11m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7a46fb21e6794772852795127cea8032 : SUCCESS in 13m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfdd91a9e65c43dea5652bb386d91158 : SUCCESS in 16m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c68247e4df148f0b1c5dc8f7798feb8 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c12a90fc805a4cbfbc7a6cbc08325116 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6765f1422bb848dd86ff1e559e34fc95 : SUCCESS in 10m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/937dcdaa56ed4a87beecf70de71509a9 : SUCCESS in 1h 19m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/506ced428420438f9cebfe668089d381 : SUCCESS in 1h 19m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e2b705f2c914e3fa52a4b0091280359 : SUCCESS in 1h 54m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5e59514acb4f45bab388a7eaa6812ce4 : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":26},{"id":"1b8a05a7e26201e79cc716ad3349200dee31cf0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 15:03:31.000000000","message":"Patch Set 26: -Code-Review","accounts_in_message":[],"_revision_number":26},{"id":"4535feb0f82fa9e3bd95a519c4534a65b1030b23","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 15:40:11.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"716dde3d3d51218deec4a90765770ca86e10e952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:14:49.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ddbba84170415eb7ebf2853e7c70c8 : SUCCESS in 4m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ac6d102197934ed4810f99f6a0daad63 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f8e5c2a20949e8be9517055d84682b : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34d0c461b2a34723b408f17e34881a6a : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95501d60a5df4f35a7486104264fc076 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1facc739819743a99be05c69d8cf0c34 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9d87e9d9d084a2dabe742ee35c6e82c : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3a4a612ae374d6a9592a4eedf7c86a2 : SUCCESS in 2m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ce53bfda6c6a42708c4e9f5ed4a3c670 : FAILURE in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a755ff1e4be4b3da4535f3feb9cfb99 : FAILURE in 7m 50s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/194c5918c8ea4100b0aeb4aa15d6a733 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef7eeac717c34bda8582b9bab20b74bc : SUCCESS in 12m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/36c5ed47fae244159c21b7bfaa57658d : SUCCESS in 5m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0f758322e8c4d73abda15e0d4c1d6a4 : SUCCESS in 6m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9b3f2000cff84c9795b5d3babad1b180 : SUCCESS in 9m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a4dec76bc77f401c9df482e66037f9e3 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72d360c5910b40e6aa9db5a94a715599 : SUCCESS in 8m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/23e4fe0e1ecd445d95c5f22a766455a0 : SUCCESS in 16m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3d931c0abd2c45e49e73e0441b3e091e : SUCCESS in 12m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0648d6caf8f54bf4a61ed7f213b007b3 : SUCCESS in 28m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78286ae1854e4b5cab700b873d6ba074 : SUCCESS in 23m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b8bed816c4ca461a9e75337183756c8f : SUCCESS in 15m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cc92578bbdc0412bbf7d3df394a41d14 : SUCCESS in 17m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eaebc01ff44a4da4b0121d6123b0bfac : SUCCESS in 16m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8637ab87ce94c4a978b7fd7d7120c4c : SUCCESS in 21m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c876577e32ab42ed9f90dbe916bd7cb2 : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/304cc79f84c7475e84e88c79fb00af4c : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c1da56e2ff94969abe5a65f14ae1a13 : SUCCESS in 11m 14s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a9c117b0bf5a4244acbac907871e6282 : SUCCESS in 1h 07m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b66cb163df3b43fdbf9c1382f866d825 : FAILURE in 59m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b4f1e9ad21d04c8e8890c1f2a98bfa53 : SUCCESS in 1h 45m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ea07d7024cf54f5187443349437e44ab : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":27},{"id":"82d501a4de16e6df83048d9f7e7fa33092c10300","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-07 20:46:36.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"77a9093fe29fc08858e416548a45cc8769a2cf02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:39:25.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7136f8b2273243d88e4a3ec820d8ce0d : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4dbd1d6a8fbb48249ad50d55a452941d : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1109620cd8b45fbac360743308164f2 : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33acc02c86ad45e0ab1fa3b960a252a0 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35a7417028e34a05b2a2ed2256b7a86e : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35e82291a9cc4ca6809d8026befe1553 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/299606a662794151a96ef5159245b8c4 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/31fdc4a4cefc478798c2867f5b1526e3 : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d5f05e2dec9b43a6a94aa33d7afdd63a : SUCCESS in 7m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb3a6f16b8d64b439d2677ede8da7132 : SUCCESS in 7m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bc3b9b9a5d8e4f77883c62dc3986df66 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/84623ff0c65a4c5288f7730250691782 : SUCCESS in 8m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e53f4e27410440fdb3d5685a1b8c8246 : SUCCESS in 4m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4db04f3c779437f865ac089a4b75f8b : SUCCESS in 4m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/23615cbb14e6484e960402ae7a70b5af : SUCCESS in 11m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a636e89d0f1462298563c12e0fe413a : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c7b963981fef41ff94b42a228fc2f231 : SUCCESS in 8m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fc4daee6f1324e71b476107c1804ca76 : SUCCESS in 12m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f86af0c17f1245b7ace2a1947d54f7d9 : SUCCESS in 14m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d32ae574d53f45fca6dbfa6f77c4643e : SUCCESS in 22m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2c9297fdef44fafa755a0c697644eb9 : SUCCESS in 23m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ae0ac89103e84479ae7547be1c8e6f91 : SUCCESS in 14m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b1154a11c4043169431417fb8ed80fe : SUCCESS in 13m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c00c9cdf84545a697ed7441892135ba : SUCCESS in 12m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e6a9af13043440da0e5705abf68ac30 : SUCCESS in 16m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9f738ef588e4c8692ff74af4a12b196 : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a903149b42b54cefbbc51b08bf94b131 : SUCCESS in 11m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a2d094f58ce44f4928500dda2882780 : SUCCESS in 9m 44s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/13c5aa2a3112427bbf7873c271a389b9 : SUCCESS in 48m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1819a2542c1b48a6b485f40f44665848 : SUCCESS in 1h 15m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b097966dd6d64fa6ac099e4f6f9ef0d9 : SUCCESS in 1h 46m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8a333388909b4d0f9a26059722710c9b : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":28},{"id":"cfe5448fc85efe70cc5adb23767bb8647c0fe71f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 00:05:58.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":28},{"id":"5e9617136596da370d1ccf004b33d5f27ba36bac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 00:06:06.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"dcf641c54ace8750592f11d59a487b7dc1404a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:06:30.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"a24a8fa554d9e1cc0335fa4ec2217703111b5675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 02:38:29.000000000","message":"Patch Set 28: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1aa3782f3ea44549d179d1e58492d04 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ec1fba6ec245a08b099ab2b868cc22 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17e6aa6fac3c4a34a89f00ae34acf277 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f6cb25c76064dadbc1699d2c42a3bce : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/704906e033494c47b8f3e0cdeaad0e15 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bff90d559c194908b0d346252275ddb4 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df54192b40d54321804e7ddf87a74bb7 : SUCCESS in 3m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f35cafc92c724966bd9aa86efe5e5102 : SUCCESS in 7m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ab58796a763a4fc9966a55f446d4df44 : SUCCESS in 8m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/00a4d15f93b04dbebb991d97a4f8ce1c : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/07adbbc3e14049a182f76ec7d987a06a : SUCCESS in 8m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5db169dae76c45389ad9a316dbe6b876 : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7c8b2b63c9b44c41a308ea47962f0081 : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/83534e11c1774ec5814371903f09c482 : SUCCESS in 11m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17a293ade783484bbcbfe5ff41758854 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4f1999ae09ce45c89ac3390649842771 : SUCCESS in 9m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/36c68e028e9b4aa98b09f12c88da19eb : SUCCESS in 13m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2062b43b0043438dae59ce7fabeb016e : SUCCESS in 19m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1778a21d4ae748ac9b2ec86758894ac5 : SUCCESS in 24m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/07b4f71081444ccc923d79a3c7f1cc65 : SUCCESS in 20m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/160494ab8eff4d92832f34e2ccf3099c : SUCCESS in 13m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d50531ddb2684d5c80dd388ddae08168 : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5ee0689caf694636a03616393154c2db : SUCCESS in 12m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e392d9b51b254442b99dacb259e28a89 : SUCCESS in 18m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a47891315de54278a797b0a627f099fa : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a309a069d66d4c9bb21fbe7a8bf7fd3d : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73ca3dc639e442c8aafb13c82f9f8c7a : SUCCESS in 9m 17s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a3fc43d3de2d43d1b8798918d689ebe5 : SUCCESS in 54m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2f226be518a4645b4cd0d294b538d4d : FAILURE in 1h 11m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/53f32bc188d0497490e8a6a918ef41b8 : TIMED_OUT in 2h 11m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9bda5762c35f42e093f18179b52c80f8 : FAILURE in 37m 11s","accounts_in_message":[],"_revision_number":28},{"id":"f03b4e265d71dc93cf52043004cf4a323d8804d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 02:48:36.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"d5f0cd3f650bc161a4fe139f56c721ebf3cf7946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 04:35:43.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27153a4fddbd4b3fb17aeb2e0bd9a1b5 : SUCCESS in 3m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/aadcfeb75c604f47a05811ec4ff6b928 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/839eb185081b42bc8b9e011dd3dcb643 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57ba9682d1ef4ca083ac2e10099f32b2 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51b9bbd1b23d4911ab43f9c12e97a7f5 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8970ac726464f0d9081f5e9df7dc957 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6f27c249caa44ef858750938dbf7841 : SUCCESS in 9m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7434c6274ca46c0ba91e23585246e2f : SUCCESS in 5m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7369f091f4904b329677c375b425943f : SUCCESS in 13m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be377689592547128af08fb75965bb38 : SUCCESS in 7m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b693c35532204573a6c81e18f99e3607 : SUCCESS in 8m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1f4af16c1bc34f5c9104609260b2e75c : SUCCESS in 8m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a933d91a70ca46fa91cd119e50b7a80f : SUCCESS in 6m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68cd7f28e4b9402ba9f629b4113e4e23 : SUCCESS in 4m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/43758458564747caad9e07ae33b6d527 : SUCCESS in 9m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8eadfd9883eb499ba25e7b13b3696a8f : SUCCESS in 7m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9961b89651c4202a97ec7d1b759a9a8 : SUCCESS in 6m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/276cebb17ee449f1a36f4a287931be84 : SUCCESS in 12m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70608a9306b74ddab0c194b334aac5a6 : SUCCESS in 22m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bd90bc512d624087a8a8a7e3b636c4c4 : SUCCESS in 24m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/beb9c6fe2b9d49ada5934039996decea : SUCCESS in 24m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f5361210a53f40f2a0ab8b253b723e4b : SUCCESS in 14m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fb4f17138e1a4209824cd96437e6ca85 : SUCCESS in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/53a3f8478a66478bb5b592148bd1781a : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a00af53dfde94cac8193d03b973c80f2 : SUCCESS in 19m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/959d2e5810f344ceae4aa57833cdb659 : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7bcd34cef4a2496e9f0d07fdd88f8282 : SUCCESS in 9m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fa52265c97c14bd7ac4a34d8d674879a : SUCCESS in 8m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/26434c75cefe440ba4c934f4cf486570 : FAILURE in 32m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04940bf527b4451581a4c12652a1e7da : SUCCESS in 1h 18m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/87858a77867548b2886c84c03efc824f : SUCCESS in 1h 41m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27eff13a97cc422687e2e6a8398c8bce : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":28},{"id":"c5e7689b625a31d56d774a37d35dab61e53e6dfd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 04:56:27.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"ab44dcecc2572bf696d2076d273587b4e6ed6a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 06:40:51.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da742fda05d4208b41dd645fbd0c913 : SUCCESS in 3m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/eb4e7294881f4838b60dd60323dc417d : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84346fcf0e824e2e98a09f8e96fba311 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f876d0dc540140ba8d45a0a337738abd : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1f8c1ac8ba640e1a172f53245d84fc0 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e6a0ff0fe32474893a39218b9b71692 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4ae5edd734c445299efffb2f20b2f9f : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c5262515d54405eba5b272ceb926af7 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3724faf695a044a59d18aa3d89f9afc3 : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58adb64bd485432e876a0000f68bb6e6 : SUCCESS in 6m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fb9ff7205ffa4e059e2e37aaa585c93c : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/15c05241a88e4b27bc16cb45d09d1a09 : SUCCESS in 7m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/48371d369f554880b154d38544894fc7 : SUCCESS in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/774b0c6ef2f444279d580803372dbb91 : SUCCESS in 3m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/33f456f19a704622b262201593836be0 : SUCCESS in 8m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/072230d1c5a743e1b4e485c903750740 : SUCCESS in 7m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a84687076924f3e91fb19bcb9f7d34f : SUCCESS in 5m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f1feb699bbb34c36bdc77ec40318ed14 : SUCCESS in 17m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a18258d22d76484da80c0f68db1bccc9 : SUCCESS in 19m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/66dabfbeceaf40bdae3cf2f06ff07b27 : SUCCESS in 24m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/275daa35ea14481faf808dac8af85d12 : SUCCESS in 21m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4cc5621f27f649728eac62950e3b50b2 : SUCCESS in 12m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/453e9035c6c5434b9e123c7623b028fa : SUCCESS in 11m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a4874e529ed4e0d945319461301f39a : SUCCESS in 12m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba9e9d94e0314cc1ac8062fb1e9e7cb8 : SUCCESS in 16m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c4cf14b6998c4e0b9e7f484335cba70d : SUCCESS in 4m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/75ddb56e3a5947f6b7885e7ad3a624bb : SUCCESS in 8m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6aebe358f5574d96a5769748e0a84f33 : SUCCESS in 8m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4b0b9f17777a421e9cb5e24eb133a484 : FAILURE in 33m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd9aba211cd94f2da969609bcc0210c2 : SUCCESS in 1h 07m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e27e78dedf1e446089a4171bfe12d85d : SUCCESS in 1h 39m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bcba2058219142b08373e25384df618d : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":28},{"id":"5d9ccaaa70cf4f910cef4bf8d3d03061ca5dbeff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-10-08 06:45:57.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"81173fe7eb3b9afb4285625a89c4240597206e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:30:14.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a42753c5e946eaac67b5fc93154729 : SUCCESS in 4m 45s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/626868bf052b4db08fd6ed76843d2210 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a46025896e42b1b6f65b0313aab93f : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de9776765e894ea5a02a9f24470ecff6 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99efdd33b604ede9ae191a258e95565 : SUCCESS in 14m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/945a45400acb4c9189d75cbfdfea16dd : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b361a3e989494a47b58024a9822b848d : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76e3d73738574e29bced682ce7a98d1a : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d16b7e97fe77469f999b5204158eb12a : SUCCESS in 8m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2656031387c41e39c40a33dd19b9faf : SUCCESS in 7m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f885b23ee81b4056a0ee488d72be3af1 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fce35125add64535bc00e5078d72100e : SUCCESS in 9m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6baa4c08b31045149d4783fae4e18513 : SUCCESS in 5m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ef56da56e5b4d238d5c050d82c87690 : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d60bf5cb60ad4d759659d2a3816319a1 : SUCCESS in 10m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f384aad3440043a59249743d5795465f : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7cbcc482c8584a489c33bb39228dcbe3 : SUCCESS in 6m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c6f266fdde7e4838a5354cddb2450363 : SUCCESS in 22m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/962ff750fe6c42b3922487f707edc2d6 : SUCCESS in 21m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c7ebd1cb3dfe4b109a5d62654b766032 : SUCCESS in 23m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/43afb90975704dc887fe2a8dace0b96a : SUCCESS in 21m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ad64ef27ed7f47d99a4fe20e2b5f6fe3 : SUCCESS in 16m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d92c890ea4e49868ae82f6451b335bc : SUCCESS in 14m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4b6a3b091db64b88a7bd6a0b16e59a42 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a14caa999cf84a489198a6014817d8bb : SUCCESS in 19m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/222775cd725d43d0802d3aad4c1c60bc : SUCCESS in 5m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/02efeb9edd6a43599726658d7c470c54 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c6b449b00ae7483a9f4aaef3dfd0868f : SUCCESS in 9m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d4627705c92b4edfb7ae73a9521f4842 : SUCCESS in 1h 00m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c32d0a7d5444177ac32a1365cef7d81 : SUCCESS in 1h 27m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/adb7946f027d48a6bc619eb85fa9b2a8 : SUCCESS in 1h 38m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/47f59c28a11f4186886be78c4adf667b : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":28},{"id":"6f8be4cb06b9afba3c072519908c1bcf5347f8c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:30:28.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"b6914f517e99f0ff8bffa5ea879c4d1e9594e12a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 10:12:53.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2950d10493745cdb3f5eb480bc79a61 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbbded72a73047199240a32865bc0817 : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e146a30c62047e49315e92c5637315c : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6152c4caf8464fbaa7eb610aa98ac5e9 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1011ae983074861a492624a3c7f33e0 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ba3111cad1942aea4d3146ab919dfa9 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ec95bf0b3d44bd996f2bcd0d1d7bfb1 : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ec3f827fe9a94fdb83d05b9886aa7e24 : SUCCESS in 8m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5b6b778a32d4ffe84cd5dcc6ea88199 : SUCCESS in 7m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/68e2aa7400d24e33a00072e139c281e6 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a745265359834d4d97b32cb4593f431f : SUCCESS in 9m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/94f1d62054114a698bfacd34d229a368 : SUCCESS in 5m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff5b11d5778d426ea45c6eb4cbbf32c8 : SUCCESS in 4m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cfe41f8b01014519af470c6007c5edc3 : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9c340aadc002470199ce343fc765eef4 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c49b68b1f31240f8a750faaaa09b71e6 : SUCCESS in 7m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/82bee2f5bd964d15b759280813b3238e : SUCCESS in 14m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/789cd1f7996f4dd1be221b5f842e181f : SUCCESS in 14m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ce69bdaef5304b68a754613b51729061 : SUCCESS in 24m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9ba313a26cd64187ba9c18aed9887dc0 : SUCCESS in 26m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2aa70a806a87448e916415e304bbdc48 : SUCCESS in 14m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2ef89c1325b47ccab95036eb5bc3291 : SUCCESS in 12m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ee2b8bbec5764a7aa574c759c51e96b1 : SUCCESS in 13m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc162351a07f4fbe8c9e5ce3e7828841 : SUCCESS in 18m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fde5606a18b43b7b9a876db7aa83e18 : SUCCESS in 5m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/787d710496314caba608d815903d4599 : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/750a83719247407abd304eb11a3cea4a : SUCCESS in 10m 18s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a8678aaee5484ee39d5bc688f9afcf14 : SUCCESS in 1h 04m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/153934a62c8d4549bc6df8f7463eff98 : SUCCESS in 1h 16m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/05e004f708024bec824435bfaf84b87d : SUCCESS in 1h 34m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2a48286c9fb4074a9c582ede1034372 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":28},{"id":"dd354258ccb3fb825b866fc30a6cd4e51305f938","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 10:12:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":28},{"id":"e3f96d04444ab501fe1606a9672a3d0ad3cf702b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 10:14:22.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a212dad9d54c9794632bf35a3701e1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"696228742fe3f4d76740f56c4cf2a9e02c72deab","revisions":{"ef6f106e745a6b9eba662b6a74f7203243046f5f":{"kind":"REWORK","_number":1,"created":"2019-09-11 06:16:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/1"}}},"commit":{"parents":[{"commit":"2c49134f0a4c60d30c06943c006de714abe1adab","subject":"Bump timeout for keystone tox jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c49134f0a4c60d30c06943c006de714abe1adab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-11 06:16:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-11 06:16:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f106e745a6b9eba662b6a74f7203243046f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f106e745a6b9eba662b6a74f7203243046f5f"}]},"branch":"refs/heads/master"},"123b544be30fdb8575d3a8369c26449f98e3107f":{"kind":"REWORK","_number":2,"created":"2019-09-12 06:14:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/2"}}},"commit":{"parents":[{"commit":"81deebb7be87df1a6c3cb4cf82e3bbf593e710de","subject":"Revert configparser update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81deebb7be87df1a6c3cb4cf82e3bbf593e710de"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 06:14:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-12 06:14:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/123b544be30fdb8575d3a8369c26449f98e3107f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/123b544be30fdb8575d3a8369c26449f98e3107f"}]},"branch":"refs/heads/master"},"548bcb37af3c0e3d268c3b006424276a80437626":{"kind":"REWORK","_number":3,"created":"2019-09-13 06:34:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/3"}}},"commit":{"parents":[{"commit":"e8ff05b17776fececb7d8d78c4cbb3ac397aa501","subject":"update constraint for python-watcherclient to new release 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8ff05b17776fececb7d8d78c4cbb3ac397aa501"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 06:34:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-13 06:34:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/548bcb37af3c0e3d268c3b006424276a80437626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/548bcb37af3c0e3d268c3b006424276a80437626"}]},"branch":"refs/heads/master"},"6cd1b8588f918e31c2ecd2ed3a67cb9ea4342377":{"kind":"REWORK","_number":4,"created":"2019-09-14 06:11:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/4"}}},"commit":{"parents":[{"commit":"af3c9ff0053d71144ca8d04bfb32c4394d32cce3","subject":"Merge \"update constraint for python-keystoneclient to new release 3.21.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af3c9ff0053d71144ca8d04bfb32c4394d32cce3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-14 06:11:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-14 06:11:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cd1b8588f918e31c2ecd2ed3a67cb9ea4342377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cd1b8588f918e31c2ecd2ed3a67cb9ea4342377"}]},"branch":"refs/heads/master"},"085536db4de480198fd122fa429722da7eafc46c":{"kind":"REWORK","_number":5,"created":"2019-09-15 06:14:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/5"}}},"commit":{"parents":[{"commit":"9fa1683b4f50aefc94eb2b5bba99b1c320001ecb","subject":"Merge \"update constraint for python-blazarclient to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fa1683b4f50aefc94eb2b5bba99b1c320001ecb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-15 06:14:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-15 06:14:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/085536db4de480198fd122fa429722da7eafc46c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/085536db4de480198fd122fa429722da7eafc46c"}]},"branch":"refs/heads/master"},"317f65422b17e49b880d08f21a2762f1ce7ac88a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-09-16 06:26:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/6"}}},"commit":{"parents":[{"commit":"2c42d7f4d3e31ddb78a2fa42c29a0475249c0871","subject":"Merge \"update constraint for tricircleclient to new release 0.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c42d7f4d3e31ddb78a2fa42c29a0475249c0871"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-16 06:26:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-16 06:26:06.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317f65422b17e49b880d08f21a2762f1ce7ac88a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/317f65422b17e49b880d08f21a2762f1ce7ac88a"}]},"branch":"refs/heads/master"},"5bbb251ccfb12b79ed05723ccff58005926eef5a":{"kind":"REWORK","_number":7,"created":"2019-09-17 06:38:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/7"}}},"commit":{"parents":[{"commit":"ae07219610bda5dcc1f3aa21fef8702228db6767","subject":"Merge \"update constraint for python-manilaclient to new release 1.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae07219610bda5dcc1f3aa21fef8702228db6767"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-17 06:38:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-17 06:38:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bbb251ccfb12b79ed05723ccff58005926eef5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bbb251ccfb12b79ed05723ccff58005926eef5a"}]},"branch":"refs/heads/master"},"66ee197846ff634fb607fb2744ae609c1994e702":{"kind":"REWORK","_number":8,"created":"2019-09-18 06:17:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/8"}}},"commit":{"parents":[{"commit":"ae07219610bda5dcc1f3aa21fef8702228db6767","subject":"Merge \"update constraint for python-manilaclient to new release 1.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae07219610bda5dcc1f3aa21fef8702228db6767"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 06:17:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-18 06:17:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66ee197846ff634fb607fb2744ae609c1994e702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66ee197846ff634fb607fb2744ae609c1994e702"}]},"branch":"refs/heads/master"},"035d78ffbe2776224f6fd49a2347a767ee2aa18e":{"kind":"REWORK","_number":9,"created":"2019-09-19 06:47:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/9"}}},"commit":{"parents":[{"commit":"70354893f2108e4e8076705291133bc7e15348ba","subject":"Merge \"update constraint for openstacksdk to new release 0.36.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70354893f2108e4e8076705291133bc7e15348ba"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-19 06:47:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-19 06:47:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/035d78ffbe2776224f6fd49a2347a767ee2aa18e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/035d78ffbe2776224f6fd49a2347a767ee2aa18e"}]},"branch":"refs/heads/master"},"9f5876573fd8516e1463fdee9db642b56b0c0363":{"kind":"REWORK","_number":10,"created":"2019-09-20 06:27:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/10"}}},"commit":{"parents":[{"commit":"9ab183ab3992232c9baed2137ab822a6891a3bf3","subject":"Merge \"update constraint for kuryr-lib to new release 1.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ab183ab3992232c9baed2137ab822a6891a3bf3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-20 06:27:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-20 06:27:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f5876573fd8516e1463fdee9db642b56b0c0363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f5876573fd8516e1463fdee9db642b56b0c0363"}]},"branch":"refs/heads/master"},"f16f386746c140cd878db378931c026ac0c5c784":{"kind":"REWORK","_number":11,"created":"2019-09-21 06:36:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/11"}}},"commit":{"parents":[{"commit":"f1ce61b86ec7226acd620024ffd1857b15105c8b","subject":"Merge \"update constraint for paunch to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1ce61b86ec7226acd620024ffd1857b15105c8b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-21 06:36:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-21 06:36:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f16f386746c140cd878db378931c026ac0c5c784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f16f386746c140cd878db378931c026ac0c5c784"}]},"branch":"refs/heads/master"},"5876deeae164c049a5ff151bab97ba05c79ab89d":{"kind":"REWORK","_number":12,"created":"2019-09-22 06:22:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/12"}}},"commit":{"parents":[{"commit":"f1ce61b86ec7226acd620024ffd1857b15105c8b","subject":"Merge \"update constraint for paunch to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1ce61b86ec7226acd620024ffd1857b15105c8b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-22 06:22:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-22 06:22:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5876deeae164c049a5ff151bab97ba05c79ab89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5876deeae164c049a5ff151bab97ba05c79ab89d"}]},"branch":"refs/heads/master"},"83d758b16e024af6a1075465285b5be544121c48":{"kind":"REWORK","_number":13,"created":"2019-09-23 06:29:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/13"}}},"commit":{"parents":[{"commit":"bf19087eac2578f85bc7f2ef09e9e88b0eb610c5","subject":"Merge \"update constraint for os-apply-config to new release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf19087eac2578f85bc7f2ef09e9e88b0eb610c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-23 06:29:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-23 06:29:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83d758b16e024af6a1075465285b5be544121c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83d758b16e024af6a1075465285b5be544121c48"}]},"branch":"refs/heads/master"},"a65c460947e9ad6e75e54aee039ce2342ee48654":{"kind":"REWORK","_number":14,"created":"2019-09-24 06:15:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/14"}}},"commit":{"parents":[{"commit":"bf19087eac2578f85bc7f2ef09e9e88b0eb610c5","subject":"Merge \"update constraint for os-apply-config to new release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf19087eac2578f85bc7f2ef09e9e88b0eb610c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-24 06:15:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-24 06:15:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a65c460947e9ad6e75e54aee039ce2342ee48654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a65c460947e9ad6e75e54aee039ce2342ee48654"}]},"branch":"refs/heads/master"},"8638795eb0332e32db38624faedbbc290407b384":{"kind":"REWORK","_number":15,"created":"2019-09-25 06:20:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/15"}}},"commit":{"parents":[{"commit":"bf19087eac2578f85bc7f2ef09e9e88b0eb610c5","subject":"Merge \"update constraint for os-apply-config to new release 10.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf19087eac2578f85bc7f2ef09e9e88b0eb610c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-25 06:20:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-25 06:20:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8638795eb0332e32db38624faedbbc290407b384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8638795eb0332e32db38624faedbbc290407b384"}]},"branch":"refs/heads/master"},"54768c1c4661755f138e18599af4d4da4da1fd00":{"kind":"REWORK","_number":16,"created":"2019-09-26 06:32:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/16"}}},"commit":{"parents":[{"commit":"23fbf0ac452a7a2edb0414f3c793d94472bc5780","subject":"Merge \"Skip alembic 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23fbf0ac452a7a2edb0414f3c793d94472bc5780"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-26 06:32:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-26 06:32:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54768c1c4661755f138e18599af4d4da4da1fd00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54768c1c4661755f138e18599af4d4da4da1fd00"}]},"branch":"refs/heads/master"},"bb6f4e4b02d0a829c5e61dff57f3e59ca24acdc7":{"kind":"REWORK","_number":17,"created":"2019-09-27 06:17:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/17"}}},"commit":{"parents":[{"commit":"130daa293612969e3e4b2ef168151737c22ec5da","subject":"Merge \"Update upper-constraint for stestr to 2.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130daa293612969e3e4b2ef168151737c22ec5da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-27 06:17:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-27 06:17:14.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb6f4e4b02d0a829c5e61dff57f3e59ca24acdc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb6f4e4b02d0a829c5e61dff57f3e59ca24acdc7"}]},"branch":"refs/heads/master"},"78439fff4a1af526d5f02445cc6c29882a0c8848":{"kind":"REWORK","_number":18,"created":"2019-09-28 06:18:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/18"}}},"commit":{"parents":[{"commit":"34dd20b9caa8463dcdaa50189946930d5b30a2e6","subject":"Merge \"update constraint for python-tackerclient to new release 0.16.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34dd20b9caa8463dcdaa50189946930d5b30a2e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-28 06:18:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-28 06:18:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78439fff4a1af526d5f02445cc6c29882a0c8848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78439fff4a1af526d5f02445cc6c29882a0c8848"}]},"branch":"refs/heads/master"},"f4adb1a142d9cea6a94fc555113e4bac88652967":{"kind":"REWORK","_number":19,"created":"2019-09-29 06:45:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/19"}}},"commit":{"parents":[{"commit":"34dd20b9caa8463dcdaa50189946930d5b30a2e6","subject":"Merge \"update constraint for python-tackerclient to new release 0.16.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34dd20b9caa8463dcdaa50189946930d5b30a2e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-29 06:45:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-29 06:45:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4adb1a142d9cea6a94fc555113e4bac88652967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4adb1a142d9cea6a94fc555113e4bac88652967"}]},"branch":"refs/heads/master"},"48f966215af7e6dcc1450c217108d99fb9307a5b":{"kind":"REWORK","_number":20,"created":"2019-09-30 06:18:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/20"}}},"commit":{"parents":[{"commit":"34dd20b9caa8463dcdaa50189946930d5b30a2e6","subject":"Merge \"update constraint for python-tackerclient to new release 0.16.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34dd20b9caa8463dcdaa50189946930d5b30a2e6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-30 06:18:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-30 06:18:40.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48f966215af7e6dcc1450c217108d99fb9307a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48f966215af7e6dcc1450c217108d99fb9307a5b"}]},"branch":"refs/heads/master"},"692e5946975b93ba60cf093dc537ce494ee08bf5":{"kind":"REWORK","_number":21,"created":"2019-10-01 06:19:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/21"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-01 06:19:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-01 06:19:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/692e5946975b93ba60cf093dc537ce494ee08bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/692e5946975b93ba60cf093dc537ce494ee08bf5"}]},"branch":"refs/heads/master"},"f4d2a24cf4bdbea4022b9c32042aa32485a6f354":{"kind":"REWORK","_number":22,"created":"2019-10-02 06:22:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/22"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-02 06:21:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-02 06:21:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4d2a24cf4bdbea4022b9c32042aa32485a6f354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4d2a24cf4bdbea4022b9c32042aa32485a6f354"}]},"branch":"refs/heads/master"},"de511d389396555cfed33278ae7d8c612195fbe4":{"kind":"REWORK","_number":23,"created":"2019-10-03 06:14:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/23"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-03 06:14:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-03 06:14:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de511d389396555cfed33278ae7d8c612195fbe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de511d389396555cfed33278ae7d8c612195fbe4"}]},"branch":"refs/heads/master"},"3c033ae28f5fa3ed26644539d33be4e1847d2e68":{"kind":"REWORK","_number":24,"created":"2019-10-04 06:12:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/24"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-04 06:12:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-04 06:12:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c033ae28f5fa3ed26644539d33be4e1847d2e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c033ae28f5fa3ed26644539d33be4e1847d2e68"}]},"branch":"refs/heads/master"},"d414391930e9c5072353e280316de3007e0f3714":{"kind":"REWORK","_number":25,"created":"2019-10-05 06:53:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/25"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-05 06:53:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-05 06:53:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d414391930e9c5072353e280316de3007e0f3714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d414391930e9c5072353e280316de3007e0f3714"}]},"branch":"refs/heads/master"},"720c0a299c538c5573aff73b55175e9d31e758d6":{"kind":"REWORK","_number":26,"created":"2019-10-07 06:11:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/681382/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/26"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 06:11:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 06:11:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/720c0a299c538c5573aff73b55175e9d31e758d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/720c0a299c538c5573aff73b55175e9d31e758d6"}]},"branch":"refs/heads/master"},"8cd122e71553a4d6078385f9b9ab83872f24a307":{"kind":"REWORK","_number":27,"created":"2019-10-07 15:40:11.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/681382/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/27"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 06:11:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-07 15:39:29.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back amqp-2.5.2\nwaiting for https://github.com/celery/kombu/commit/b51d1d678e198a80d7e5fd95f32674c7d8e04a75\nto be in a release...\n\nhard pinning sucks\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cd122e71553a4d6078385f9b9ab83872f24a307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cd122e71553a4d6078385f9b9ab83872f24a307"}]},"branch":"refs/heads/master"},"696228742fe3f4d76740f56c4cf2a9e02c72deab":{"kind":"REWORK","_number":28,"created":"2019-10-07 20:46:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/681382/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/681382/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/681382/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/681382/28"}}},"commit":{"parents":[{"commit":"aaf24f51199e298c89c73fcb1bbb4e488c52bde0","subject":"update constraint for glance_store to new release 1.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aaf24f51199e298c89c73fcb1bbb4e488c52bde0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-07 06:11:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-10-07 20:46:12.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back amqp-2.5.2\nwaiting for https://github.com/celery/kombu/commit/b51d1d678e198a80d7e5fd95f32674c7d8e04a75\nto be in a release...\n\nhard pinning sucks\n\nsalt also caps pyyaml, I\u0027m siding with pyyaml\n\nChange-Id: I0b2eae0e848416e05430278a5a934a256b0ee001\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/696228742fe3f4d76740f56c4cf2a9e02c72deab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/696228742fe3f4d76740f56c4cf2a9e02c72deab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
