)]}'
{"id":"openstack%2Frequirements~681399","triplet_id":"openstack%2Frequirements~master~I774a6c4d0454c17e180a5ebd77f97fd2dc0617bd","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I774a6c4d0454c17e180a5ebd77f97fd2dc0617bd","subject":"Fix misspell word","status":"ABANDONED","created":"2019-09-11 08:05:16.000000000","updated":"2019-09-18 13:47:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b6ec93a17a2f1b670f4dd11f34956d19bf8747f","_number":681399,"virtual_id_number":681399,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-11 12:52:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 12:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b04706f8cf7bcde7f0c13c915e32095d8cd63b4","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-11 08:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53464d630e47d650f3955a4f5a786215b969c075","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 12:52:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd758582e5d4b86b54d1be40c34868e : SUCCESS in 3m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d90dc4f55c51411eafa0461fe61b6403 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3c8e8c882d4f6c9f8b7a823e90180a : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6932575351ba42fa901fa3f49a716b45 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/299505cfa1444e5cb7bfe6d5c209ac48 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20e6095344b54cb99c45d7ae2783cfb3 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c5e211bd7554c678e57f09854a39a7e : SUCCESS in 3m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e608b0321a5c47178fafc1593cfb3fa0 : SUCCESS in 1h 11m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/42f4a0886a104506a8d2b0486ec91f4e : FAILURE in 1h 49m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a47135194c14268b45a8ae0c568cabd : TIMED_OUT in 2h 08m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1d501d2eb9404a3289a2a9c3bc75933f : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6b6ec93a17a2f1b670f4dd11f34956d19bf8747f","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-18 13:47:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c90e28f175685ccd34b53f98b699f7f32ab34ec","revisions":{"3c90e28f175685ccd34b53f98b699f7f32ab34ec":{"kind":"REWORK","_number":1,"created":"2019-09-11 08:05:16.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/99/681399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/681399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/681399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/681399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/681399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/681399/1"}}},"commit":{"parents":[{"commit":"2c49134f0a4c60d30c06943c006de714abe1adab","subject":"Bump timeout for keystone tox jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c49134f0a4c60d30c06943c006de714abe1adab"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-11 08:04:22.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-11 08:04:54.000000000","tz":480},"subject":"Fix misspell word","message":"Fix misspell word\n\nChange-Id: I774a6c4d0454c17e180a5ebd77f97fd2dc0617bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c90e28f175685ccd34b53f98b699f7f32ab34ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c90e28f175685ccd34b53f98b699f7f32ab34ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
